)]}'
{"id":"openstack%2Fopenstack-ansible-os_keystone~785393","triplet_id":"openstack%2Fopenstack-ansible-os_keystone~stable%2Fussuri~Ic52fc7dbdb0324ab8f4b71d25398f23a05df05d7","project":"openstack/openstack-ansible-os_keystone","branch":"stable/ussuri","hashtags":[],"change_id":"Ic52fc7dbdb0324ab8f4b71d25398f23a05df05d7","subject":"Use absolute path for uwsgi_params include","status":"MERGED","created":"2021-04-09 06:25:33.000000000","updated":"2021-07-20 21:20:49.000000000","submitted":"2021-07-20 21:19:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"785393","cherry_pick_of_change":775911,"cherry_pick_of_patch_set":1,"meta_rev_id":"4e3c5cb8b073b52d0b9ba4d87c2f06d504f87bbd","_number":785393,"virtual_id_number":785393,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-20 21:19:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"all":[{"value":2,"date":"2021-07-20 21:20:49.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-07-20 21:20:49.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":1,"date":"2021-07-20 17:55:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-09 08:56:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-10 10:52:53.000000000","updated_by":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"reviewer":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"state":"REVIEWER"},{"updated":"2021-07-12 14:19:28.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"6e7f843912e939991b102cf8364bf3e5727bacca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-04-09 06:25:33.000000000","message":"Patch Set 1: Cherry Picked from branch stable/victoria.","accounts_in_message":[],"_revision_number":1},{"id":"e97a3ff8883b32d09872740b903de125c39499e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-09 08:56:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/4ce21aa31d5e4b29ae6ecee58a187033 : SUCCESS in 13m 07s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/4d3e2159f4f74f66b819d52de6c887e3 : SUCCESS in 1h 06m 50s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/6150f41dc3ec4273ae9468a3f0ac0e61 : SUCCESS in 1h 18m 24s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/c81c833b75294e08a7a0251f2d911067 : SUCCESS in 1h 16m 59s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e551439b68e347ea88e3b22ab5320783 : SUCCESS in 1h 16m 00s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e52276aa41d04d8b89544768690d77e0 : SUCCESS in 1h 05m 19s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/fca0f6b0c9994126bbe0ac37b38ce327 : SUCCESS in 2h 15m 55s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/28922383003048239925354c9e921e9b : SUCCESS in 1h 43m 19s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/698fd81dc12340d69671a6dc7a18db50 : SUCCESS in 54m 26s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/98164e2c194d4fa48447258681defdc4 : SUCCESS in 57m 56s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e6c394f3f1a74a948c75ef9114cfec15 : SUCCESS in 1h 47m 59s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f081de62262c4f47aec7e776ee18a990 : SUCCESS in 56m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8942923c51d42b3ba01cff6f6371497 : SUCCESS in 8m 24s\n- openstack-ansible-uw_apache-centos-7 https://zuul.opendev.org/t/openstack/build/a11ff2de5bb54500ad00241affef5f1e : FAILURE in 38m 40s\n- openstack-ansible-uw_apache-centos-8 https://zuul.opendev.org/t/openstack/build/01034eabb4b64014889cee3d3be18fc8 : FAILURE in 44m 11s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5bb53c634af24764baafbdc67fd84cdd : FAILURE in 34m 16s","accounts_in_message":[],"_revision_number":1},{"id":"9a455b39e0eb5d6f8ef9703ae8fa4af5e20872c1","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-06-03 11:13:02.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"81b50831b719f08bf0ea32ea03e7d67703582c2c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-03 13:38:49.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/310ba6be0dc341dd87700af145fabe7a : SUCCESS in 9m 58s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/e68dd4219dfe4947b8d4d6f21954a7d2 : SUCCESS in 1h 09m 06s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/ae53c14248b7449ab141cbb77cd90867 : SUCCESS in 1h 17m 27s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/07f33f0d0f22433e846d9daffb4ba645 : SUCCESS in 1h 14m 25s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/59cb7d8cbd674caaa28f552039e51dfc : FAILURE in 28m 11s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/33cc68d56da542c295e113a41e54fe70 : SUCCESS in 1h 03m 28s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/f56373617e9a4063809f3131e07aaefa : SUCCESS in 2h 21m 29s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/eb9489f7b5434fcbbb9f381a84605264 : SUCCESS in 1h 27m 39s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/899bb47815f3465db01161713c23f7cc : SUCCESS in 1h 06m 16s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/4b3ffe42143540e2af7e09cad703aac3 : SUCCESS in 50m 42s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/464ce24437ed4b5aa2bf66916b3e87a1 : SUCCESS in 1h 05m 18s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8a6f084bfa754d98bc91e8c0d7e51cb5 : SUCCESS in 46m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b4010ec95b24c6b9c8a168bd1e8e4db : SUCCESS in 7m 14s\n- openstack-ansible-uw_apache-centos-7 https://zuul.opendev.org/t/openstack/build/97a0a7014e2142cbbb018dde71d786f3 : FAILURE in 35m 19s\n- openstack-ansible-uw_apache-centos-8 https://zuul.opendev.org/t/openstack/build/d9ff379d4a044401a650cda3ee658616 : FAILURE in 42m 35s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/203c7dfd28aa409ca85f468908cc4b4e : FAILURE in 34m 44s","accounts_in_message":[],"_revision_number":1},{"id":"88ea8fcee5f856e5b5082e704d044542d67e18ae","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-07-07 19:49:18.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e0fd042742e0e8d1c86d27cee43b5bdc40ff0b10","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-07 22:08:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/78aded72d9994cf594fce2e93d7fc4a5 : SUCCESS in 9m 49s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/0bda8bc472c44fa384c0b1c555512106 : SUCCESS in 1h 06m 39s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/315b96a478e5414683cd211e2d077c6e : SUCCESS in 1h 22m 32s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/eca8bfe0fbe14a1d885ef22532ede064 : SUCCESS in 1h 17m 53s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/02740c4b7fc04ed2b754cc1f2d932041 : SUCCESS in 1h 13m 05s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cfd9086f9bfd41f9af0a32a9195ae65e : SUCCESS in 1h 10m 00s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/29a0dd2cfcaa45e9acd854cc699661b8 : SUCCESS in 2h 16m 46s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c2aa2a6297f1417d85597852e391d0f7 : SUCCESS in 1h 52m 17s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/5ffcde54c19b44e5921e46682dfd3898 : SUCCESS in 1h 09m 25s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/b4baf23793674c6880d99b9553760f0a : SUCCESS in 58m 31s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6fcac1f99bb44cdaa4052872f0b3542c : SUCCESS in 1h 01m 32s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c9d0cf3988f149148fa75ce70ce6bae6 : SUCCESS in 58m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1899f683d59f48dea4ed79b2768f5a12 : SUCCESS in 6m 53s\n- openstack-ansible-uw_apache-centos-7 https://zuul.opendev.org/t/openstack/build/0afe22b8a3dc4534a0723164e01330e7 : SUCCESS in 38m 44s\n- openstack-ansible-uw_apache-centos-8 https://zuul.opendev.org/t/openstack/build/5f7ac2b91d274ced8e057a59865a80bf : SUCCESS in 46m 34s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/52575ae556b949c1b2d3951abdddcc18 : SUCCESS in 35m 55s","accounts_in_message":[],"_revision_number":1},{"id":"58e2e738a6d7ea16125b0fc5751964d6bd61f9b8","author":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"date":"2021-07-10 10:52:53.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"23b4ae2af0000db36633eb0fb1ca36768b2c0d1d","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-07-12 14:19:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"aa118bfeb09c01469c82fd8096d3a561b9c1b8ee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-12 14:20:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d666ee2fa2ed5241dae4a5c684a82192863391a5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-12 15:57:47.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/dc8bb764103b40c495503504daf4d0ca : SUCCESS in 9m 33s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/a05460f82fc746cebae511647de6a37c : SUCCESS in 1h 13m 55s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/63ba2f2a14b441d2bc0c6000546e7487 : SUCCESS in 1h 37m 26s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/269e1734a9274b228b09e0dac463231a : SUCCESS in 1h 30m 25s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1cc2894d3e404fbe9f99de1d90778294 : SUCCESS in 1h 19m 06s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6a2002dc22134bbbac1bfff6b1e5d338 : SUCCESS in 1h 17m 56s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/3c9ad4e987924f2dbe20ef7749cbe160 : SUCCESS in 1h 05m 23s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/76b3ea9f92b243a4997a0b86a1f1ed26 : SUCCESS in 1h 07m 59s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d038fcf69c724c04ad26d26b71a69995 : SUCCESS in 1h 04m 05s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e17c9b5c8bd24a71b805518ac718f036 : SUCCESS in 1h 03m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7a6fc12681f43238152341dcce2fd24 : SUCCESS in 8m 56s\n- openstack-ansible-uw_apache-centos-7 https://zuul.opendev.org/t/openstack/build/1bcab8e83f5d444bb9f7b3a3b711b48a : FAILURE in 15m 18s\n- openstack-ansible-uw_apache-centos-8 https://zuul.opendev.org/t/openstack/build/881c2a40fc53431cbe292e6cc39a9309 : SUCCESS in 43m 56s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ae66ced6863f44e7b697f2019ac17ee6 : SUCCESS in 38m 13s","accounts_in_message":[],"_revision_number":1},{"id":"a2d82c1cfadbd37940482c514ff34927c7f7b4b4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-07-20 17:23:35.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"cc3a27c750147c4b6f91eabbf22c9121602ff2dc","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-07-20 17:55:29.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"38263226466f33dfec380c7102ae898686ef2da3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-20 19:47:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/9c6fa5c0db3d44e7ae9d19827d77b108 : SUCCESS in 8m 34s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/a9bd081d0493496e9a36fba424e1baa2 : SUCCESS in 1h 15m 25s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/9275f331bc5948a3ba82670fee4e94a8 : SUCCESS in 1h 44m 52s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/3063ca2f294443cfbecbb9daf3765a81 : SUCCESS in 1h 20m 25s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1040b0ffd11a44058379e6b62af51926 : SUCCESS in 1h 07m 11s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c8681b2b043445648d0cf165ddbf0b0c : SUCCESS in 1h 10m 50s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/9f5f31cef4fe4f259fce2a48effc7b14 : SUCCESS in 2h 21m 48s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e8f5e24ef80841f2926b52e3d986fc5d : SUCCESS in 1h 50m 17s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/870679049e8d45feb617e0e3efce98d0 : SUCCESS in 1h 09m 29s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/49fdf5d564544b198fe29fc8d91fa83b : SUCCESS in 1h 07m 47s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2cdf77aa42094e35a1658c3e0979c846 : SUCCESS in 1h 03m 11s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e324d787bf794366b0eb12f2d35553b3 : SUCCESS in 1h 05m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8d68eb2a26a4db1b9320387c967881d : SUCCESS in 6m 11s\n- openstack-ansible-uw_apache-centos-7 https://zuul.opendev.org/t/openstack/build/a20feef58b164b3299e12f3d0faed9f5 : SUCCESS in 42m 25s\n- openstack-ansible-uw_apache-centos-8 https://zuul.opendev.org/t/openstack/build/4f1dd867c1cd4e1ba4be13f5894fa579 : SUCCESS in 41m 10s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b1d7106506bd4b83b305d013a3fda811 : SUCCESS in 37m 22s","accounts_in_message":[],"_revision_number":2},{"id":"10805a7e040d63b5bbc7081740964fb425c368c9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-20 19:49:04.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9b95e37b7a8e74d15b84cb726b217b060198c1be","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-20 21:19:23.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/d859dae0829441009be46bbe15d04b3e : SUCCESS in 9m 12s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/d3a011e5b3604faf9cb02860b011ff9c : SUCCESS in 1h 01m 19s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/c9e6c1926a3c4f85a6b19def0db72774 : SUCCESS in 1h 26m 52s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/b43d7c789ae44203a6b4a06a82b5e92f : SUCCESS in 1h 22m 42s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/073cd4b4450747968de5fc06d9970dc5 : SUCCESS in 1h 17m 44s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/37f90acd4e434283a3b324dd4e64999b : SUCCESS in 1h 07m 52s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/d371a6cdafea4fcda93fe7a89a688a5f : SUCCESS in 1h 11m 58s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/5ce4181485af4bd8b8bbaba557047dd4 : SUCCESS in 59m 01s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f7925aa42543478ca67c25d822c3b213 : SUCCESS in 55m 58s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8e253dc50c6a40a08442702cd1311155 : SUCCESS in 55m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9642c6188e14b11bd0ecd7e0f4929a7 : SUCCESS in 5m 39s\n- openstack-ansible-uw_apache-centos-7 https://zuul.opendev.org/t/openstack/build/8116ea0fb3954d2ca4ec2d56f9a9ec7c : SUCCESS in 37m 01s\n- openstack-ansible-uw_apache-centos-8 https://zuul.opendev.org/t/openstack/build/f4553cc366d5410195579cc87905f576 : SUCCESS in 44m 31s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/585826ff47bc4f3198fcb9484b411c45 : SUCCESS in 36m 47s","accounts_in_message":[],"_revision_number":2},{"id":"a7eaaa977c1a49b2e3e9ef33aa71e375caaffba7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-20 21:19:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"87e6c217599e362c3133afb3207d7bf4d7f704fc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-20 21:20:49.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1afde64334e476fb6eba8019cd24a4c : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b8154166bf7c91502b9f088261f0ec409cccc5d1","revisions":{"5b5fef649f85ab9f657d961c00610dc3c3a415a0":{"kind":"REWORK","_number":1,"created":"2021-04-09 06:25:33.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/93/785393/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/93/785393/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/93/785393/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/93/785393/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/93/785393/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/93/785393/1"}}},"commit":{"parents":[{"commit":"9b75e5071520533bd0e5083029dd344d16beafcc","subject":"Allow OIDCClaimDelimiter to be set in the apache config file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/9b75e5071520533bd0e5083029dd344d16beafcc"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2021-02-16 09:18:08.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-04-09 06:25:33.000000000","tz":0},"subject":"Use absolute path for uwsgi_params include","message":"Use absolute path for uwsgi_params include\n\nNginx config verification that is performed by ansible [1] is made\nin tmp \"on fly\" which fails because of the relative import.\n\nWe also move task that replaces ports for nginx.conf to the end\nso that config validation was accomplished after all configurations\nare applied.\n\n[1] https://opendev.org/openstack/openstack-ansible-os_keystone/src/branch/master/tasks/keystone_nginx.yml#L44\n\nChange-Id: Ic52fc7dbdb0324ab8f4b71d25398f23a05df05d7\n(cherry picked from commit e377209297dd359c2191c8b1c23eae2ca21fe5eb)\n(cherry picked from commit ad079a5bade97424b69849ce675743f8440a518b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/5b5fef649f85ab9f657d961c00610dc3c3a415a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/5b5fef649f85ab9f657d961c00610dc3c3a415a0"}]},"branch":"refs/heads/stable/ussuri"},"b8154166bf7c91502b9f088261f0ec409cccc5d1":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-07-20 17:23:35.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/93/785393/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/93/785393/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/93/785393/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/93/785393/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/93/785393/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/93/785393/2"}}},"commit":{"parents":[{"commit":"be97806eff60725216aa86197748fdfc988e306d","subject":"Fix typo in keystone-httpd template","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/be97806eff60725216aa86197748fdfc988e306d"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2021-02-16 09:18:08.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-07-20 17:23:35.000000000","tz":0},"subject":"Use absolute path for uwsgi_params include","message":"Use absolute path for uwsgi_params include\n\nNginx config verification that is performed by ansible [1] is made\nin tmp \"on fly\" which fails because of the relative import.\n\nWe also move task that replaces ports for nginx.conf to the end\nso that config validation was accomplished after all configurations\nare applied.\n\n[1] https://opendev.org/openstack/openstack-ansible-os_keystone/src/branch/master/tasks/keystone_nginx.yml#L44\n\nChange-Id: Ic52fc7dbdb0324ab8f4b71d25398f23a05df05d7\n(cherry picked from commit e377209297dd359c2191c8b1c23eae2ca21fe5eb)\n(cherry picked from commit ad079a5bade97424b69849ce675743f8440a518b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/b8154166bf7c91502b9f088261f0ec409cccc5d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/b8154166bf7c91502b9f088261f0ec409cccc5d1"}]},"branch":"refs/heads/stable/ussuri","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
