)]}'
{"id":"openstack%2Fopenstack-ansible-os_keystone~792610","triplet_id":"openstack%2Fopenstack-ansible-os_keystone~master~I22c4f987aedba049a953c955395b06803de37c27","project":"openstack/openstack-ansible-os_keystone","branch":"master","topic":"integrated_linters","hashtags":[],"change_id":"I22c4f987aedba049a953c955395b06803de37c27","subject":"Replace linters test with integarted one","status":"MERGED","created":"2021-05-21 12:53:01.000000000","updated":"2021-05-26 02:45:31.000000000","submitted":"2021-05-26 02:42:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"792610-integrated_linters","meta_rev_id":"7a5f74faf9c2a5fa7eeae62272f8306320bbd39b","_number":792610,"virtual_id_number":792610,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-26 02:42:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2021-05-25 08:55:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-05-25 14:19:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-05-25 14:19:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-21 15:35:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-25 08:55:43.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2021-05-25 14:19:24.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"090d5748d3c8aec9f40b4760b220ab740b099c02","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-05-21 12:53:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e84edc4d216bbee7943d1b39910070306afdc58e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-21 15:35:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c9156e94afc74129a2ec3340a5a8cacf : SUCCESS in 26m 16s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/b380cf005e874373882fe93083910c19 : SUCCESS in 1h 17m 42s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/f8db19e70ece4d069af49749c4692fd5 : SUCCESS in 1h 30m 24s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/5eab9283e62846939220452f0130278d : SUCCESS in 1h 26m 19s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5e3607b2a26d4b0da02012d238e2020e : SUCCESS in 1h 27m 27s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7fb8a3699bee45c8b851cf9f745cf890 : SUCCESS in 1h 15m 58s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/7164002230214713a9e988d690f1c227 : SUCCESS in 1h 54m 39s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/651cf39be6c446d2a538b1256f031c9e : SUCCESS in 1h 47m 57s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/55e14d4acb7f4e03b5ca1baaa35f21cc : SUCCESS in 1h 03m 46s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6b17128cbf6e4956b1991d19ef7c3302 : SUCCESS in 1h 03m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1fa5e73f0cd04e98a149af7257116bc5 : SUCCESS in 6m 35s\n- openstack-ansible-uw_apache-centos-8 https://zuul.opendev.org/t/openstack/build/bc912a2c4ef741ab9129bd35eadb4cfb : SUCCESS in 49m 45s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/65bbffda74b8433abdc2f562a4026cdb : SUCCESS in 44m 01s\n- openstack-ansible-uw_apache-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4592ca38535d4711aaab480ba03ce9a9 : SUCCESS in 50m 24s","accounts_in_message":[],"_revision_number":1},{"id":"691972abcbeedd7ce6e8829ea8095151dc488e57","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2021-05-25 08:55:43.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a80b8e413ef620b12d92128d076b720d87ccfc89","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-05-25 14:19:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"dfc54831f3d76a79736f0a4dee661f207b233c01","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-25 14:20:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c63201e90581b8f21ee7ff04431e9c73fc5141c0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-25 15:58:07.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aed7ad1abdb849dfb5cda0bdfcc18e37 : SUCCESS in 30m 55s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/9f322494a3694e5c8f78a9e70929fc49 : SUCCESS in 1h 20m 19s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/d1dee3e53ae94d4292e44f2c10fc7ba0 : SUCCESS in 1h 31m 30s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/c97dd7deb2bd4969a283af732cbbde6e : FAILURE in 1h 19m 01s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/196ca3627a9749a491643be18a977ef1 : SUCCESS in 1h 24m 37s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/68426c7a67884697af4271f8a96c4dbd : SUCCESS in 1h 10m 21s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/1f650965126240a78c2d267ea99a5c4d : FAILURE in 1h 01m 17s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f8cb7002774346d4bb5c2b66f94b0e51 : SUCCESS in 1h 02m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fded9e2e55fc4511913d69f2041c4cea : SUCCESS in 5m 52s\n- openstack-ansible-uw_apache-centos-8 https://zuul.opendev.org/t/openstack/build/b640a561fb9a4089a7a14b07494a866a : SUCCESS in 46m 08s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c5ad476db71f437aa1a39977273112ce : SUCCESS in 44m 27s\n- openstack-ansible-uw_apache-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ddf8a84d787c4f319548dedf523b7da7 : SUCCESS in 44m 21s","accounts_in_message":[],"_revision_number":1},{"id":"ccf5c8fe4c85a8b6e21cdf4dc8fbc87d0531ed7e","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-05-25 23:29:59.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"366ea21cc1b18c49c480621b22dac34bfcd22b48","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-26 01:24:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/206f180bc1874f7782e6282ac87faa29 : SUCCESS in 24m 45s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/e38ea20fd9b143a5a90861ca70f6cf71 : SUCCESS in 1h 12m 21s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/cd3e1556237a4de98c3cacd66c0e0b85 : SUCCESS in 1h 23m 01s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/f4a32a5827624582a71a2ca4045b7be9 : SUCCESS in 1h 10m 28s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3b4815075053485ea2ad140f82dd4fbb : SUCCESS in 1h 22m 45s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/53d88163333542be91de651b74612520 : SUCCESS in 1h 08m 17s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/b154c491856745eab2c79d1dce71cb66 : SUCCESS in 1h 36m 17s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8f2c16190f7c4a7f828b9c0f46d1ff8f : SUCCESS in 1h 29m 54s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/55f17fd2e4eb436db012ceb4687d5a0b : SUCCESS in 1h 00m 27s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/67fdaeee5dcd49f9bbceb041ad2fe6e3 : SUCCESS in 53m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6aba8aeda3fe4eaab9766fae7e1ce9ce : SUCCESS in 6m 35s\n- openstack-ansible-uw_apache-centos-8 https://zuul.opendev.org/t/openstack/build/43a35e932a5d49b895b425f406c84b70 : SUCCESS in 48m 32s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5d82ca63154e4894a1ba02247586d385 : SUCCESS in 41m 00s\n- openstack-ansible-uw_apache-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4783984891224276adf8bbf3bca0be91 : SUCCESS in 41m 22s","accounts_in_message":[],"_revision_number":1},{"id":"cb4e1513dd68c38e35c00c60ee4fa723267d62cf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-26 01:25:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c2135e039f926041ed068759f4475367f99ff338","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-26 02:42:37.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b211f5c90941475fbfc6e50046c90d2a : SUCCESS in 21m 32s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/b0ea20b5ea844c86a2a811c06f5f0a8d : SUCCESS in 1h 08m 53s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/bac6026d01454f03a09749304d5d9829 : SUCCESS in 1h 14m 28s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/d4ee22d6f004491782b5f369d0c197f6 : SUCCESS in 1h 11m 44s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/22fa6f88e397499eaeb486e2e090c4ac : SUCCESS in 1h 15m 54s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5d70a16c30624ab88614d55b29c203f8 : SUCCESS in 1h 03m 10s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/71ac25e3820842b090e72d3263609723 : SUCCESS in 52m 18s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bdb232046d2641d3b1a9af4b669e5060 : SUCCESS in 53m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1e8680e34bf4bdd83c20228eb2eb7a1 : SUCCESS in 5m 32s\n- openstack-ansible-uw_apache-centos-8 https://zuul.opendev.org/t/openstack/build/464eb2b5b07542e890b36666bc7aa646 : SUCCESS in 40m 23s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/66587c092439413eb0ce5f562a189ca6 : SUCCESS in 35m 20s\n- openstack-ansible-uw_apache-ubuntu-focal https://zuul.opendev.org/t/openstack/build/257ff68a8ea446b0b9f9f3f521518505 : SUCCESS in 39m 50s","accounts_in_message":[],"_revision_number":1},{"id":"af9be137c179db9974eba7d18a3d949d9c323308","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-26 02:42:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"7a5f74faf9c2a5fa7eeae62272f8306320bbd39b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-26 02:45:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/216cf00ebf2a4f50a114f727028e6726 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"713a0576d882ff3f13da7d5ac3cefb9775ce90ed","revisions":{"713a0576d882ff3f13da7d5ac3cefb9775ce90ed":{"kind":"REWORK","_number":1,"created":"2021-05-21 12:53:01.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/10/792610/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/10/792610/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/10/792610/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/10/792610/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/10/792610/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/10/792610/1"}}},"commit":{"parents":[{"commit":"fc2bc15904de56d1abc01c0a647d802c1ba7e98b","subject":"Merge \"Add variables for rabbitmq ssl configuration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/fc2bc15904de56d1abc01c0a647d802c1ba7e98b"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2021-05-21 12:52:58.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2021-05-21 12:52:58.000000000","tz":180},"subject":"Replace linters test with integarted one","message":"Replace linters test with integarted one\n\nWe\u0027ve created integrated linters check job a while back and it\u0027s successfully\nworking for several releases. At the moment we experience difficulties\nwith future maintenance of the linters check from the openstack-ansible-tests\nrepo. So instead of fixing current one, we replace it with modern version of\nthe test.\n\n\nChange-Id: I22c4f987aedba049a953c955395b06803de37c27","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/713a0576d882ff3f13da7d5ac3cefb9775ce90ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/713a0576d882ff3f13da7d5ac3cefb9775ce90ed"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
