)]}'
{"id":"openstack%2Fopenstack-ansible-os_keystone~811402","triplet_id":"openstack%2Fopenstack-ansible-os_keystone~master~I9f46466943544168335b07198264adb8be49d261","project":"openstack/openstack-ansible-os_keystone","branch":"master","topic":"bug/1945274","hashtags":[],"change_id":"I9f46466943544168335b07198264adb8be49d261","subject":"Fix apache ProxyPass arguments","status":"MERGED","created":"2021-09-28 12:04:42.000000000","updated":"2021-11-12 21:08:52.000000000","submitted":"2021-11-12 21:06:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"811402-bug/1945274","meta_rev_id":"57490cd9991635acf1c774e5ebb91968a421ad93","_number":811402,"virtual_id_number":811402,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-12 21:06:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2021-10-06 08:30:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":2,"date":"2021-10-05 19:45:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2021-10-06 08:30:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}],"REVIEWER":[{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-28 15:02:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-28 17:00:45.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2021-10-05 19:45:03.000000000","updated_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"reviewer":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"state":"REVIEWER"},{"updated":"2021-10-06 08:30:03.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"0d641dcf3942ea7bae26809db2c36473c542e8ab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-09-28 12:04:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"136a7f62bd4ebf6816ff5561d3c0087531f2a0d0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-09-28 12:53:42.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"b48811e42df06d5e1ed2d23557020c4a86add305","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-28 15:02:33.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/27ce28e008334ee08aebedf2c1cfc4c2 : SUCCESS in 18m 39s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/99f8a22888394ea88b121af8111534da : FAILURE in 1h 04m 45s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/79865833846b40a2bdd5449d17521a00 : FAILURE in 1h 05m 24s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/c85db09bd32e4a748139615e47acb4c9 : FAILURE in 1h 11m 45s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/88e1292b097f435391b8e62cae6eb2de : FAILURE in 1h 21m 39s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dc0abb5cd88344d6add0445f02f2ab1a : FAILURE in 1h 13m 22s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/672af3ca68e84928869cb003d1bdc643 : FAILURE in 1h 09m 13s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/98e6c286b4f8444fba675c801d7cce6e : FAILURE in 1h 58m 33s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7a352b25be584e82902bf75dff26706b : FAILURE in 1h 35m 11s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/374cc0a6bb1e443cb60ea659307a0323 : FAILURE in 45m 50s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e38b73a28eac4980972ac3ad6e4588ae : FAILURE in 52m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/133aa4d78231454da456ede7f92f3bae : SUCCESS in 7m 04s\n- openstack-ansible-uw_apache-centos-8 https://zuul.opendev.org/t/openstack/build/de22020f64e5402ca6adc5fc2f368aaa : SUCCESS in 45m 17s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/face4c7e94924ed4a60ebe54ecacf60b : SUCCESS in 43m 14s\n- openstack-ansible-uw_apache-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7bf6d1582750415bb035cfb1f5996b04 : SUCCESS in 42m 47s","accounts_in_message":[],"_revision_number":2},{"id":"566d1e216eeb6993df8dd1b4ba84a7539c7902e5","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-09-28 17:00:45.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"003477d68cd00f25d724e409564e44b018f13c78","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-09-28 17:05:57.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"85f5058e6ef7a7f6d3b4aecaa1d3de8d151da4b7","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-09-28 17:06:34.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"df4bc0601cccbf2231aad53d87425cb7bb87ccd4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-28 19:27:49.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/badefb563f8e44698f829d08e584f0cd : SUCCESS in 12m 36s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/9bbfb46e9b0f40259b90d54b15993545 : FAILURE in 1h 04m 25s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/011f4b74b59e44568591c76f3107328a : FAILURE in 56m 22s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/d49993718f1a4d828f298c8bec2764eb : FAILURE in 1h 15m 35s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/4329a2ec9e4d43db98a9e63880ae4a22 : FAILURE in 1h 11m 56s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/948c75d377994c0185e2878f755cd694 : FAILURE in 1h 09m 31s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2db4e0a3350447f38a3c8585408c024e : FAILURE in 1h 02m 10s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/a8494a5f57a84bd2b04b0808d624bcc1 : FAILURE in 2h 10m 03s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a219ee651d0948f7a12b0d572e852538 : FAILURE in 1h 32m 17s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/cac0403e88c94bafa041689ea822945e : FAILURE in 42m 10s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4025ee68b10f43a5a6ff862abd9abaf2 : FAILURE in 46m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc2fcdd7464e4711b1b4fabfbdcab4c6 : SUCCESS in 9m 11s\n- openstack-ansible-uw_apache-centos-8 https://zuul.opendev.org/t/openstack/build/4848bf8c36a94272b7513bf0a624405f : SUCCESS in 44m 29s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ea708cbf8476472db0faf5a811b8d67d : SUCCESS in 41m 33s\n- openstack-ansible-uw_apache-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0e2adf39e5f04bc0b2849e95a8eb5c60 : SUCCESS in 44m 57s","accounts_in_message":[],"_revision_number":3},{"id":"361b9cb2854667a3eba2212d5a7e39991b6d380d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-09-30 14:11:40.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"0f3c6f10550ca785c919483fd76e79549d53c9a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-30 16:21:03.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/13157705d1ec4000a48884ddc18b2b19 : SUCCESS in 17m 31s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/1650bc812ac9464c83e2fc594bbca237 : SUCCESS in 1h 17m 01s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/7018a476023243b48f2c736d57b4900a : SUCCESS in 1h 09m 00s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/0a183606657d49819874f902872274c0 : SUCCESS in 1h 21m 55s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/6bd75311529945228d80a30553a5a8b3 : SUCCESS in 1h 20m 53s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b0c43d5f1aef4874a594cd2fe9d17fae : SUCCESS in 1h 18m 24s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c6e5a1f61dcf4dfd91700cdd92418600 : SUCCESS in 1h 10m 28s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/4bd78890af9d4e3886ed7840bc8333dd : SUCCESS in 2h 00m 39s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4a82fc61856b422ebe33415449bba0fa : SUCCESS in 1h 39m 57s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/9c76b1c14669447c893fa9b4a4d26d65 : FAILURE in 43m 01s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8c47ea5025564dba9f7ab10080f1d145 : SUCCESS in 56m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99b154fc5ba44475acffb4dee771d9a6 : SUCCESS in 7m 15s\n- openstack-ansible-uw_apache-centos-8 https://zuul.opendev.org/t/openstack/build/b944fcb9432f4e10bd5910acd8fb6c7d : SUCCESS in 44m 27s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/00b9177da68047aca7f9969ca363f218 : SUCCESS in 42m 13s\n- openstack-ansible-uw_apache-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f55b3bca2fbe4fb19c0fbed22876eb5b : SUCCESS in 42m 08s","accounts_in_message":[],"_revision_number":4},{"id":"f94788333ea8f4aa46df107589a5acafc1b65546","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2021-10-05 19:45:03.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"3d3286289410ce3b2129c6e6b0526d3fdb13d8cb","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2021-10-06 08:30:03.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"1ffe23c2ebe6184c47ba1616ca1ac44d0a7ba041","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-06 08:35:32.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"21424901a2ec30d0c4d3f9347b7281d3187c8e59","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-06 10:12:30.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6c0fb82d5e0d4595bf7d0545b0c65742 : SUCCESS in 15m 56s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/2ecd9055a7c941b8aa5a57792b330d30 : SUCCESS in 1h 20m 19s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/831ff052051e4c599af4cfe190e39ec8 : SUCCESS in 1h 15m 20s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/4c4cacbedeaa4f49981c86e280c26c06 : SUCCESS in 1h 34m 48s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/20f9fa7347394b8f9cbc77e27487a9a0 : SUCCESS in 1h 23m 19s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e15e708435564ed6bf23065b2117238a : SUCCESS in 1h 23m 39s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b7688af1a00d44ebab6bf0bf4f93ca4c : SUCCESS in 1h 13m 57s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f0b8a65c40c14187b42c057c485b366f : SUCCESS in 1h 04m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13a647b16539493db4c46125095ef155 : SUCCESS in 8m 24s\n- openstack-ansible-uw_apache-centos-8 https://zuul.opendev.org/t/openstack/build/e33e4966a4244ae1ab6d86445520c376 : FAILURE in 27m 08s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8fe00345f52c49819af4adfc7fdefe51 : FAILURE in 23m 51s\n- openstack-ansible-uw_apache-ubuntu-focal https://zuul.opendev.org/t/openstack/build/428117642d874bb498c6c6bb492e31fc : FAILURE in 23m 29s","accounts_in_message":[],"_revision_number":4},{"id":"284d428450032014d5339ba117b07a214c013347","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-10-06 12:00:30.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"f40565d0df91d40a2f72c82bae51d55d8e4d4435","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-06 14:11:01.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7a49ee9b64304ba99e8c57e0c42155c5 : SUCCESS in 15m 57s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/0962d0bbea414de6b517dd36a9ca4aa6 : SUCCESS in 1h 13m 48s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/ec56513bb1e546e8b65848459ea49fdd : SUCCESS in 1h 08m 46s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/4f917d2413bb4fe2b520ed95e09e5180 : SUCCESS in 1h 23m 58s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/c297282503484fafa174535d64848c47 : SUCCESS in 1h 29m 30s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/41bc65f73d084d47af5abd046bfa918e : SUCCESS in 1h 27m 52s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/115961ead3194e7c8676864425ae5675 : SUCCESS in 1h 20m 28s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/8781f9770ee7453f8c721e2edf5fa364 : SUCCESS in 1h 57m 54s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a5119131cc534b1e89c8f844bc218e30 : SUCCESS in 1h 47m 58s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/a4645da17e3c4d8cb552230ff0aa3baf : FAILURE in 45m 35s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bd332bc674744c59ab0cf57968da2c29 : SUCCESS in 1h 01m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b223d99d5474e28b2d16f39192749bb : SUCCESS in 7m 46s\n- openstack-ansible-uw_apache-centos-8 https://zuul.opendev.org/t/openstack/build/26026353aa6442d98d4a1dfa8bd2a738 : FAILURE in 29m 08s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/820369f23ba84c84b8f4b1c7cdf3d388 : FAILURE in 24m 59s\n- openstack-ansible-uw_apache-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3ef1cb9629024e6295455bf7d8a9e3ad : FAILURE in 26m 59s","accounts_in_message":[],"_revision_number":4},{"id":"412fae4c0c92f1cf0f82fa26f32d2b602a878553","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-11-12 17:19:39.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ac21cedd0d431a0b482f521a3a0ca91f01c9fb78","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-12 19:23:29.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ae27ea81ddf84b29858d5c6e4b625081 : SUCCESS in 15m 56s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/e1d3e44153094449a052211a33f14d48 : SUCCESS in 1h 24m 18s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/06af0685ba004ae39e43d4710d606e96 : SUCCESS in 1h 12m 46s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/c252fc6b77b5472b83da575eb9976407 : SUCCESS in 1h 18m 26s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/ea9a6b7b769043f99d863c69afc4ea3a : SUCCESS in 1h 20m 52s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/89dea6751fc14318892eae7a3154bfe2 : SUCCESS in 1h 31m 19s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8efb0b209a9a492aa5508e222e65eafa : SUCCESS in 1h 26m 20s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/1ff7cce5b4554988b95f6f4c349c8555 : SUCCESS in 1h 56m 35s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1ac7d4496302469c98befa11dcb5b143 : SUCCESS in 1h 50m 37s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/408992ff043b4eb7a154cbd1845f33b7 : FAILURE in 40m 13s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7d993e0cbf6348678bcb2f84285bf010 : SUCCESS in 56m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16f4446a3f7741b8b30aa80d9f781f3a : SUCCESS in 6m 34s\n- openstack-ansible-uw_apache-centos-8 https://zuul.opendev.org/t/openstack/build/9633f397b58b48e6966327ed4b682d0c : SUCCESS in 57m 21s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ce3bf277854045ac863580c53f084b6e : SUCCESS in 39m 58s\n- openstack-ansible-uw_apache-ubuntu-focal https://zuul.opendev.org/t/openstack/build/76d88f983ddb4421ba05af5c86460b6c : SUCCESS in 41m 24s","accounts_in_message":[],"_revision_number":4},{"id":"eb1410b328e8d4c35091299421c073e6a502bf08","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-12 19:32:33.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"6c8b279405d7fcfa9fcde549bd88a8575a810f2b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-12 21:06:49.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/480cd4a1cbf44dcb8ac44142b233218d : SUCCESS in 15m 19s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/a18ca9b5a1b244f889174441c756b438 : SUCCESS in 1h 29m 01s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/d25dcf7ae78241a5bdd082ee5e54b56f : SUCCESS in 1h 06m 22s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/48cc56b0635846589228dbf52ee9ee17 : SUCCESS in 1h 17m 57s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/193c770361df4bb3bd84f3b16f955a28 : SUCCESS in 1h 19m 30s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/69a62f352a6045749ad4013e4b8e4107 : SUCCESS in 1h 27m 51s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/72ec01dd7c204faea22dfb3a936571d6 : SUCCESS in 1h 20m 33s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1635aaffb9aa4ed2a19c24e34dbad380 : SUCCESS in 53m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c13624c826fb4bc5a51f9204ff869c06 : SUCCESS in 6m 38s\n- openstack-ansible-uw_apache-centos-8 https://zuul.opendev.org/t/openstack/build/564f33c308b94bd2be5b2780cefa47a0 : SUCCESS in 41m 09s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6677cde898814dbc9ca7bc217872cd8e : SUCCESS in 38m 39s\n- openstack-ansible-uw_apache-ubuntu-focal https://zuul.opendev.org/t/openstack/build/28098b562212487da10966f95ffa8043 : SUCCESS in 39m 36s","accounts_in_message":[],"_revision_number":4},{"id":"a80d52eb9707df7c7324b96eb68c1e64791d3aa8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-12 21:06:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"57490cd9991635acf1c774e5ebb91968a421ad93","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-12 21:08:52.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24835f0c2ffe4073a3bbf97c99b79db0 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"c2281d2f2b53cef219c8e9e393460001db3df64e","revisions":{"84064de9faf88ed69997303c231abe2d9c1f3bcd":{"kind":"REWORK","_number":1,"created":"2021-09-28 12:04:42.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/02/811402/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/02/811402/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/02/811402/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/02/811402/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/02/811402/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/02/811402/1"}}},"commit":{"parents":[{"commit":"eeef3a48f413a702a1c0eb8c4b40db9a5f9d85bb","subject":"Merge \"Fix shibboleth compatibility for ubuntu 18.04\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/eeef3a48f413a702a1c0eb8c4b40db9a5f9d85bb"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2021-09-28 12:04:33.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2021-09-28 12:04:33.000000000","tz":180},"subject":"Fix apache ProxyPass arguments","message":"Fix apache ProxyPass arguments\n\nWith elimination of CVE-2021-36160 apache stops interpreting trailing\nwhitespaces in ProxyPass arguments. This is directly not supported based\non the apache documentation [1] and should be fixed\n\n[1] https://httpd.apache.org/docs/trunk/mod/mod_proxy.html#proxypass\n\nChange-Id: I9f46466943544168335b07198264adb8be49d261\nRelated-Bug: #1945274\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/84064de9faf88ed69997303c231abe2d9c1f3bcd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/84064de9faf88ed69997303c231abe2d9c1f3bcd"}]},"branch":"refs/heads/master"},"da74c9ddefe73675dbfadf46ba4df5f96b2cb0e5":{"kind":"REWORK","_number":2,"created":"2021-09-28 12:53:42.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/02/811402/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/02/811402/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/02/811402/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/02/811402/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/02/811402/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/02/811402/2"}}},"commit":{"parents":[{"commit":"eeef3a48f413a702a1c0eb8c4b40db9a5f9d85bb","subject":"Merge \"Fix shibboleth compatibility for ubuntu 18.04\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/eeef3a48f413a702a1c0eb8c4b40db9a5f9d85bb"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2021-09-28 12:04:33.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-09-28 12:53:39.000000000","tz":0},"subject":"Fix apache ProxyPass arguments","message":"Fix apache ProxyPass arguments\n\nWith elimination of CVE-2021-36160 apache stops interpreting trailing\nwhitespaces in ProxyPass arguments. This is directly not supported based\non the apache documentation [1] and should be fixed\n\n[1] https://httpd.apache.org/docs/trunk/mod/mod_proxy.html#proxypass\n\nChange-Id: I9f46466943544168335b07198264adb8be49d261\nRelated-Bug: #1945274\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/da74c9ddefe73675dbfadf46ba4df5f96b2cb0e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/da74c9ddefe73675dbfadf46ba4df5f96b2cb0e5"}]},"branch":"refs/heads/master"},"a05a8caa2cfdf7fceb9a8f1ad5611eaf6f5b703f":{"kind":"REWORK","_number":3,"created":"2021-09-28 17:05:57.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/02/811402/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/02/811402/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/02/811402/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/02/811402/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/02/811402/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/02/811402/3"}}},"commit":{"parents":[{"commit":"eeef3a48f413a702a1c0eb8c4b40db9a5f9d85bb","subject":"Merge \"Fix shibboleth compatibility for ubuntu 18.04\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/eeef3a48f413a702a1c0eb8c4b40db9a5f9d85bb"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2021-09-28 12:04:33.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-09-28 17:05:46.000000000","tz":0},"subject":"Fix apache ProxyPass arguments","message":"Fix apache ProxyPass arguments\n\nWith elimination of CVE-2021-36160 apache stops interpreting trailing\nwhitespaces in ProxyPass arguments. This is directly not supported based\non the apache documentation [1] and should be fixed\n\n[1] https://httpd.apache.org/docs/trunk/mod/mod_proxy.html#proxypass\n\nChange-Id: I9f46466943544168335b07198264adb8be49d261\nRelated-Bug: #1945274\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/a05a8caa2cfdf7fceb9a8f1ad5611eaf6f5b703f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/a05a8caa2cfdf7fceb9a8f1ad5611eaf6f5b703f"}]},"branch":"refs/heads/master"},"c2281d2f2b53cef219c8e9e393460001db3df64e":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2021-09-30 14:11:40.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/02/811402/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/02/811402/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/02/811402/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/02/811402/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/02/811402/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/02/811402/4"}}},"commit":{"parents":[{"commit":"eeef3a48f413a702a1c0eb8c4b40db9a5f9d85bb","subject":"Merge \"Fix shibboleth compatibility for ubuntu 18.04\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/eeef3a48f413a702a1c0eb8c4b40db9a5f9d85bb"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2021-09-28 12:04:33.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-09-30 14:11:37.000000000","tz":0},"subject":"Fix apache ProxyPass arguments","message":"Fix apache ProxyPass arguments\n\nWith elimination of CVE-2021-36160 apache stops interpreting trailing\nwhitespaces in ProxyPass arguments. This is directly not supported based\non the apache documentation [1] and should be fixed\n\n[1] https://httpd.apache.org/docs/trunk/mod/mod_proxy.html#proxypass\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/811742\nChange-Id: I9f46466943544168335b07198264adb8be49d261\nRelated-Bug: #1945274\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/c2281d2f2b53cef219c8e9e393460001db3df64e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/c2281d2f2b53cef219c8e9e393460001db3df64e"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
