)]}'
{"id":"openstack%2Fopenstack-ansible-os_keystone~812768","triplet_id":"openstack%2Fopenstack-ansible-os_keystone~master~Ic8e867b5263c60c47cf513adc574a018d1dd3811","project":"openstack/openstack-ansible-os_keystone","branch":"master","topic":"wsoyinka-patch1","attention_set":{},"removed_from_attention_set":{"32197":{"account":{"_account_id":32197,"name":"ws","email":"wsoyinka@gmail.com","username":"soyinka"},"last_update":"2022-09-06 08:18:02.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ic8e867b5263c60c47cf513adc574a018d1dd3811","subject":"Add vars file for Rocky Linux distribution","status":"ABANDONED","created":"2021-10-07 02:31:08.000000000","updated":"2022-09-06 08:18:02.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"4a71364a66f591203a88d0c3abe05b034370346e","_number":812768,"virtual_id_number":812768,"owner":{"_account_id":32197,"name":"ws","email":"wsoyinka@gmail.com","username":"soyinka"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-10-07 04:33:45.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"all":[{"value":-1,"date":"2021-10-19 09:19:09.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"all":[{"value":-1,"date":"2021-10-19 09:19:09.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-07 04:33:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-07 08:30:27.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"CC"},{"updated":"2021-10-19 09:19:09.000000000","updated_by":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"reviewer":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"state":"REVIEWER"}],"messages":[{"id":"551104fd332729b06035c6cb26cf192e0a2a585b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32197,"name":"ws","email":"wsoyinka@gmail.com","username":"soyinka"},"date":"2021-10-07 02:31:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9001875843dc08e2d1b29f93e7cc7b8139ad4dfe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-07 04:33:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6fd68351ba17402695652064d3b99bb0 : SUCCESS in 14m 35s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/11e026bea0634847a14e7cb3f45af3a4 : SUCCESS in 1h 21m 11s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/f425f2b7238f4648849ff8aca8555191 : SUCCESS in 1h 12m 11s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/b20fc0ca0b6f470ebf59308e55c19645 : SUCCESS in 1h 23m 09s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/227978d2f12940cb971d063db93b129c : SUCCESS in 1h 27m 26s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d0b7633dd777423ab3b1e0adf4ba5745 : SUCCESS in 1h 21m 54s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/16febca23d4b49368fb5cd87daba2db0 : SUCCESS in 1h 14m 19s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/e316e8a115ee47f0bbfdd4f1b5dc5d67 : SUCCESS in 1h 58m 59s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/99551e6145d14c058964d304d5813676 : SUCCESS in 1h 42m 08s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/32a93c2b0abb495cbdebe947bc879281 : FAILURE in 43m 23s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3f4f9a89562947669336cb0a4b2050aa : SUCCESS in 59m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b881630e5b5b4bf08552bfd442528bde : SUCCESS in 6m 28s\n- openstack-ansible-uw_apache-centos-8 https://zuul.opendev.org/t/openstack/build/7c3cdcb60904443cbd8aa8a8505e9036 : FAILURE in 39m 53s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/326ddac23fe44d2f8e36b7a437b43a1b : FAILURE in 22m 28s\n- openstack-ansible-uw_apache-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ce07fa5969d3478e96aa0b78f474a22b : FAILURE in 24m 08s","accounts_in_message":[],"_revision_number":1},{"id":"c26c76ad03ace8f66cd7516ea41353e4e830cfd6","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-10-07 08:30:27.000000000","message":"Patch Set 1:\n\nI guess this is smth we should discuss on the nearest PTG. Because right now it feels hard to maintain support of rocky linux considering that neither of really active contributors are currently running systems on it. Also I guess there\u0027re no opendev CI images for it? And adding without CI means it will be broken pretty soon.\n\nAlso I wonder if just symlink will work here as well?","accounts_in_message":[],"_revision_number":1},{"id":"9ec33fc3331558618e1c5ba4611754ad953bed9e","author":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"date":"2021-10-19 09:19:09.000000000","message":"Patch Set 1: Code-Review-1 Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4a71364a66f591203a88d0c3abe05b034370346e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"date":"2022-09-06 08:18:02.000000000","message":"Abandoned\n\nMore recent ansible releases fix this issue. No need to introduce technical debt.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f7fdfdae2cda8ca61592423b47af37ffdbbff18f","revisions":{"f7fdfdae2cda8ca61592423b47af37ffdbbff18f":{"kind":"REWORK","_number":1,"created":"2021-10-07 02:31:08.000000000","uploader":{"_account_id":32197,"name":"ws","email":"wsoyinka@gmail.com","username":"soyinka"},"ref":"refs/changes/68/812768/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/68/812768/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/68/812768/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/68/812768/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/68/812768/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/68/812768/1"}}},"commit":{"parents":[{"commit":"eeef3a48f413a702a1c0eb8c4b40db9a5f9d85bb","subject":"Merge \"Fix shibboleth compatibility for ubuntu 18.04\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/eeef3a48f413a702a1c0eb8c4b40db9a5f9d85bb"}]}],"author":{"name":"wsoyinka","email":"wsoyinka@gmail.com","date":"2021-10-07 02:11:35.000000000","tz":-240},"committer":{"name":"wsoyinka","email":"wsoyinka@gmail.com","date":"2021-10-07 02:11:35.000000000","tz":-240},"subject":"Add vars file for Rocky Linux distribution","message":"Add vars file for Rocky Linux distribution\n\nRocky Linux is a community enterprise operating system that is compatible\nwith Redhat enterprise Linux. This vars file begins the work for\nadding support for the Rocky distro to the openstack ansible project.\n\nChange-Id: Ic8e867b5263c60c47cf513adc574a018d1dd3811\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/f7fdfdae2cda8ca61592423b47af37ffdbbff18f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/f7fdfdae2cda8ca61592423b47af37ffdbbff18f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
