)]}'
{"id":"openstack%2Fopenstack-ansible-os_keystone~817390","triplet_id":"openstack%2Fopenstack-ansible-os_keystone~master~Ib3f90a72dfc8f78cf304b0f130883befdeb09220","project":"openstack/openstack-ansible-os_keystone","branch":"master","topic":"keystone_apache_only","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2021-11-26 15:14:18.000000000","reason":"Change was submitted"},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2021-11-24 17:02:27.000000000","reason":"Jonathan Rosser replied on the change"}},"hashtags":[],"change_id":"Ib3f90a72dfc8f78cf304b0f130883befdeb09220","subject":"Drop Nginx webserver support","status":"MERGED","created":"2021-11-10 11:54:23.000000000","updated":"2021-11-26 15:15:58.000000000","submitted":"2021-11-26 15:14:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"817390-keystone_apache_only","meta_rev_id":"c7f28941691b2b835898281a6cc0c745edcd9b0b","_number":817390,"virtual_id_number":817390,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-26 15:14:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":2,"date":"2021-11-26 13:21:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":1,"date":"2021-11-25 09:28:11.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-11-24 17:02:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":1,"date":"2021-11-26 13:21:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-10 13:27:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-12 18:15:28.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"CC"},{"updated":"2021-11-24 17:02:27.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2021-11-25 09:28:11.000000000","updated_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"reviewer":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"state":"REVIEWER"},{"updated":"2021-11-26 13:21:58.000000000","updated_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"reviewer":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"state":"REVIEWER"}],"messages":[{"id":"1be16f3c19c446cc7cb9b9e5546d902d36362e09","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-11-10 11:54:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d4a2c22ad74dfe77bead6424e88903b1453902a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-10 13:27:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9dbd273a8ebb4b4d96abe57e5cc8a4de : SUCCESS in 22m 14s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/c249faa763364acf836a300966c12746 : FAILURE in 51m 24s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/6d6f977f65ec4c68b894cfd204022b14 : FAILURE in 28m 49s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/b562464b57574f279c06d8aee5361efc : FAILURE in 35m 47s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/7a8f8f75408d424491dfa852e5395554 : FAILURE in 48m 28s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0faab50b47034a4e86728520d278635c : FAILURE in 38m 37s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7eb48a069cce4f70994e167f7e6c4ded : FAILURE in 36m 58s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/0949309a61e64e5d850b624da7e670c7 : FAILURE in 1h 24m 44s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d8e89a3d1eb1414d9845f3176f671c4b : FAILURE in 1h 23m 54s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/62147d92eca941819e27f7d9287507f4 : FAILURE in 34m 51s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8ba9b129133742f58e2c3cfa8837004d : FAILURE in 34m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99da4ea021174b56bec1369185dd0287 : SUCCESS in 5m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/21839ecc3a9d450ea658fc8272a9b6f5 : SUCCESS in 4m 26s\n- openstack-ansible-uw_apache-centos-8 https://zuul.opendev.org/t/openstack/build/b5c0ba94b23a4ca5a021000af752214c : FAILURE in 26m 39s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b80e323a0ebd40c2b4a6d37b2c1d29d5 : FAILURE in 20m 46s\n- openstack-ansible-uw_apache-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e0972aa63acd40f3891ce0ca16f66bff : FAILURE in 24m 40s","accounts_in_message":[],"_revision_number":1},{"id":"c43b48ac6382e648aa07233c5edca0d058d11986","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-11-12 17:25:39.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"88072930095866606da037ece3d573ba78114774","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-11-12 18:15:28.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"88e2e5c18d279b645b6a3c61d419a4c1d179cc33","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-12 19:04:58.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9ca7926291a342b58a9d2fe827b63004 : SUCCESS in 16m 29s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/1832f7d009e14da2b11309681daf7cd3 : FAILURE in 38m 32s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/56e4418a32094ec084e1e12b72b88030 : FAILURE in 38m 38s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/ec2abdc2134449448a42cb269b2d66e9 : FAILURE in 47m 55s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/2d3e9b3d3d844b628e7005087e532f8b : FAILURE in 42m 53s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3afcb6f1827947aa8fc268cc073e5b05 : FAILURE in 44m 42s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4c16b3d55eb24486a5cf12c1c6b173e9 : FAILURE in 41m 33s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/703d158e2a914843937161f211d903cd : FAILURE in 1h 20m 35s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/af7e8c8324ca4f8eb7f61499fc431ea5 : FAILURE in 1h 14m 13s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/9d3b612914ab4f38b46c4d8941b39cae : FAILURE in 34m 16s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e4226e1dddec43a38988622ebeb997a0 : FAILURE in 32m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0fe5db84c7674dfea31606a15da2bc70 : SUCCESS in 7m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2ac8786893af4d91b2386740d6c7e77e : SUCCESS in 4m 35s\n- openstack-ansible-uw_apache-centos-8 https://zuul.opendev.org/t/openstack/build/e23e5255b03e494b8d83e7452c83ab14 : SUCCESS in 1h 00m 21s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a954f07c36214ea99fd5076c6fc0962c : SUCCESS in 43m 51s\n- openstack-ansible-uw_apache-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c13f06db85f74f2b98d4731dd1a6bf02 : SUCCESS in 38m 53s","accounts_in_message":[],"_revision_number":2},{"id":"c0592dc60018d7e35b3d56578536a1d375a8be99","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-11-16 15:12:25.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d065c105895df6140432f45161dfb6929488d78f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-11-19 13:04:29.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"c7170314642bf1b2d9fa642ea31a4c3a37ccf0d1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-11-19 13:10:37.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"254e90ee42dd9960c73c472f4b77cf6e2b64c4bd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-11-19 13:13:53.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"7de5faa473e4ac3f3f2c816ead4490bc932c5319","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-19 15:08:00.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/57d1b1c7ed6042bfb2b9f2eacfcf8ae6 : SUCCESS in 15m 35s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/82fdef086e5c463e9a0df497f38b9281 : FAILURE in 55m 32s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/9f977d03af6345cb8bd25c7972331653 : FAILURE in 44m 41s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/79337d0556fe4669ba123842d43e27fd : FAILURE in 56m 17s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/327d0bc06b92405dbdc53a612a78f303 : FAILURE in 48m 21s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b94782a533e14ff1ba2a1cd08e827b81 : SUCCESS in 1h 44m 41s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5b49837dbf764509b75483378d55ae67 : FAILURE in 50m 07s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/7e0877eac7bb4b65af62776f55c4cc19 : FAILURE in 1h 38m 31s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/09806ec542fc4d65baaaa0089792aff7 : FAILURE in 1h 20m 32s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/bbdf64423bc4455c9254b479bcbc6887 : FAILURE in 40m 59s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a487286bf1714a149986c5c55db4dad9 : FAILURE in 43m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9e8bdb902314b31a7f3ef19b53203b3 : SUCCESS in 7m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3c3569e274364d7795c9ed0234c3feaa : SUCCESS in 3m 40s\n- openstack-ansible-uw_apache-centos-8 https://zuul.opendev.org/t/openstack/build/5d71fd7382024f7f961083841cad81f9 : FAILURE in 13m 26s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/262f0a0482ca4ac0bf11d1f054426688 : SUCCESS in 39m 53s\n- openstack-ansible-uw_apache-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0e081a0a33df454da0aadd809e952cf9 : SUCCESS in 47m 26s","accounts_in_message":[],"_revision_number":5},{"id":"3b17080e3ace79908cf5dbde01436ea74a307635","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-11-19 15:52:31.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"f1e833d7d7b149efcd353fa81eb8255e93284cad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-11-19 15:58:27.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"26131b1a3ed9302dcc8a1d03ceb7ec6794f64ffa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-19 17:40:36.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3bdc0ca278aa41a2bd44789c8f57fa1c : SUCCESS in 15m 53s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/89e6e2ea60954800b029a89270110110 : FAILURE in 37m 04s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/d114182c0f7047a9902585ebc860912a : FAILURE in 57m 11s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/89d901faf0e64aecb2b501a1ba50d352 : FAILURE in 49m 10s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/2d09f12d8e5749e19b8331984f959e75 : FAILURE in 45m 58s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d1385e2a06ef4898b4b0f70c518f5798 : SUCCESS in 1h 34m 12s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3d0f7d933a994ccbadf550f7c3005e35 : FAILURE in 1h 09m 15s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/2e777958de584ca988709e40c9dd1d65 : FAILURE in 1h 20m 34s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b3fab13b4a4f497d84bc49f909d828ab : FAILURE in 1h 10m 12s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/43deb75614994306a08bd181b1a8c153 : FAILURE in 35m 07s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/77b38697188a41b7a940b1d441eb224e : FAILURE in 35m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d6b1ec6b6ea45bdaf98f137f95d354a : SUCCESS in 5m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bd287c236ae94012ad6fff975003a796 : SUCCESS in 4m 15s\n- openstack-ansible-uw_apache-centos-8 https://zuul.opendev.org/t/openstack/build/48b0d66b64d64d0cbe708509bb5651b1 : SUCCESS in 38m 07s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dda1373060344186a6f207829188452c : SUCCESS in 36m 55s\n- openstack-ansible-uw_apache-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9982233cbba8419a853b020e3a3d37d9 : SUCCESS in 39m 48s","accounts_in_message":[],"_revision_number":7},{"id":"3587ba2d8e5af9298a956278dc9df4c2f2685c2b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-11-20 17:35:24.000000000","message":"Patch Set 8: Commit message was updated.","accounts_in_message":[],"_revision_number":8},{"id":"f0138fbd1248aff015dc24a2e6a9c7766219fa4b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-20 19:35:40.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2352009359a847ab8071dc5e51e815f6 : SUCCESS in 19m 01s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/727df921bf4e400387bbec9bc6736612 : SUCCESS in 1h 22m 55s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/1598c7aa129b4b4e99c6719b3d2d8505 : SUCCESS in 1h 12m 37s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/47d8c49045174a618c99d4ec3c3bb8dc : SUCCESS in 1h 23m 17s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/1a4ba3eb6b9a42b391e19e2f41ee3937 : SUCCESS in 1h 55m 58s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/eb3cd7bf09b14d8b8881f6ed1807e5f6 : SUCCESS in 1h 36m 32s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5c5e5cfcfab94d7fb7a5155a8071fd6a : SUCCESS in 1h 33m 59s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/baf8f8d0e5a842fc93da7ab017ad7939 : FAILURE in 1h 22m 18s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c672d993ff094860a3521fe6a8130246 : FAILURE in 1h 36m 37s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/457ac13f18f443a2a3adc660000a87b2 : FAILURE in 43m 29s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/742ec8e43f5544bd8dbf97d38e129de2 : SUCCESS in 44m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3081c4ce7214c72a88ca16c0144f107 : SUCCESS in 10m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cf546f104c3a431ba83d393bc6edab57 : SUCCESS in 3m 32s\n- openstack-ansible-uw_apache-centos-8 https://zuul.opendev.org/t/openstack/build/80d4177304ac45fa850b9f33a99a81f1 : SUCCESS in 43m 29s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e0df33596dbf4e95ac551350fc792dce : SUCCESS in 43m 57s\n- openstack-ansible-uw_apache-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3dec12ddced34b7788ab45f5c085e417 : SUCCESS in 39m 27s","accounts_in_message":[],"_revision_number":8},{"id":"e9a607f354ad1c41b2a874bc92aa551928da6427","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-11-22 10:45:43.000000000","message":"Patch Set 9: Published edit on patch set 8.","accounts_in_message":[],"_revision_number":9},{"id":"e91ddd8932ce1c97446e49ef93baa2da5f90b7e0","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-11-22 12:28:31.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"b181251397cc18e685286a46a20feb151c286aba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-22 12:49:07.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d41a9001b8ad4b96bf3c024c12335a4e : SUCCESS in 23m 48s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/968a2fb1fab24501ae363401f493176f : SUCCESS in 1h 24m 11s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/5581974fcd8c41d89b476b19321caf9c : SUCCESS in 1h 04m 24s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/96edb62ed46547edbcffe921367abf04 : SUCCESS in 1h 05m 56s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/a3a77a1e5d534ec88fc0737956aa7362 : SUCCESS in 1h 23m 16s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6dd0e1435c2746d7bb761a4749c6eadb : SUCCESS in 1h 28m 42s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/56efa590d6c249abbd117b26b975ae0f : SUCCESS in 1h 55m 33s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/8ebf6d23c1ba41838beb1aaa467b109a : FAILURE in 1h 22m 22s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c59708d7992c4575b1ef9911bc14907e : FAILURE in 1h 58m 38s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/aab0409133dd41d68fddd9a9dd855ed2 : FAILURE in 41m 42s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f4867bc0657449089002cb0b0880184a : SUCCESS in 57m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e10c962a500434eaeb17c578c547fdb : SUCCESS in 6m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dbd7c81e02ed400d81216af28743dc29 : SUCCESS in 7m 17s\n- openstack-ansible-uw_apache-centos-8 https://zuul.opendev.org/t/openstack/build/da4969559c234f899657e8e83dcab246 : SUCCESS in 48m 27s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/763b9f5b19eb4317a68ff1d9e68520a5 : SUCCESS in 33m 16s\n- openstack-ansible-uw_apache-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fb5b033ce3ba4d13b5f81e44366785c1 : SUCCESS in 44m 41s","accounts_in_message":[],"_revision_number":9},{"id":"4ae927d520d3bd7ff2e444c1d4c90707886280af","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-11-23 18:38:48.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"136ef2b577284a0d3e4568dbec799fa854a40da1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-23 20:56:04.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/03c259500a6141b494f35f88b98aa3af : SUCCESS in 16m 29s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/198a7b3270234c86a1b2a89af3ab52c8 : SUCCESS in 1h 34m 22s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/0f29528aecb140d7945e2c2bf7addb0e : SUCCESS in 1h 16m 44s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/5ceaa6cab32341578aa61101fb4611cf : SUCCESS in 1h 35m 30s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/a8d76dbb70db4e00bcda03f63effc125 : SUCCESS in 1h 34m 27s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dc3f528f54eb4df182b8f72a57845d09 : SUCCESS in 1h 29m 28s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b4aac12b90194578b5b6eab7e056a99d : SUCCESS in 1h 24m 55s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/901c15ae37494d5ba941d3c49ba743a1 : SUCCESS in 1h 41m 20s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/beb5f07419314777ac223bda34c1120d : SUCCESS in 2h 06m 05s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/65e2c9dd8d274c2abf1b96ed60850cde : FAILURE in 46m 29s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/08ec520abc52465ab0f36e1e59c771af : SUCCESS in 55m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f4364c11d79412287e483031439fe6c : SUCCESS in 5m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4e8a2049ed3a41efa451d09ae43c6e48 : SUCCESS in 3m 35s\n- openstack-ansible-uw_apache-centos-8 https://zuul.opendev.org/t/openstack/build/8cb2fa2afd3949cca830606a00ade720 : SUCCESS in 41m 25s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/18a28a6939664e1c91523e69341bad2e : SUCCESS in 38m 01s\n- openstack-ansible-uw_apache-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c67c072e481e4a7da9827ea320a502d9 : SUCCESS in 41m 23s","accounts_in_message":[],"_revision_number":9},{"id":"583ed7c934dd68fa1a65e2b4518684c6f4bc66ac","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-11-24 17:02:27.000000000","message":"Patch Set 9: Code-Review+2","accounts_in_message":[],"_revision_number":9},{"id":"7f741c255dd203458475888a3cf1b7d611426e73","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2021-11-25 09:28:11.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"c1445caa8769753988ed1b9ebc66995fc78b2ab3","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2021-11-26 13:21:58.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"0e91fb7a1eeb0af7fad2fe695cf19838da72fae9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-26 13:25:52.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"6d8e62d3463c7f98eed9c2b43bf21424fc21ced3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-26 15:14:18.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/07817e68e5ea41bebd28c0d099b6febb : SUCCESS in 30m 51s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/6ae1162ee8b94e8eb8328012a682f5a3 : SUCCESS in 1h 29m 26s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/119114f23fb54336996e95fc1dcba7a4 : SUCCESS in 1h 13m 44s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/54badf1fa35f4d30bac1902d25c778f1 : SUCCESS in 1h 22m 13s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/36e98a194d164f60a00c8d508549da61 : SUCCESS in 1h 22m 55s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c3aa26e46a124e3d91e0160e1230490c : SUCCESS in 1h 34m 15s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d502fe7d7ca9421abdab064e5eac1074 : SUCCESS in 1h 42m 14s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ee465ba1769f4d14a28495918eb64fb4 : SUCCESS in 1h 02m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55949bdf84764e22b79d836440e64151 : SUCCESS in 5m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fecfbc332c2245b296a649cb5e4e6ab5 : SUCCESS in 3m 36s\n- openstack-ansible-uw_apache-centos-8 https://zuul.opendev.org/t/openstack/build/5004c44153a2459ab98b418c4435ac3f : SUCCESS in 39m 25s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c6063bf8d00a486fb058650fdf1b7d72 : SUCCESS in 37m 19s\n- openstack-ansible-uw_apache-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4af603f20c484db4a93ce843a4e21f3f : SUCCESS in 44m 48s","accounts_in_message":[],"_revision_number":9},{"id":"271c79cd01bed700af4935a3d451ad8f323cfbd9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-26 15:14:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":9},{"id":"c7f28941691b2b835898281a6cc0c745edcd9b0b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-26 15:15:58.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4df2a54a0345420cb8ca915983cca9c4 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"eb9a0c6ceaae98fe992bab1f9b5be797ba8a99b4","revisions":{"56e263f4970718e62a7fdc7b62c4738b5c44c57a":{"kind":"REWORK","_number":1,"created":"2021-11-10 11:54:23.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/90/817390/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/90/817390/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/90/817390/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/90/817390/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/90/817390/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/90/817390/1"}}},"commit":{"parents":[{"commit":"eeef3a48f413a702a1c0eb8c4b40db9a5f9d85bb","subject":"Merge \"Fix shibboleth compatibility for ubuntu 18.04\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/eeef3a48f413a702a1c0eb8c4b40db9a5f9d85bb"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2021-11-10 11:54:03.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2021-11-10 11:54:15.000000000","tz":120},"subject":"Drop Nginx webserver support","message":"Drop Nginx webserver support\n\nThere is no reason to support multiple web servers as a proxy for\nkeystone. Nginx is missing modules to support federation. With it\u0027s\nremoval we simplify code and reduce maintaining efforts needed.\n\nChange-Id: Ib3f90a72dfc8f78cf304b0f130883befdeb09220\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/56e263f4970718e62a7fdc7b62c4738b5c44c57a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/56e263f4970718e62a7fdc7b62c4738b5c44c57a"}]},"branch":"refs/heads/master"},"7897677d6a44546800cba57a72f1ed8661dcfade":{"kind":"REWORK","_number":2,"created":"2021-11-12 17:25:39.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/90/817390/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/90/817390/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/90/817390/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/90/817390/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/90/817390/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/90/817390/2"}}},"commit":{"parents":[{"commit":"eeef3a48f413a702a1c0eb8c4b40db9a5f9d85bb","subject":"Merge \"Fix shibboleth compatibility for ubuntu 18.04\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/eeef3a48f413a702a1c0eb8c4b40db9a5f9d85bb"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2021-11-10 11:54:03.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-11-12 17:25:39.000000000","tz":0},"subject":"Drop Nginx webserver support","message":"Drop Nginx webserver support\n\nThere is no reason to support multiple web servers as a proxy for\nkeystone. Nginx is missing modules to support federation. With it\u0027s\nremoval we simplify code and reduce maintaining efforts needed.\n\nChange-Id: Ib3f90a72dfc8f78cf304b0f130883befdeb09220\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/7897677d6a44546800cba57a72f1ed8661dcfade"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/7897677d6a44546800cba57a72f1ed8661dcfade"}]},"branch":"refs/heads/master"},"28d53d456ed648783c0505d97581a2d5670d8a7c":{"kind":"REWORK","_number":3,"created":"2021-11-19 13:04:29.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/90/817390/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/90/817390/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/90/817390/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/90/817390/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/90/817390/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/90/817390/3"}}},"commit":{"parents":[{"commit":"eeef3a48f413a702a1c0eb8c4b40db9a5f9d85bb","subject":"Merge \"Fix shibboleth compatibility for ubuntu 18.04\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/eeef3a48f413a702a1c0eb8c4b40db9a5f9d85bb"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2021-11-10 11:54:03.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-11-19 13:04:29.000000000","tz":0},"subject":"Drop Nginx webserver support","message":"Drop Nginx webserver support\n\nThere is no reason to support multiple web servers as a proxy for\nkeystone. Nginx is missing modules to support federation. With it\u0027s\nremoval we simplify code and reduce maintaining efforts needed.\n\nChange-Id: Ib3f90a72dfc8f78cf304b0f130883befdeb09220\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/28d53d456ed648783c0505d97581a2d5670d8a7c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/28d53d456ed648783c0505d97581a2d5670d8a7c"}]},"branch":"refs/heads/master"},"088171166a732efe357cb34e59b5dfdf976280ab":{"kind":"REWORK","_number":4,"created":"2021-11-19 13:10:37.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/90/817390/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/90/817390/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/90/817390/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/90/817390/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/90/817390/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/90/817390/4"}}},"commit":{"parents":[{"commit":"eeef3a48f413a702a1c0eb8c4b40db9a5f9d85bb","subject":"Merge \"Fix shibboleth compatibility for ubuntu 18.04\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/eeef3a48f413a702a1c0eb8c4b40db9a5f9d85bb"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2021-11-10 11:54:03.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-11-19 13:10:37.000000000","tz":0},"subject":"Drop Nginx webserver support","message":"Drop Nginx webserver support\n\nThere is no reason to support multiple web servers as a proxy for\nkeystone. Nginx is missing modules to support federation. With it\u0027s\nremoval we simplify code and reduce maintaining efforts needed.\n\nChange-Id: Ib3f90a72dfc8f78cf304b0f130883befdeb09220\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/088171166a732efe357cb34e59b5dfdf976280ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/088171166a732efe357cb34e59b5dfdf976280ab"}]},"branch":"refs/heads/master"},"20103ab5c78cad3871ddaec877737d91e5cfa084":{"kind":"REWORK","_number":5,"created":"2021-11-19 13:13:53.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/90/817390/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/90/817390/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/90/817390/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/90/817390/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/90/817390/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/90/817390/5"}}},"commit":{"parents":[{"commit":"eeef3a48f413a702a1c0eb8c4b40db9a5f9d85bb","subject":"Merge \"Fix shibboleth compatibility for ubuntu 18.04\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/eeef3a48f413a702a1c0eb8c4b40db9a5f9d85bb"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2021-11-10 11:54:03.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-11-19 13:13:53.000000000","tz":0},"subject":"Drop Nginx webserver support","message":"Drop Nginx webserver support\n\nThere is no reason to support multiple web servers as a proxy for\nkeystone. Nginx is missing modules to support federation. With it\u0027s\nremoval we simplify code and reduce maintaining efforts needed.\n\nChange-Id: Ib3f90a72dfc8f78cf304b0f130883befdeb09220\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/20103ab5c78cad3871ddaec877737d91e5cfa084"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/20103ab5c78cad3871ddaec877737d91e5cfa084"}]},"branch":"refs/heads/master"},"c0cadb0584108761ea90aded80a474cf08359581":{"kind":"REWORK","_number":6,"created":"2021-11-19 15:52:31.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/90/817390/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/90/817390/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/90/817390/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/90/817390/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/90/817390/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/90/817390/6"}}},"commit":{"parents":[{"commit":"eeef3a48f413a702a1c0eb8c4b40db9a5f9d85bb","subject":"Merge \"Fix shibboleth compatibility for ubuntu 18.04\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/eeef3a48f413a702a1c0eb8c4b40db9a5f9d85bb"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2021-11-10 11:54:03.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2021-11-19 15:52:19.000000000","tz":120},"subject":"Drop Nginx webserver support","message":"Drop Nginx webserver support\n\nThere is no reason to support multiple web servers as a proxy for\nkeystone. Nginx is missing modules to support federation. With it\u0027s\nremoval we simplify code and reduce maintaining efforts needed.\n\nChange-Id: Ib3f90a72dfc8f78cf304b0f130883befdeb09220\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/c0cadb0584108761ea90aded80a474cf08359581"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/c0cadb0584108761ea90aded80a474cf08359581"}]},"branch":"refs/heads/master"},"ac214a3419adc616163a86b5d6c697ea5722177c":{"kind":"REWORK","_number":7,"created":"2021-11-19 15:58:27.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/90/817390/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/90/817390/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/90/817390/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/90/817390/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/90/817390/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/90/817390/7"}}},"commit":{"parents":[{"commit":"eeef3a48f413a702a1c0eb8c4b40db9a5f9d85bb","subject":"Merge \"Fix shibboleth compatibility for ubuntu 18.04\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/eeef3a48f413a702a1c0eb8c4b40db9a5f9d85bb"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2021-11-10 11:54:03.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2021-11-19 15:58:18.000000000","tz":120},"subject":"Drop Nginx webserver support","message":"Drop Nginx webserver support\n\nThere is no reason to support multiple web servers as a proxy for\nkeystone. Nginx is missing modules to support federation. With it\u0027s\nremoval we simplify code and reduce maintaining efforts needed.\n\nChange-Id: Ib3f90a72dfc8f78cf304b0f130883befdeb09220\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/ac214a3419adc616163a86b5d6c697ea5722177c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/ac214a3419adc616163a86b5d6c697ea5722177c"}]},"branch":"refs/heads/master"},"a4750e3b6ba2943406d3c5b35f4336c68c2e2976":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2021-11-20 17:35:24.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/90/817390/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/90/817390/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/90/817390/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/90/817390/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/90/817390/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/90/817390/8"}}},"commit":{"parents":[{"commit":"eeef3a48f413a702a1c0eb8c4b40db9a5f9d85bb","subject":"Merge \"Fix shibboleth compatibility for ubuntu 18.04\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/eeef3a48f413a702a1c0eb8c4b40db9a5f9d85bb"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2021-11-10 11:54:03.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-11-20 17:35:23.000000000","tz":0},"subject":"Drop Nginx webserver support","message":"Drop Nginx webserver support\n\nThere is no reason to support multiple web servers as a proxy for\nkeystone. Nginx is missing modules to support federation. With it\u0027s\nremoval we simplify code and reduce maintaining efforts needed.\n\nChange-Id: Ib3f90a72dfc8f78cf304b0f130883befdeb09220\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/818674\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/a4750e3b6ba2943406d3c5b35f4336c68c2e2976"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/a4750e3b6ba2943406d3c5b35f4336c68c2e2976"}]},"branch":"refs/heads/master","description":"Edit commit message"},"eb9a0c6ceaae98fe992bab1f9b5be797ba8a99b4":{"kind":"REWORK","_number":9,"created":"2021-11-22 10:45:43.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/90/817390/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/90/817390/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/90/817390/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/90/817390/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/90/817390/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/90/817390/9"}}},"commit":{"parents":[{"commit":"eeef3a48f413a702a1c0eb8c4b40db9a5f9d85bb","subject":"Merge \"Fix shibboleth compatibility for ubuntu 18.04\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/eeef3a48f413a702a1c0eb8c4b40db9a5f9d85bb"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2021-11-10 11:54:03.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-11-22 10:36:35.000000000","tz":0},"subject":"Drop Nginx webserver support","message":"Drop Nginx webserver support\n\nThere is no reason to support multiple web servers as a proxy for\nkeystone. Nginx is missing modules to support federation. With it\u0027s\nremoval we simplify code and reduce maintaining efforts needed.\n\nChange-Id: Ib3f90a72dfc8f78cf304b0f130883befdeb09220\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/818674\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/eb9a0c6ceaae98fe992bab1f9b5be797ba8a99b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/eb9a0c6ceaae98fe992bab1f9b5be797ba8a99b4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
