)]}'
{"id":"openstack%2Fopenstack-ansible-os_keystone~819860","triplet_id":"openstack%2Fopenstack-ansible-os_keystone~master~I9c4caf10192e2a25b1819d47065163ec78356a91","project":"openstack/openstack-ansible-os_keystone","branch":"master","topic":"config_template_collection","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2021-12-02 03:50:46.000000000","reason":"Dmitriy Rabotyagov replied on the change"}},"hashtags":[],"change_id":"I9c4caf10192e2a25b1819d47065163ec78356a91","subject":"Use config_template as a collection","status":"MERGED","created":"2021-11-30 13:17:21.000000000","updated":"2021-12-02 07:41:31.000000000","submitted":"2021-12-02 07:40:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"819860-config_template_collection","meta_rev_id":"3cfecbd29b12d19dd3e0e2f22a633a9e1541c342","_number":819860,"virtual_id_number":819860,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-02 07:40:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2021-12-01 08:26:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-12-01 07:55:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2021-12-01 08:26:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-12-01 17:28:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-30 13:26:19.000000000","updated_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"reviewer":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"state":"REVIEWER"},{"updated":"2021-11-30 18:06:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-01 07:55:05.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2021-12-01 08:26:04.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"65d403c6f47cdee1e7b6ed80af28896a7551abe3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-11-30 13:17:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5b7e513e88e335aea7d206f97f551de929e38eda","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2021-11-30 13:26:19.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"58bfd2de80dcbef5462af6bced02fc240b18d8fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-11-30 15:56:09.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"0d03dac63a9327af58754060fb54508b11eec1a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-30 18:06:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2605128bd26f4fbda6672c18902494f4 : SUCCESS in 12m 39s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/4d14ad33ebea444fb021ff9abce8df27 : SUCCESS in 1h 43m 43s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/fc22a9a85e6d4a34b709a73231bf7482 : SUCCESS in 1h 16m 59s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/19c5209855fb477ba181f3cebcab0ff9 : SUCCESS in 1h 33m 51s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/3790691355594f258e6520e967109b67 : SUCCESS in 1h 34m 57s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8ae0fb37267748b290f7f76050e31867 : SUCCESS in 1h 36m 17s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0a296cb00e904e60b9886d227a7c7e39 : SUCCESS in 1h 31m 32s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/f454924bb5af4cc0acab3de9dd486745 : SUCCESS in 2h 05m 49s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d1d54a67a27b452d8081875427c6a661 : SUCCESS in 1h 55m 38s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/2ccf68e800cd4bbe99b558789a807143 : FAILURE in 47m 08s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e15c8c6519e344769013e2bb9d25bf6d : SUCCESS in 1h 14m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c04641baebe04628bcf678df1a776b13 : SUCCESS in 9m 16s\n- openstack-ansible-uw_apache-centos-8 https://zuul.opendev.org/t/openstack/build/47baddc0e6c2411aad8fe9c42482976f : SUCCESS in 41m 02s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f4464ee66f1146329b7d2126104089b2 : SUCCESS in 39m 03s\n- openstack-ansible-uw_apache-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8c72997135ba448eb0d1d8e8b9c0ed72 : SUCCESS in 44m 06s","accounts_in_message":[],"_revision_number":2},{"id":"b44eecbe0b9d203c65de9e5e58fb0dbda6843725","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-12-01 07:55:05.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"bb232cc57d7f9b07febc73d8eac39b70041f425c","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2021-12-01 08:26:04.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"fb47598650dc9c5782490179fd1e2df569fee7ef","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-12-01 17:28:33.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"10c499181691e73bfa1d0bf95c8507f6afe0b943","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-01 17:36:02.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5ac94c17f2871f404965e7b419a544d328c7c23d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-01 19:42:33.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9e47b67996584acaae5b391e537ae4eb : SUCCESS in 18m 16s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/e02b2e2d29c042478a5baddc6f83c0b9 : SUCCESS in 1h 31m 05s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/add7f43a31c24fb79e65da472b3363a4 : SUCCESS in 1h 23m 19s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/45b53c087db24ad2a5bf84d448ec5176 : SUCCESS in 1h 12m 15s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/0c11d8219ddf42eea450adb686a199c9 : SUCCESS in 1h 37m 35s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9097a6c99ce64f00aadcd0fb946614ee : SUCCESS in 1h 35m 07s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/88531c9ce51e45a0965af507c1905b99 : SUCCESS in 1h 10m 49s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/73535d782eab43958d8f8b990d41aea0 : SUCCESS in 55m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5ef30b2bd324d7dbe6d88613753e7f5 : SUCCESS in 7m 53s\n- openstack-ansible-uw_apache-centos-8 https://zuul.opendev.org/t/openstack/build/55a0e10088a94fad84342accf762bcfc : TIMED_OUT in 1h 36m 07s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f6d1b8c5be7949fcb3faeac488654fc5 : SUCCESS in 38m 43s\n- openstack-ansible-uw_apache-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bc479c25b3e64a2eb27b953c53446179 : SUCCESS in 50m 10s","accounts_in_message":[],"_revision_number":2},{"id":"b5019544ea7d072650b9296d09216d61ac4ef372","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-12-02 03:50:46.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"47f008ae129b44377cbbbde34df7c29cd5989ce6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-02 06:03:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e45de5184fdb4be8b720a70dc1a55cee : SUCCESS in 12m 23s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/cdd22fde948c4666bb286b0a5f2c6eb1 : SUCCESS in 1h 33m 52s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/37e7c431c1cf4479bc75a6429d59eb9d : SUCCESS in 1h 13m 09s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/bcdfe02526d843598dd421c76a86fa25 : SUCCESS in 1h 17m 51s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/8451e4ff59164901bfa0f16927cfbb8e : SUCCESS in 1h 25m 35s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0f39db44a76947cf939c0c6f42e39328 : SUCCESS in 1h 25m 58s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cac84783bf114f96a7445445eb724b37 : SUCCESS in 1h 22m 02s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/86d8c3d2e3d94a4fb4e49bf01f253a1f : SUCCESS in 2h 04m 54s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/91be6c2f45f746469c3471bb1a569aef : SUCCESS in 1h 43m 45s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/0b852a89430f483e813d41ff30b5a83a : FAILURE in 44m 10s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2a934b89126e464ea31137685a7fcedd : SUCCESS in 55m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2ae6aba071242ba8aff9ad70752950d : SUCCESS in 5m 02s\n- openstack-ansible-uw_apache-centos-8 https://zuul.opendev.org/t/openstack/build/130d101dce8d4347be783203a948851e : SUCCESS in 41m 51s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e3726280da3c449e85a75e0cb94613b3 : SUCCESS in 37m 12s\n- openstack-ansible-uw_apache-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ab94c67f0d1e4a92aef59af3094c6918 : SUCCESS in 41m 12s","accounts_in_message":[],"_revision_number":2},{"id":"5eb255529f32791878f6b2d91062eef26b1a2b6f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-02 06:05:32.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b8f4928345fc0f32fcd2874635d5860afa2421e0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-02 07:40:13.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a7ffd3396ebb4633b807f155323b9a1b : SUCCESS in 13m 56s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/9c4f8748bb9844f7b09f8a2645e2ab60 : SUCCESS in 1h 29m 30s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/db3ce748735d459e998f4db304083c3e : SUCCESS in 1h 14m 27s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/60c02d97d2714559a8b6d53a15c21de4 : SUCCESS in 1h 19m 22s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/f1080aa680f14a80b5da40eaaa1f36f8 : SUCCESS in 1h 21m 31s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/911c3f64956244e998a00b18327a7579 : SUCCESS in 1h 29m 55s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a8ec92179b9047f386f00148120f4a92 : SUCCESS in 1h 22m 10s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d8d0548e5f4b491f968710a39fa913fe : SUCCESS in 56m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e23dbc91c9e4d549bf99d5e5140538b : SUCCESS in 5m 15s\n- openstack-ansible-uw_apache-centos-8 https://zuul.opendev.org/t/openstack/build/2cb9c00a9abc4037807fce93470c5e76 : SUCCESS in 37m 53s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8022c7907b824155bc277af9e17a41c3 : SUCCESS in 37m 09s\n- openstack-ansible-uw_apache-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bead97c3f7dc48e18344752b21f4740b : SUCCESS in 34m 37s","accounts_in_message":[],"_revision_number":2},{"id":"348c00269a730c57283890d93be99d376422a54e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-02 07:40:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"3cfecbd29b12d19dd3e0e2f22a633a9e1541c342","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-02 07:41:31.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/268f751de81a477fa4d4b4e48f90814a : SUCCESS in 52s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"59e879d28ab54d6bac480cb3e8e50f06f37f0d37","revisions":{"b19c506240294a1fd4963b3b4ae821bcb9936c21":{"kind":"REWORK","_number":1,"created":"2021-11-30 13:17:21.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/60/819860/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/60/819860/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/60/819860/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/60/819860/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/60/819860/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/60/819860/1"}}},"commit":{"parents":[{"commit":"91c397dc8bcb408abbd9a168f5f8279ec117eaaf","subject":"Merge \"Drop Nginx webserver support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/91c397dc8bcb408abbd9a168f5f8279ec117eaaf"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2021-11-30 13:17:19.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2021-11-30 13:17:19.000000000","tz":120},"subject":"Use config_template as a collection","message":"Use config_template as a collection\n\nSince we still use ceph-ansible that has their own implementation of\nconfig_template module it\u0027s worth to use mentioned module as a collection\nexplicitly.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/819814\n\n\n\nChange-Id: I9c4caf10192e2a25b1819d47065163ec78356a91","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/b19c506240294a1fd4963b3b4ae821bcb9936c21"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/b19c506240294a1fd4963b3b4ae821bcb9936c21"}]},"branch":"refs/heads/master"},"59e879d28ab54d6bac480cb3e8e50f06f37f0d37":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-11-30 15:56:09.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/60/819860/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/60/819860/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/60/819860/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/60/819860/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/60/819860/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/60/819860/2"}}},"commit":{"parents":[{"commit":"91c397dc8bcb408abbd9a168f5f8279ec117eaaf","subject":"Merge \"Drop Nginx webserver support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/91c397dc8bcb408abbd9a168f5f8279ec117eaaf"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2021-11-30 13:17:19.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-11-30 15:56:09.000000000","tz":0},"subject":"Use config_template as a collection","message":"Use config_template as a collection\n\nSince we still use ceph-ansible that has their own implementation of\nconfig_template module it\u0027s worth to use mentioned module as a collection\nexplicitly.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/819814\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-tests/+/819897\n\n\nChange-Id: I9c4caf10192e2a25b1819d47065163ec78356a91\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/59e879d28ab54d6bac480cb3e8e50f06f37f0d37"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/59e879d28ab54d6bac480cb3e8e50f06f37f0d37"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
