)]}'
{"id":"openstack%2Fopenstack-ansible-os_keystone~820225","triplet_id":"openstack%2Fopenstack-ansible-os_keystone~master~Ia507320ca552ec60d893e398ad7f68d4283539be","project":"openstack/openstack-ansible-os_keystone","branch":"master","topic":"db-pooling","attention_set":{},"removed_from_attention_set":{"31542":{"account":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"last_update":"2021-12-04 14:41:11.000000000","reason":"Change was submitted"},"32666":{"account":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"last_update":"2021-12-04 14:41:11.000000000","reason":"Change was submitted"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2021-12-04 07:27:28.000000000","reason":"Dmitriy Rabotyagov replied on the change"},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2021-12-04 13:09:52.000000000","reason":"Jonathan Rosser replied on the change"}},"hashtags":[],"change_id":"Ia507320ca552ec60d893e398ad7f68d4283539be","subject":"Database connection pooling improvements","status":"MERGED","created":"2021-12-02 18:18:38.000000000","updated":"2021-12-04 14:42:27.000000000","submitted":"2021-12-04 14:41:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"820225-db-pooling","meta_rev_id":"0b8e5fac52fec0ab0bcd29467d82d4bb09d20463","_number":820225,"virtual_id_number":820225,"owner":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-04 14:41:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2021-12-04 07:27:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-12-04 13:09:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-12-04 13:09:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}],"CC":[{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-02 20:40:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-03 08:09:08.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2021-12-03 08:44:18.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"CC"},{"updated":"2021-12-03 11:21:38.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"}],"messages":[{"id":"14f2b426991bc33d36a130f1ea219afd5dc2e57b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2021-12-02 18:18:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f3f37c1526cb6d0237494bce701fcf0808924223","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-02 20:40:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d517441bcb5646afb675fc5a08e2795b : SUCCESS in 19m 56s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/e66014528a9544a19acf6b74f1d14545 : SUCCESS in 1h 39m 37s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/78dcb104293d47458878afecd8d9ea75 : SUCCESS in 1h 17m 53s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/5e3f8b2d500a4fd6b8d8120161079838 : SUCCESS in 1h 33m 14s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/6a089ffe53ec41f387e727de9548ebe0 : SUCCESS in 1h 30m 19s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e211728fa98440ce894600a78b921b81 : SUCCESS in 1h 36m 33s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/62f1171c99874d22b33bc7ccaecb2744 : SUCCESS in 1h 32m 53s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/506e48622fc0407b949b6aab00ef3a30 : SUCCESS in 2h 17m 35s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/36c2e68610c14813add9c8cd702f2017 : SUCCESS in 2h 01m 31s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/7e9d8adbf8c2445cb15cf159774cdcb8 : FAILURE in 46m 11s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/699726eb714a4593b49922d3991cba59 : SUCCESS in 1h 08m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e6f223044d74042a058d0803d0e0ea6 : SUCCESS in 7m 02s\n- openstack-ansible-uw_apache-centos-8 https://zuul.opendev.org/t/openstack/build/dfc67680dff3477f858e1c2b6188e00d : SUCCESS in 48m 42s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e41bc904c69743339cc72d0efa0a626e : SUCCESS in 47m 01s\n- openstack-ansible-uw_apache-ubuntu-focal https://zuul.opendev.org/t/openstack/build/965e348afc0b4e55bb4744698af91b83 : SUCCESS in 50m 48s","accounts_in_message":[],"_revision_number":1},{"id":"cb6d14aff6ecbc8f9d1e9112f4b591cb3053fcaf","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-12-03 08:09:08.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"387c06109e8837a439cb1cdd9fcdb55ad116facd","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2021-12-03 08:44:18.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"93b3cf251a7bd8ea0efd941246db3d4cbf934091","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-12-03 11:21:38.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"64db6ef3e7355e67a13eab6d4fefd902f30e80d3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2021-12-03 15:43:46.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1db1b1ea63380c14b98b3c9372f8f2605f537605","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-12-03 15:45:31.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f4789c84b1c324136bcdd96b9d93e4e4fabb854c","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-12-03 15:49:03.000000000","message":"Removed Code-Review-1 by Dmitriy Rabotyagov \u003cnoonedeadpunk@ya.ru\u003e\n","accounts_in_message":[],"_revision_number":2},{"id":"3012246de4b46b90b4c2a47fb28f311b7dab38cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2021-12-03 15:53:52.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"2eed0e7e6852ce55af9ed411787b1c6b6aa6c3da","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2021-12-03 15:54:07.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"faf358d658a8abb5261b9ea79833cf6b93fc6503","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2021-12-03 15:54:45.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"ac3c77ba34cf8d0e95cb7031eb40c5f0fd4a8e82","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2021-12-03 15:57:07.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b03f593ea9fb64a5090ee2e3cf9ea8494c4f0eca","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2021-12-03 15:57:40.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"cbac9c29cc0bb3640451a900a1c436dbeee8ce33","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-03 18:01:10.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/73ce87f26d1a40d8b20182e3079209da : SUCCESS in 13m 48s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/55b54507b7844cabaf4de331f19a98a0 : SUCCESS in 1h 30m 46s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/a9f36d1b831748cfa0998ca57ef80de6 : SUCCESS in 1h 11m 52s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/ec008a538acd4c3e886da505a3378e7c : SUCCESS in 1h 34m 13s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/0bb17e608cb14c35bfdc14b214d89c80 : SUCCESS in 1h 22m 35s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/226d72ee737e43c090e530082803027c : SUCCESS in 1h 41m 28s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b8045c46457b4ceaab1766106f48f55a : SUCCESS in 1h 23m 05s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/a89d10fef75740e6aea1a5127d983350 : SUCCESS in 2h 02m 32s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/295b295bd666421d85a7dd35ab290f92 : SUCCESS in 1h 56m 23s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/c249d42fdd994ab282eb09885312b1c0 : FAILURE in 46m 34s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/61ee170345e44c44a07c5a29bf115b1a : SUCCESS in 57m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00f8c659b425406487984086eb4223b9 : SUCCESS in 6m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/24f18b61cc3f4335a8d01dfb9ed4402a : SUCCESS in 3m 54s\n- openstack-ansible-uw_apache-centos-8 https://zuul.opendev.org/t/openstack/build/d6fa44c359b34d6ab5de4e9e5158d0da : SUCCESS in 38m 41s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/46afb183d1804c47ae94c644ad489e06 : SUCCESS in 38m 18s\n- openstack-ansible-uw_apache-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bf0f9363a6f04844a5ad596fc4978fdd : SUCCESS in 37m 28s","accounts_in_message":[],"_revision_number":4},{"id":"e8ba24c5a65c7717d5d924263d5686c7d1c4d111","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-12-04 07:27:28.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"a7f156814dedc39775d0763b629e2d31095a3d49","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-12-04 13:09:52.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"b9a6e1fe008a262ab043044954833cfdd39b9831","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-04 13:11:57.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"310eaaf5f9390dd7d7d28d9fecc06fec09ff3f9d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-04 14:41:11.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9e1a6610d2e0400ea6fd604b4d48ada5 : SUCCESS in 14m 22s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/66e19c67681c43ef87bd567cacdad3bf : SUCCESS in 1h 27m 42s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/7a5c9068b7db409c8450fe966243bfd3 : SUCCESS in 1h 10m 21s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/0aead55c052b43e68c79ea4d7c234ff2 : SUCCESS in 1h 23m 37s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/86b0e759866c4b8498eff3507c80bf82 : SUCCESS in 1h 26m 08s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/29bd4bdf078442a6a8b5c6ab399e44cd : SUCCESS in 1h 29m 05s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1dbe5f01a9574310b050aba0dd656106 : SUCCESS in 1h 25m 51s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1621e690fb98422ea805e06385060978 : SUCCESS in 1h 00m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39a8cdea67b844b68b619fce8d0c4d70 : SUCCESS in 10m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c8cb0a806ca4491a8e90337a31c78cad : SUCCESS in 3m 46s\n- openstack-ansible-uw_apache-centos-8 https://zuul.opendev.org/t/openstack/build/eb316176c0da435eba4aa3ceedf22bc0 : SUCCESS in 44m 22s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d9801ab6aedf4360abbb12174e2dfd8e : SUCCESS in 38m 13s\n- openstack-ansible-uw_apache-ubuntu-focal https://zuul.opendev.org/t/openstack/build/80d7649a09a3472c8f8bbeb3e83ecaca : SUCCESS in 47m 43s","accounts_in_message":[],"_revision_number":4},{"id":"e55e669c854d3ca4396bf6e8c79c3e3f876bd4b1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-04 14:41:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"0b8e5fac52fec0ab0bcd29467d82d4bb09d20463","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-04 14:42:27.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1c31387a78c43b1b1c56784d82eae8a : SUCCESS in 53s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"b36b942aed1892910d73e8927ce83d890fdb29be","revisions":{"867f3c168fe217cb91339e7a7049ed8184568cb2":{"kind":"REWORK","_number":1,"created":"2021-12-02 18:18:38.000000000","uploader":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"ref":"refs/changes/25/820225/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/25/820225/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/25/820225/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/25/820225/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/25/820225/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/25/820225/1"}}},"commit":{"parents":[{"commit":"59e879d28ab54d6bac480cb3e8e50f06f37f0d37","subject":"Use config_template as a collection","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/59e879d28ab54d6bac480cb3e8e50f06f37f0d37"}]}],"author":{"name":"Damian Dabrowski","email":"damian@dabrowski.cloud","date":"2021-12-02 18:18:27.000000000","tz":60},"committer":{"name":"Damian Dabrowski","email":"damian@dabrowski.cloud","date":"2021-12-02 18:18:36.000000000","tz":60},"subject":"Database connection pooling improvements","message":"Database connection pooling improvements\n\n- Implemented new variable ``connection_recycle_time`` responsible for SQLAlchemy\u0027s connection recycling\n- Set new default values for db pooling variables which are inherited from the global ones.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/819424\nChange-Id: Ia507320ca552ec60d893e398ad7f68d4283539be\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/867f3c168fe217cb91339e7a7049ed8184568cb2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/867f3c168fe217cb91339e7a7049ed8184568cb2"}]},"branch":"refs/heads/master"},"55bdbcde848769f688e8f06cc61e3f39ee4db1c8":{"kind":"REWORK","_number":2,"created":"2021-12-03 15:43:46.000000000","uploader":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"ref":"refs/changes/25/820225/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/25/820225/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/25/820225/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/25/820225/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/25/820225/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/25/820225/2"}}},"commit":{"parents":[{"commit":"59e879d28ab54d6bac480cb3e8e50f06f37f0d37","subject":"Use config_template as a collection","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/59e879d28ab54d6bac480cb3e8e50f06f37f0d37"}]}],"author":{"name":"Damian Dabrowski","email":"damian@dabrowski.cloud","date":"2021-12-02 18:18:27.000000000","tz":60},"committer":{"name":"Damian Dabrowski","email":"damian@dabrowski.cloud","date":"2021-12-03 15:43:39.000000000","tz":60},"subject":"Database connection pooling improvements","message":"Database connection pooling improvements\n\n- Implemented new variable ``connection_recycle_time`` responsible for SQLAlchemy\u0027s connection recycling\n- Set new default values for db pooling variables which are inherited from the global ones.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/819424\nChange-Id: Ia507320ca552ec60d893e398ad7f68d4283539be\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/55bdbcde848769f688e8f06cc61e3f39ee4db1c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/55bdbcde848769f688e8f06cc61e3f39ee4db1c8"}]},"branch":"refs/heads/master"},"f12da6e036d9c88a698cb5c78015e0b56c2ce0c8":{"kind":"REWORK","_number":3,"created":"2021-12-03 15:53:52.000000000","uploader":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"ref":"refs/changes/25/820225/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/25/820225/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/25/820225/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/25/820225/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/25/820225/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/25/820225/3"}}},"commit":{"parents":[{"commit":"59e879d28ab54d6bac480cb3e8e50f06f37f0d37","subject":"Use config_template as a collection","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/59e879d28ab54d6bac480cb3e8e50f06f37f0d37"}]}],"author":{"name":"Damian Dabrowski","email":"damian@dabrowski.cloud","date":"2021-12-02 18:18:27.000000000","tz":60},"committer":{"name":"Damian Dabrowski","email":"damian@dabrowski.cloud","date":"2021-12-03 15:53:44.000000000","tz":60},"subject":"Database connection pooling improvements","message":"Database connection pooling improvements\n\n- Implemented new variable ``connection_recycle_time`` responsible for SQLAlchemy\u0027s connection recycling\n- Set new default values for db pooling variables which are inherited from the global ones.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/819424\nChange-Id: Ia507320ca552ec60d893e398ad7f68d4283539be\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/f12da6e036d9c88a698cb5c78015e0b56c2ce0c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/f12da6e036d9c88a698cb5c78015e0b56c2ce0c8"}]},"branch":"refs/heads/master"},"b36b942aed1892910d73e8927ce83d890fdb29be":{"kind":"REWORK","_number":4,"created":"2021-12-03 15:54:45.000000000","uploader":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"ref":"refs/changes/25/820225/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/25/820225/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/25/820225/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/25/820225/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/25/820225/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/25/820225/4"}}},"commit":{"parents":[{"commit":"59e879d28ab54d6bac480cb3e8e50f06f37f0d37","subject":"Use config_template as a collection","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/59e879d28ab54d6bac480cb3e8e50f06f37f0d37"}]}],"author":{"name":"Damian Dabrowski","email":"damian@dabrowski.cloud","date":"2021-12-02 18:18:27.000000000","tz":60},"committer":{"name":"Damian Dabrowski","email":"damian@dabrowski.cloud","date":"2021-12-03 15:54:38.000000000","tz":60},"subject":"Database connection pooling improvements","message":"Database connection pooling improvements\n\n- Implemented new variable ``connection_recycle_time`` responsible for SQLAlchemy\u0027s connection recycling\n- Set new default values for db pooling variables which are inherited from the global ones.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/819424\nChange-Id: Ia507320ca552ec60d893e398ad7f68d4283539be\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/b36b942aed1892910d73e8927ce83d890fdb29be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/b36b942aed1892910d73e8927ce83d890fdb29be"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
