)]}'
{"id":"openstack%2Fopenstack-ansible-os_keystone~820999","triplet_id":"openstack%2Fopenstack-ansible-os_keystone~master~I09f9fda3699ece3300a666d0c103da4e022d70e2","project":"openstack/openstack-ansible-os_keystone","branch":"master","topic":"osa/common_tasks","attention_set":{},"removed_from_attention_set":{"31542":{"account":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"last_update":"2022-01-13 14:20:44.000000000","reason":"Andrew Bonney replied on the change"},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2022-01-14 10:30:21.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I09f9fda3699ece3300a666d0c103da4e022d70e2","subject":"Use common service setup tasks from a collection rather than in-role","status":"MERGED","created":"2021-12-08 10:41:56.000000000","updated":"2022-01-14 10:32:05.000000000","submitted":"2022-01-14 10:30:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"820999-osa/common_tasks","meta_rev_id":"c4ccc22e986956df8c57ba8c91109f6527ee8f38","_number":820999,"virtual_id_number":820999,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-14 10:30:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2022-01-13 14:20:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":2,"date":"2022-01-13 11:31:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2022-01-14 08:48:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-08 11:53:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-11 19:37:26.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2022-01-12 12:54:05.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"7b8397cfa75e07338df043e9d45e6f5f243e2193","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-12-08 10:41:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"16af7fa23ed8258d194a6a9421b4ade89706294b","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-12-08 10:42:40.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"52edf030c3d82230b6575da28353d7ad7b7ff0df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-08 11:53:48.000000000","message":"Patch Set 1: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"c136910c8545ec49be877f4155ffd5526929fb47","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-12-08 11:56:37.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"cec2caf49d512fe343c59786ca781f6e780ac0c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-08 13:22:57.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1ca1e0ee43b0454aa2b89a1b2d5e5dda : FAILURE in 17m 39s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/0f8b957da0984141b2f8a767d7c6784a : FAILURE in 33m 30s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/92e3d66644ab4b6ca186701684ecd645 : FAILURE in 43m 21s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fb758b1e497b40829f407ea6864ce3ae : FAILURE in 35m 25s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/53dbf30ac17f460589436c01f02849a2 : FAILURE in 1h 12m 26s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6d28293048574b2d989d682235ad0eab : FAILURE in 1h 19m 47s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/ab8fbeb0303b4be5a43932af40167c68 : FAILURE in 39m 16s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/127cbea78f56483d95fb965fc35dacb9 : FAILURE in 20m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7232c71a0de74c01af77ab7535ce55ce : SUCCESS in 6m 10s","accounts_in_message":[],"_revision_number":2},{"id":"cfa87e8191678b7c45afe4e1c71683c56bfbcf2b","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-12-08 13:28:35.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"18dcb0a9111ee6c4d1134152d27f7483ef6ee4bb","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-12-08 13:28:41.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":2},{"id":"e1de67d362bad0ed0c3241454945cf61bedc0e77","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-08 15:43:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8e89720436b14232bf147cf5bf6ff675 : SUCCESS in 20m 28s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/04319c7e707b4c1c9f25295840e9cb21 : SUCCESS in 1h 34m 42s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/2110fa97f3a84fdbaecbde79571bea3d : SUCCESS in 1h 27m 02s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/246448caab3c450eb309120a17357cb6 : SUCCESS in 1h 15m 23s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/672abb5387f947b59482b4e19ec6916e : SUCCESS in 2h 08m 07s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/12c470415d274476979d39189ae111fe : SUCCESS in 2h 03m 54s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/f0884381b149482c8a124da207ca74ff : FAILURE in 41m 51s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/78585d70dcb74ddeaae335e4f2e4444e : SUCCESS in 56m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/016c936c5eb347aab4b0fecff0a44bcd : SUCCESS in 6m 23s","accounts_in_message":[],"_revision_number":2},{"id":"0f34985320fe938efe8eccd11adbe7664c2db888","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-01-11 09:46:38.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"f16ca1690c173b493ba4f0e5f70d94ffd6b1f83e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-01-11 10:53:55.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"c4e1bcdb641169413ec7ba2150611c40b2158c0f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-01-11 12:17:55.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"e61615f4c92e1c449046d808fa0f690f2b16cb55","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-11 13:44:48.000000000","message":"Patch Set 5: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":5},{"id":"1424903e6f61dd58903ffe5cf29954f56186d48c","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-01-11 13:46:41.000000000","message":"Patch Set 5:\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":5},{"id":"87f8564babee4d0eb95390d2299d04f879b80845","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-11 13:53:21.000000000","message":"Patch Set 5:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":5},{"id":"eb350885da5155312cdaebe5e89686f5340fbf3c","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-01-11 15:41:09.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"baf9ec59bbba0476ec5b1bb8a6a7afce1e449ca7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-11 18:08:35.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/78c6a1893c834c5b94647ca1ce970a05 : SUCCESS in 26m 44s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/63fa1d93a27b45d1bb75ecc96680e0e2 : SUCCESS in 1h 25m 09s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/239d130cff7f45e9ac5b7f1e94b4ad21 : SUCCESS in 1h 27m 04s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/620b8832209d4b1ca4b756076c8fdea5 : SUCCESS in 1h 27m 21s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/45ca4ac7766e424994d8b501ad12e519 : SUCCESS in 1h 37m 49s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/c69d42ecc3034a4f8eccf4c2e4eac282 : SUCCESS in 2h 11m 39s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e28ac224d7254a28877788762f2f93b6 : SUCCESS in 2h 19m 32s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/fd0555c5c5a44f95b67646c8772f41b5 : FAILURE in 34m 52s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/401e4d609b3e46528f6b38b80d9a92e3 : SUCCESS in 1h 07m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b80b67724df43ac9d1b5f9b897b57ea : SUCCESS in 7m 32s","accounts_in_message":[],"_revision_number":5},{"id":"1e26d6fd2445fcac7bd82bf5d4c24d99c36ad2a5","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-01-11 19:37:26.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"d81eec519eafcd15f82f0e7c40ec9cb622815aa3","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-01-12 11:32:28.000000000","message":"Topic set to osa/common_tasks","accounts_in_message":[],"_revision_number":5},{"id":"09d648f15ca51ac046f0043be4fc6fe86285976d","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2022-01-12 12:54:05.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"16c163f307f09d8d84836e6a837a504360b8d3d3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-01-12 17:07:11.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"b75e91263afaa7dec88fd6a8f76e3b512085ad3b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-01-12 17:15:08.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"18479389369b4fb2d30e17174e50829b0ed8c707","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-01-12 17:20:32.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"41779239c4ac70a66581516521e78d98ca63abb8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 20:13:44.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/38583b9d8e7a4b62b5c30c163b76567b : SUCCESS in 21m 09s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/de92d9f04eb544e8b4e7a768c9af2d85 : SUCCESS in 1h 16m 37s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/64771d1e31a34efcb6e99aaa1e5cf1ba : SUCCESS in 1h 22m 56s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/44ee71c447684a32878ece832bddb97c : SUCCESS in 1h 24m 53s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cea4cc36a8084834a2ce88bd711b8131 : SUCCESS in 1h 33m 21s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/1f211d92c3614662867bd9999e1c248f : SUCCESS in 2h 10m 37s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1478f767bae745f39bff3d5700e1861a : SUCCESS in 2h 41m 45s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/25e24e39a061487a95c7acde926e71a4 : FAILURE in 34m 53s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1557a29e904d4eb283d95ca9f441a949 : SUCCESS in 1h 09m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/690077e12e2e4ec49f8399506935420c : SUCCESS in 7m 41s","accounts_in_message":[],"_revision_number":8},{"id":"575e4e86130a7b37584d304e295169caa1b664d8","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-01-13 10:59:57.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"429820c919c42d7a35bfddb1bbd916830c208125","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2022-01-13 11:26:37.000000000","message":"Patch Set 9: Published edit on patch set 8.","accounts_in_message":[],"_revision_number":9},{"id":"27db855089d16671c7a049052ae296b8a6ecfd28","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-01-13 11:31:35.000000000","message":"Patch Set 9: Code-Review+2","accounts_in_message":[],"_revision_number":9},{"id":"73d613d649fa7d66659847057d50963fcc67b4ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-13 14:13:09.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/df0569de7d3a46c3b2de18bbc4f574c7 : SUCCESS in 19m 55s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/1fd6d1ec13f4421ebd3e44ce4bf75112 : SUCCESS in 1h 22m 24s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/99ecdad0d86c4e7bb30aa80a9410ba08 : SUCCESS in 1h 34m 33s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/61f9559454bd4abea0c5992eafce4e6f : SUCCESS in 1h 36m 28s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a808422040e1449f8b42b44431801114 : SUCCESS in 1h 38m 00s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/85b5ea16e64349f0be3a0913dbe3b0b9 : SUCCESS in 2h 13m 51s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0221e6f0b4ab4806bd4d27fc4d747fd0 : SUCCESS in 2h 24m 54s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/25a1313aae59454fa74804913fea3eb5 : FAILURE in 37m 36s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3fcba1fd66ef4ae8947a6692d5456e00 : SUCCESS in 1h 09m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48741e8e1bce4528b7ab54fefae87157 : SUCCESS in 7m 33s","accounts_in_message":[],"_revision_number":9},{"id":"2a5ba472fa17ca2658acfac72798694e10b15a85","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2022-01-13 14:20:44.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"62fbda5f73efd828b23ffad97d64a7108a3aca43","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2022-01-14 08:47:58.000000000","message":"Removed Workflow+1 by Andrew Bonney \u003candrew.bonney@bbc.co.uk\u003e\n","accounts_in_message":[],"_revision_number":9},{"id":"dfd28a86897bf8bfb1344da2d1fa023fc10ed7d6","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2022-01-14 08:48:01.000000000","message":"Patch Set 9: Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"4db4ca6a72c950548eadacafd7db7331eaf152c8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-14 08:51:47.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"7e7bddc99e3190cd5a86ed4d87e82c448c529472","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-14 10:30:20.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/af07772b6c2c4f46ad420b126d5991bd : SUCCESS in 15m 58s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/32c925ab2a704cd2930218fa169341c9 : SUCCESS in 1h 20m 43s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/7e64d375b8bf4e9f96692331d51f3328 : SUCCESS in 1h 23m 36s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/134b51c4bfe14bcd917acf9fa90983ba : SUCCESS in 1h 31m 10s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c1763d00a67e42d6a0badfa94d7a3ad0 : SUCCESS in 54m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b58229141d44fabb63930b43b30cb35 : SUCCESS in 5m 39s","accounts_in_message":[],"_revision_number":9},{"id":"731871a893b6e47d4c78b80c953a0c2901ebdc8c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-14 10:30:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":9},{"id":"c4ccc22e986956df8c57ba8c91109f6527ee8f38","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-14 10:32:05.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/037b9b824d4241f6a16b1e29896c8d38 : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"0985cfa47d25c09e1bd63d8da5950e6e568ebbc1","revisions":{"6ea2e4d23e861649f61226a28537bf1b7fd833c0":{"kind":"REWORK","_number":1,"created":"2021-12-08 10:41:56.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/99/820999/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/99/820999/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/99/820999/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/99/820999/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/99/820999/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/99/820999/1"}}},"commit":{"parents":[{"commit":"d5542a5b81147698df75a5bc2ffcba47a2bd95f3","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/d5542a5b81147698df75a5bc2ffcba47a2bd95f3"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-12-08 10:38:14.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-12-08 10:41:40.000000000","tz":0},"subject":"Use common service setup tasks from a collection rather than in-role","message":"Use common service setup tasks from a collection rather than in-role\n\nDepends-On: I8c2e9394dc51d0930b2c340afd0b28fde86a966b\nChange-Id: I09f9fda3699ece3300a666d0c103da4e022d70e2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/6ea2e4d23e861649f61226a28537bf1b7fd833c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/6ea2e4d23e861649f61226a28537bf1b7fd833c0"}]},"branch":"refs/heads/master"},"1e5f7caeccb88b22c2c314eeaad59564be1dfff8":{"kind":"REWORK","_number":2,"created":"2021-12-08 11:56:37.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/99/820999/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/99/820999/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/99/820999/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/99/820999/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/99/820999/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/99/820999/2"}}},"commit":{"parents":[{"commit":"d5542a5b81147698df75a5bc2ffcba47a2bd95f3","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/d5542a5b81147698df75a5bc2ffcba47a2bd95f3"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-12-08 10:38:14.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-12-08 11:56:32.000000000","tz":0},"subject":"Use common service setup tasks from a collection rather than in-role","message":"Use common service setup tasks from a collection rather than in-role\n\nDepends-On: I8c2e9394dc51d0930b2c340afd0b28fde86a966b\nChange-Id: I09f9fda3699ece3300a666d0c103da4e022d70e2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/1e5f7caeccb88b22c2c314eeaad59564be1dfff8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/1e5f7caeccb88b22c2c314eeaad59564be1dfff8"}]},"branch":"refs/heads/master"},"ce61079572e37a1bec054128c53803c3b76faf34":{"kind":"REWORK","_number":3,"created":"2022-01-11 09:46:38.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/99/820999/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/99/820999/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/99/820999/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/99/820999/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/99/820999/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/99/820999/3"}}},"commit":{"parents":[{"commit":"d5542a5b81147698df75a5bc2ffcba47a2bd95f3","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/d5542a5b81147698df75a5bc2ffcba47a2bd95f3"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-12-08 10:38:14.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-01-11 09:46:34.000000000","tz":0},"subject":"Use common service setup tasks from a collection rather than in-role","message":"Use common service setup tasks from a collection rather than in-role\n\nDepends-On: I8c2e9394dc51d0930b2c340afd0b28fde86a966b\nChange-Id: I09f9fda3699ece3300a666d0c103da4e022d70e2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/ce61079572e37a1bec054128c53803c3b76faf34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/ce61079572e37a1bec054128c53803c3b76faf34"}]},"branch":"refs/heads/master"},"06c48e51d91a3278bd71b2327dd2b7f9d359c4f5":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2022-01-11 10:53:55.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/99/820999/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/99/820999/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/99/820999/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/99/820999/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/99/820999/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/99/820999/4"}}},"commit":{"parents":[{"commit":"d5542a5b81147698df75a5bc2ffcba47a2bd95f3","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/d5542a5b81147698df75a5bc2ffcba47a2bd95f3"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-12-08 10:38:14.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-01-11 10:53:55.000000000","tz":0},"subject":"Use common service setup tasks from a collection rather than in-role","message":"Use common service setup tasks from a collection rather than in-role\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/821093/7\nChange-Id: I09f9fda3699ece3300a666d0c103da4e022d70e2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/06c48e51d91a3278bd71b2327dd2b7f9d359c4f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/06c48e51d91a3278bd71b2327dd2b7f9d359c4f5"}]},"branch":"refs/heads/master","description":"Edit commit message"},"aef1ab9cf50df20564ded70616304fd38fa1e584":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2022-01-11 12:17:55.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/99/820999/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/99/820999/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/99/820999/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/99/820999/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/99/820999/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/99/820999/5"}}},"commit":{"parents":[{"commit":"d5542a5b81147698df75a5bc2ffcba47a2bd95f3","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/d5542a5b81147698df75a5bc2ffcba47a2bd95f3"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-12-08 10:38:14.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-01-11 12:17:55.000000000","tz":0},"subject":"Use common service setup tasks from a collection rather than in-role","message":"Use common service setup tasks from a collection rather than in-role\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-plugins/+/824115\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/821093/7\nChange-Id: I09f9fda3699ece3300a666d0c103da4e022d70e2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/aef1ab9cf50df20564ded70616304fd38fa1e584"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/aef1ab9cf50df20564ded70616304fd38fa1e584"}]},"branch":"refs/heads/master","description":"Edit commit message"},"7964da47be040240d2e269e2d05dfb9beb45dcd4":{"kind":"REWORK","_number":6,"created":"2022-01-12 17:07:11.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/99/820999/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/99/820999/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/99/820999/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/99/820999/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/99/820999/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/99/820999/6"}}},"commit":{"parents":[{"commit":"d5542a5b81147698df75a5bc2ffcba47a2bd95f3","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/d5542a5b81147698df75a5bc2ffcba47a2bd95f3"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-12-08 10:38:14.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-01-12 17:07:03.000000000","tz":0},"subject":"Use common service setup tasks from a collection rather than in-role","message":"Use common service setup tasks from a collection rather than in-role\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-plugins/+/824115\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/821093/7\nChange-Id: I09f9fda3699ece3300a666d0c103da4e022d70e2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/7964da47be040240d2e269e2d05dfb9beb45dcd4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/7964da47be040240d2e269e2d05dfb9beb45dcd4"}]},"branch":"refs/heads/master"},"9233d4ba316d2ad944e2d8af186aac4e4c04cac8":{"kind":"REWORK","_number":7,"created":"2022-01-12 17:15:08.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/99/820999/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/99/820999/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/99/820999/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/99/820999/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/99/820999/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/99/820999/7"}}},"commit":{"parents":[{"commit":"d5542a5b81147698df75a5bc2ffcba47a2bd95f3","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/d5542a5b81147698df75a5bc2ffcba47a2bd95f3"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-12-08 10:38:14.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-01-12 17:14:56.000000000","tz":0},"subject":"Use common service setup tasks from a collection rather than in-role","message":"Use common service setup tasks from a collection rather than in-role\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-plugins/+/824115\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/821093/7\nChange-Id: I09f9fda3699ece3300a666d0c103da4e022d70e2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/9233d4ba316d2ad944e2d8af186aac4e4c04cac8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/9233d4ba316d2ad944e2d8af186aac4e4c04cac8"}]},"branch":"refs/heads/master"},"092567426f89b42fa9a92f0b68816ed2a266ccc9":{"kind":"REWORK","_number":8,"created":"2022-01-12 17:20:32.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/99/820999/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/99/820999/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/99/820999/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/99/820999/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/99/820999/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/99/820999/8"}}},"commit":{"parents":[{"commit":"d5542a5b81147698df75a5bc2ffcba47a2bd95f3","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/d5542a5b81147698df75a5bc2ffcba47a2bd95f3"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-12-08 10:38:14.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-01-12 17:20:24.000000000","tz":0},"subject":"Use common service setup tasks from a collection rather than in-role","message":"Use common service setup tasks from a collection rather than in-role\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-plugins/+/824115\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/821093/7\nChange-Id: I09f9fda3699ece3300a666d0c103da4e022d70e2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/092567426f89b42fa9a92f0b68816ed2a266ccc9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/092567426f89b42fa9a92f0b68816ed2a266ccc9"}]},"branch":"refs/heads/master"},"0985cfa47d25c09e1bd63d8da5950e6e568ebbc1":{"kind":"REWORK","_number":9,"created":"2022-01-13 11:26:37.000000000","uploader":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"ref":"refs/changes/99/820999/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/99/820999/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/99/820999/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/99/820999/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/99/820999/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/99/820999/9"}}},"commit":{"parents":[{"commit":"d5542a5b81147698df75a5bc2ffcba47a2bd95f3","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/d5542a5b81147698df75a5bc2ffcba47a2bd95f3"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-12-08 10:38:14.000000000","tz":0},"committer":{"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","date":"2022-01-13 11:26:37.000000000","tz":0},"subject":"Use common service setup tasks from a collection rather than in-role","message":"Use common service setup tasks from a collection rather than in-role\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-plugins/+/824115\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/821093/7\nChange-Id: I09f9fda3699ece3300a666d0c103da4e022d70e2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/0985cfa47d25c09e1bd63d8da5950e6e568ebbc1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/0985cfa47d25c09e1bd63d8da5950e6e568ebbc1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
