)]}'
{"id":"openstack%2Fopenstack-ansible-os_keystone~826551","triplet_id":"openstack%2Fopenstack-ansible-os_keystone~stable%2Fwallaby~I56a6c47e4d95e86dc1e0d731f1e39eeec6ac7dc8","project":"openstack/openstack-ansible-os_keystone","branch":"stable/wallaby","topic":"osa_tls13-stable/xena-stable/wallaby","attention_set":{},"removed_from_attention_set":{"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2022-02-04 13:50:25.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I56a6c47e4d95e86dc1e0d731f1e39eeec6ac7dc8","subject":"Fix ordering error enabling/disabling Apache modules","status":"MERGED","created":"2022-01-26 21:46:17.000000000","updated":"2022-02-04 13:54:08.000000000","submitted":"2022-02-04 13:50:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"826551-osa_tls13-stable/xena-stable/wallaby","cherry_pick_of_change":826550,"cherry_pick_of_patch_set":1,"meta_rev_id":"2d872fe8f08ba8d6c66ffaa613d59ea6c59568f2","_number":826551,"virtual_id_number":826551,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-04 13:50:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2022-02-01 16:08:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-01-27 07:47:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2022-02-01 16:08:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}],"CC":[{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-26 21:46:17.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"CC"},{"updated":"2022-01-26 23:15:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-01 16:08:04.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2022-02-04 09:26:02.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"CC"}],"messages":[{"id":"f9613e020ff2cc586d4d8bc3f122f312b30cffd4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-01-26 21:46:17.000000000","message":"Patch Set 1: Cherry Picked from branch stable/xena.","accounts_in_message":[],"_revision_number":1},{"id":"5a568f2b5bc22ce76cbd33cdb4e4938e5db0466c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-26 23:15:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d56e8a75e6874a30aa149bf3e6d8a5c3 : SUCCESS in 15m 16s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/c8cadc056b7c43f4bd2f2012f869e13a : SUCCESS in 1h 09m 47s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/4603ee6985ff4844a0b870e99b90c41a : SUCCESS in 1h 01m 06s (non-voting)\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/a00fb7c52903432eabefd0aa92ba4025 : SUCCESS in 1h 13m 40s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/27667734002447a6877d5b43e2146270 : SUCCESS in 1h 06m 00s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/029e588b68c045c395da0a7325b6bcfd : SUCCESS in 1h 03m 47s\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/25168b4694af4a5b921d8c628610a94f : SUCCESS in 1h 18m 17s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/b2188895bff0414a83fdd8d6ca136ab2 : FAILURE in 44m 07s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/af4fbfe5dce5412fb806ceec4ff84a3c : SUCCESS in 54m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7bc5dc8cb6854441b71096d4a93d2840 : SUCCESS in 5m 34s\n- openstack-ansible-uw_apache-centos-8 https://zuul.opendev.org/t/openstack/build/51755e00ef324a19a97507fb575f2f68 : SUCCESS in 42m 31s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/00d043848a9d4949b049bb93c0251837 : SUCCESS in 46m 54s\n- openstack-ansible-uw_apache-ubuntu-focal https://zuul.opendev.org/t/openstack/build/95e3db6717c74c349729c891ba23931e : SUCCESS in 41m 46s","accounts_in_message":[],"_revision_number":1},{"id":"fc388647d1070ef9b7a9359f41fc41ba3bac9a11","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-01-27 07:47:20.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"eb2285e05f11bd61991c72dfc0d03f9327aa9391","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2022-02-01 16:08:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"da50a8725fdc55540cb9930c0e5a2589c3cb1b51","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-01 16:13:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5e0a1d9b73811db18b5fb2f0688cc6c8ce5a798c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-01 18:05:44.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8f6aca63d5f04f55ad4c43ad5c8ff336 : SUCCESS in 23m 35s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/e89649f321cb48d69cecf72738b3a0f9 : SUCCESS in 1h 11m 32s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/61a3f94dd40e4fd8a2d6d2704d1975b1 : SUCCESS in 1h 38m 34s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/413161b43a624e199a95a5b03606330b : SUCCESS in 1h 12m 54s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0a84ff4607ce4109a79531a1d21a0cdd : SUCCESS in 1h 16m 40s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/99852c1cbee144199db5d2cdb42ccafd : SUCCESS in 1h 03m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f035747e8af1478386bc441d68eb1086 : SUCCESS in 5m 31s\n- openstack-ansible-uw_apache-centos-8 https://zuul.opendev.org/t/openstack/build/4640d4abd15f4b42a82c571658b90b4a : RETRY_LIMIT in 6m 37s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e89baff24c1b4d9dafd7b952bda38c5e : SUCCESS in 46m 26s\n- openstack-ansible-uw_apache-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ded2879371574d518142e55db8ca3d86 : SUCCESS in 43m 34s","accounts_in_message":[],"_revision_number":1},{"id":"d78bc9579f56e2fc30845620d832e13f7a36d345","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-02-04 09:26:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d2bcc0ef21caf1feefd10de558bc7733b8941adc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 11:33:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b9ea2c7aabdd495ebe5b202ef48d18b7 : SUCCESS in 24m 06s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/7f3c2803936f40929f14cbda362d013f : SUCCESS in 1h 20m 22s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/78c20cb2e586403eb778c01a409c9628 : SUCCESS in 1h 19m 24s (non-voting)\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/eea078e5f9934074a57474d32c2c541f : SUCCESS in 1h 27m 24s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4ad910509c5b4f03ab6ba5ad4a3541fb : SUCCESS in 1h 27m 13s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b9d0fbdef99343379ea222bfa956377c : SUCCESS in 1h 16m 38s\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/307ed174fbe344d4a943e03c5e9228d7 : SUCCESS in 1h 47m 34s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/a981eb6bc0464cfbad65aabbbac943f2 : FAILURE in 44m 56s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b1b2b7bbfe7344328e479a5bbce2e357 : SUCCESS in 57m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ae3fb755732449bbb125797b5f22b73 : SUCCESS in 6m 26s\n- openstack-ansible-uw_apache-centos-8 https://zuul.opendev.org/t/openstack/build/5f92514e147c44c49da14149df0cba58 : SUCCESS in 51m 16s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2641f2476b9546d5b9f09c69176cbd49 : SUCCESS in 43m 34s\n- openstack-ansible-uw_apache-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cdc5d63186e140b4a37ab5c07fb0731e : SUCCESS in 47m 44s","accounts_in_message":[],"_revision_number":1},{"id":"6ccc63510ced02c0120137570aa20ff72f08ca3c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 12:19:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"13848beddc6fc0b89ed046886bfe7be403e4cf88","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 13:50:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f3743a9052b24d908e1ca356724632b9 : SUCCESS in 18m 15s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/26dec586bc714e99afd8495ecadeec9d : SUCCESS in 1h 17m 32s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/5f99bcf780fc4830af69484cf95f250e : SUCCESS in 1h 26m 49s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ea09cf9f369e4dcc872a8a26674516ae : SUCCESS in 1h 24m 43s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/edaef7b837174cd3ab5fe1d6685fcb11 : SUCCESS in 1h 09m 37s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2d2a8672bc0a46e180b8f4a417a638f7 : SUCCESS in 58m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7689975434148999cd47dcdbd6b17dc : SUCCESS in 7m 30s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2f416417257443918fafc6d6744ed8c4 : SUCCESS in 43m 06s\n- openstack-ansible-uw_apache-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6987ebaa79b6427b91325729ad098b29 : SUCCESS in 48m 11s","accounts_in_message":[],"_revision_number":1},{"id":"52e3975b7bc632cdbca7b3470fc084acc2cc4071","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 13:50:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"2d872fe8f08ba8d6c66ffaa613d59ea6c59568f2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 13:54:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f870f3d5495144a5aa0b25f3db5a978c : SUCCESS in 1m 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"38fd9f6349b031c5451518be13e38b0b7b90de8c","revisions":{"38fd9f6349b031c5451518be13e38b0b7b90de8c":{"kind":"REWORK","_number":1,"created":"2022-01-26 21:46:17.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/51/826551/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/51/826551/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/51/826551/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/51/826551/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/51/826551/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/51/826551/1"}}},"commit":{"parents":[{"commit":"0c7359379121e5b30078198196d45d0253735d59","subject":"Avoid Apache conflict with haproxy on metal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/0c7359379121e5b30078198196d45d0253735d59"}]}],"author":{"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","date":"2022-01-10 14:13:43.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-01-26 21:46:17.000000000","tz":0},"subject":"Fix ordering error enabling/disabling Apache modules","message":"Fix ordering error enabling/disabling Apache modules\n\nWhen site configuration already exists, a change to make a module\n\u0027absent\u0027 would fail as the module was removed before the\nconfiguration.\n\nThis change ensures modules are enabled first, before site\nconfiguration changes, and finally any required modules are\ndisabled.\n\nChange-Id: I56a6c47e4d95e86dc1e0d731f1e39eeec6ac7dc8\n(cherry picked from commit 236f11c31d6b2d395a9ec8846e240841510825ce)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/38fd9f6349b031c5451518be13e38b0b7b90de8c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/38fd9f6349b031c5451518be13e38b0b7b90de8c"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
