)]}'
{"id":"openstack%2Fopenstack-ansible-os_keystone~827603","triplet_id":"openstack%2Fopenstack-ansible-os_keystone~stable%2Fvictoria~I1457d563abcbc9e62c331375ed677758b5f73927","project":"openstack/openstack-ansible-os_keystone","branch":"stable/victoria","topic":"osa/remove-centos8","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2022-02-04 11:36:00.000000000","reason":"Dmitriy Rabotyagov replied on the change"},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2022-02-05 21:14:50.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1457d563abcbc9e62c331375ed677758b5f73927","subject":"Restore CI jobs","status":"MERGED","created":"2022-02-03 08:51:56.000000000","updated":"2022-02-05 21:18:22.000000000","submitted":"2022-02-05 21:14:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"827603-osa/remove-centos8","meta_rev_id":"3d16a9a68f18fdfbb2362425dac30f9189d6a033","_number":827603,"virtual_id_number":827603,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-05 21:14:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2022-02-03 21:35:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2022-02-05 19:44:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-03 08:52:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-03 08:53:37.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"}],"messages":[{"id":"8efdc8157800e2a36bb95494ad48abac57e3cc49","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-02-03 08:51:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2a2061de04d10f2ed534695446a25f3445894515","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 08:52:54.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/openstack-ansible-os_keystone on branch stable/victoria.  The error was:\n\n  while parsing a block mapping\n  in \"openstack/openstack-ansible-os_keystone/zuul.d/project.yaml@stable/victoria\", line 25, column 7\ndid not find expected key\n  in \"openstack/openstack-ansible-os_keystone/zuul.d/project.yaml@stable/victoria\", line 26, column 9","accounts_in_message":[],"_revision_number":1},{"id":"8a96811c438f6e6410b814e8e36018b8dcf392ab","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-02-03 08:53:37.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"fcf20bea8798ed2ddd5869871502d5a0d5bdcfc5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-02-03 08:54:42.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"57b86ce3d08fb0ca1bc1332c261594b194f67b81","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-02-03 08:55:00.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"21dda0aed6e15fe9da2dc0a8e87d3577f7185977","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-02-03 09:52:21.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"34e5ec878c94ed7e82f6af8dab557690bbbff474","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-02-03 09:52:46.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"a20acad3228409a2e3e576110bf735f66506d044","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 11:43:19.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/c9fcba312eb64c569781f0eeadbb641e : SUCCESS in 16m 49s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/d7d12fed41774786b626240bd46423b4 : SUCCESS in 1h 26m 33s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/3b46ba3818c64657a8e77c6dddfe6438 : FAILURE in 27m 22s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/714865fd39c9482c9076da8d3b99eb70 : SUCCESS in 1h 21m 58s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0dfdb82ec13c4560ae797d6c6e67748e : SUCCESS in 1h 16m 53s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/44821fb3c17548bdb6e576acfe4ec797 : FAILURE in 29m 36s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6788232d443c4c3989e364ac8b347f52 : SUCCESS in 1h 46m 14s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/728e72b1f15246a4abb0b5b2ef84cef9 : FAILURE in 28m 22s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4a9603a62206499ca29c11d5c984686b : SUCCESS in 1h 10m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db3828f48af749c19ed3115d75c30f1d : SUCCESS in 6m 27s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/773b300eae594197b8a33f28834798cc : SUCCESS in 37m 29s\n- openstack-ansible-uw_apache-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0a21be510fe74d42a0fe5937b929bf1b : SUCCESS in 39m 16s","accounts_in_message":[],"_revision_number":4},{"id":"b17624ebd1553f2c3dcdffc00ed60ce80414b469","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-02-03 21:34:42.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d122bd573b3eacefcf4b7565c27cb65270c93547","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-02-03 21:35:25.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"dbff85a8745731fe71eb9524b2c65588241ebd8a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 23:17:36.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/8e3a21e905dd40c78473c580eca8c412 : SUCCESS in 12m 07s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/057fa92feda944fb8ae0753fad71865c : SUCCESS in 1h 12m 07s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7e135702656c4538885b2785c9844788 : SUCCESS in 1h 17m 03s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8ed5d01a09fb4daf8e5c2793932dd7ac : SUCCESS in 1h 12m 49s\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d66382dfa3eb4604bfc5574945cd0770 : SUCCESS in 1h 38m 38s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/96c151fb68084e66bfc676aa2b5b9926 : SUCCESS in 1h 02m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36cd7f9539ef470297856c756b119edd : SUCCESS in 5m 59s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/948dfb24231a41da93bad34670e013e8 : SUCCESS in 40m 58s\n- openstack-ansible-uw_apache-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3e58c0909ccb409b9976af446a71edfd : FAILURE in 19m 26s","accounts_in_message":[],"_revision_number":4},{"id":"df459c6a7bf7c0b46814fcdca38de12c97fada5c","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-02-04 08:25:14.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"38f001277ad38f7695a061a978f7b58306944c1b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 10:07:00.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/f02d8182985a4ddb81530272c36c3e6d : SUCCESS in 15m 28s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/5dc712a1a168429a91cefcbdd1de66ae : SUCCESS in 1h 25m 54s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1580aed8ee244c0797b4b90f6c105565 : SUCCESS in 1h 34m 15s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f92f6b611f1a41078e989cdeb75e298f : SUCCESS in 1h 12m 22s\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/97099bda1fc64c028c072cb0db9aedf0 : SUCCESS in 1h 32m 05s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/72222a67a3114123974dbc973ea91838 : SUCCESS in 59m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ed848182e2d4ee48ecfacd82bf20e78 : SUCCESS in 6m 34s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/15e013f91f524ac9a0741f62d3c70530 : SUCCESS in 42m 00s\n- openstack-ansible-uw_apache-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eb6d97512c434661bc3153fa4bda00c5 : FAILURE in 15m 42s","accounts_in_message":[],"_revision_number":4},{"id":"95be522666a46ab6f254848a694186435ae1bd2d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-02-04 11:36:00.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"f057302143cf8a82f8bf76587f12499d15215ba4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 14:10:47.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/95c67238a92c4c4799a5cab46a297148 : SUCCESS in 11m 04s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/26e1a3efd9194b23b126d177b3e3b8cb : SUCCESS in 1h 25m 05s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/86f42e59513240609e87967a9afdef10 : SUCCESS in 1h 30m 58s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5e8202bcf8c242bfbe1550e3f6513a64 : SUCCESS in 1h 20m 11s\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d610181faaf349cfabc9eb6a31758605 : SUCCESS in 1h 35m 34s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5c5b9883173646ab9781494ede7ac059 : SUCCESS in 1h 04m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e89a42ab06fb40c5b8c99f289b94ce46 : SUCCESS in 6m 24s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/35d77fa758064786ba6a73a007bf80d0 : SUCCESS in 40m 59s\n- openstack-ansible-uw_apache-ubuntu-focal https://zuul.opendev.org/t/openstack/build/36f3950d767447bca3380a4baae1561c : SUCCESS in 42m 25s","accounts_in_message":[],"_revision_number":4},{"id":"627595fdcd5ee818cf3d1b67a0902ccaaeee7c6d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-02-05 19:44:51.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"8209b1e9bd45c34def21c4a1df4497ea29112753","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-05 19:47:49.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"149397c81777e14c21dc0f4271beafed203a2fdf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-05 21:14:50.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/ee9820cb5d1443ce8bb9e269a3979739 : SUCCESS in 12m 19s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/a6459d0fd1eb4842a23b4d6361bc2bb2 : SUCCESS in 1h 26m 14s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8de6cf13e896498891b80237d7a58383 : SUCCESS in 1h 11m 39s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6e786d2b7bc94cb48f5487e2faa876a9 : SUCCESS in 1h 07m 13s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b6e31538725b4879aa6ae799830225cf : SUCCESS in 57m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eff74f929e9140d9bdcfc9bdeefbbd34 : SUCCESS in 6m 02s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8abc7f91fa5c4bf68377ab2f4eb117c5 : SUCCESS in 40m 41s\n- openstack-ansible-uw_apache-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f90211fed3b747fe8cf4b795043a3996 : SUCCESS in 39m 53s","accounts_in_message":[],"_revision_number":4},{"id":"f08166ba78210b240827871124e1118e3239419d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-05 21:14:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"3d16a9a68f18fdfbb2362425dac30f9189d6a033","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-05 21:18:22.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36ba569d622e41519fb532238c6e7b44 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"526a3b6e89f1d4eee9065517180d677cd5410d55","revisions":{"f7e34616577ec602e8cec06f2dfefeb33b78ab30":{"kind":"REWORK","_number":1,"created":"2022-02-03 08:51:56.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/03/827603/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/03/827603/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/03/827603/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/03/827603/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/03/827603/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/03/827603/1"}}},"commit":{"parents":[{"commit":"13126f463c81e3736d4693ffc723fb79da4707d6","subject":"Remove legacy centos-8 CI jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/13126f463c81e3736d4693ffc723fb79da4707d6"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-02-03 08:51:37.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-02-03 08:51:37.000000000","tz":0},"subject":"Restore CI jobs","message":"Restore CI jobs\n\nChange-Id: I1457d563abcbc9e62c331375ed677758b5f73927\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/f7e34616577ec602e8cec06f2dfefeb33b78ab30"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/f7e34616577ec602e8cec06f2dfefeb33b78ab30"}]},"parents_data":[{"branch_name":"refs/heads/stable/victoria","commit_id":"13126f463c81e3736d4693ffc723fb79da4707d6","is_merged_in_target_branch":false,"change_id":"Icb81d95dc11462f09d5603bfd94014f1ee3f5bf1","change_number":827400,"patch_set_number":2,"change_status":"MERGED"}],"branch":"refs/heads/stable/victoria"},"4b97ac3532535d26205017bbd8c973d39c633690":{"kind":"REWORK","_number":2,"created":"2022-02-03 08:54:42.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/03/827603/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/03/827603/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/03/827603/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/03/827603/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/03/827603/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/03/827603/2"}}},"commit":{"parents":[{"commit":"13126f463c81e3736d4693ffc723fb79da4707d6","subject":"Remove legacy centos-8 CI jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/13126f463c81e3736d4693ffc723fb79da4707d6"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-02-03 08:51:37.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-02-03 08:54:42.000000000","tz":0},"subject":"Restore CI jobs","message":"Restore CI jobs\n\nChange-Id: I1457d563abcbc9e62c331375ed677758b5f73927\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/4b97ac3532535d26205017bbd8c973d39c633690"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/4b97ac3532535d26205017bbd8c973d39c633690"}]},"parents_data":[{"branch_name":"refs/heads/stable/victoria","commit_id":"13126f463c81e3736d4693ffc723fb79da4707d6","is_merged_in_target_branch":false,"change_id":"Icb81d95dc11462f09d5603bfd94014f1ee3f5bf1","change_number":827400,"patch_set_number":2,"change_status":"MERGED"}],"branch":"refs/heads/stable/victoria"},"e7b572779cacb1313e481ac96107883f111f5460":{"kind":"REWORK","_number":3,"created":"2022-02-03 09:52:21.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/03/827603/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/03/827603/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/03/827603/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/03/827603/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/03/827603/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/03/827603/3"}}},"commit":{"parents":[{"commit":"42153ba0b1e462b74ef05820586dc460f419a149","subject":"Remove legacy centos-8 CI jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/42153ba0b1e462b74ef05820586dc460f419a149"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-02-03 08:51:37.000000000","tz":0},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-02-03 09:52:03.000000000","tz":120},"subject":"Restore CI jobs","message":"Restore CI jobs\n\nChange-Id: I1457d563abcbc9e62c331375ed677758b5f73927\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/e7b572779cacb1313e481ac96107883f111f5460"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/e7b572779cacb1313e481ac96107883f111f5460"}]},"parents_data":[{"branch_name":"refs/heads/stable/victoria","commit_id":"42153ba0b1e462b74ef05820586dc460f419a149","is_merged_in_target_branch":true,"change_id":"Icb81d95dc11462f09d5603bfd94014f1ee3f5bf1","change_number":827400,"patch_set_number":3,"change_status":"MERGED"}],"branch":"refs/heads/stable/victoria"},"526a3b6e89f1d4eee9065517180d677cd5410d55":{"kind":"REWORK","_number":4,"created":"2022-02-03 09:52:46.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/03/827603/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/03/827603/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/03/827603/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/03/827603/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/03/827603/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/03/827603/4"}}},"commit":{"parents":[{"commit":"42153ba0b1e462b74ef05820586dc460f419a149","subject":"Remove legacy centos-8 CI jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/42153ba0b1e462b74ef05820586dc460f419a149"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-02-03 08:51:37.000000000","tz":0},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2022-02-03 09:52:46.000000000","tz":0},"subject":"Restore CI jobs","message":"Restore CI jobs\n\nChange-Id: I1457d563abcbc9e62c331375ed677758b5f73927\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/526a3b6e89f1d4eee9065517180d677cd5410d55"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/526a3b6e89f1d4eee9065517180d677cd5410d55"}]},"parents_data":[{"branch_name":"refs/heads/stable/victoria","commit_id":"42153ba0b1e462b74ef05820586dc460f419a149","is_merged_in_target_branch":true,"change_id":"Icb81d95dc11462f09d5603bfd94014f1ee3f5bf1","change_number":827400,"patch_set_number":3,"change_status":"MERGED"}],"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
