)]}'
{"id":"openstack%2Fopenstack-ansible-os_keystone~843714","triplet_id":"openstack%2Fopenstack-ansible-os_keystone~master~Ia454d95a48dff1fa1856137df74a548d9c7d8a11","project":"openstack/openstack-ansible-os_keystone","branch":"master","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2022-05-29 16:14:59.000000000","reason":"Jonathan Rosser replied on the change"},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2023-02-21 20:57:59.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia454d95a48dff1fa1856137df74a548d9c7d8a11","subject":"Test multiple keystone containers for os_keystone tests","status":"MERGED","created":"2022-05-28 08:03:17.000000000","updated":"2023-02-21 20:59:59.000000000","submitted":"2023-02-21 20:57:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"843714","meta_rev_id":"2880ca119f8cff33de7c370a859cbeae8c373f9f","_number":843714,"virtual_id_number":843714,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-21 20:57:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"all":[{"value":2,"date":"2023-02-21 19:10:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":2,"date":"2023-02-07 08:52:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"all":[{"value":1,"date":"2023-02-21 19:10:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}],"CC":[{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-28 09:54:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-30 21:57:03.000000000","updated_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"reviewer":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"state":"CC"},{"updated":"2023-02-07 08:52:55.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2023-02-21 19:10:04.000000000","updated_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"reviewer":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"state":"REVIEWER"}],"messages":[{"id":"f563be5fb648f3b0eb235f57a1cc4f447dff9eb6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-05-28 08:03:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fdfdb8c9e8553ce92c6481a712a4be69ba24c605","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-05-28 08:05:50.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"ed17bba3c6636aec36ec142f331959e4aa92eeab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-05-28 08:35:23.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"228769cd096c8689ff32642a18dd6fe6dc3461dc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-05-28 08:36:37.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"a17467cdfe0de531985531aee473739628c8200d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-05-28 08:39:00.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"ba8e8012c196eb93c23667bf8e6e8e30f4cb41ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-28 09:54:49.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/48dea871e73b4b52950de16e5f8caec8 : SUCCESS in 13m 30s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/7136f6a98ef545e6973f1c724d192ce4 : RETRY_LIMIT in 12m 26s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/3cfc6ef5332848fe9ae37f44ffdadd20 : FAILURE in 53m 55s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/8d52d6609619484d8622c43c3e583129 : RETRY_LIMIT in 10m 38s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1deb64566c0a48a39736e448a9457172 : FAILURE in 54m 26s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/ba700c7c599540c6af880b96bfc0c0ed : RETRY_LIMIT in 10m 57s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/afcc272dab7e4a2a85cddeae4c2cec4f : SUCCESS in 49m 34s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/23b861ec841d4260b4d5700c655ba5a5 : FAILURE in 9m 35s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/29323578d1bb495cb9fa14b9cb28a86b : SUCCESS in 1h 11m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b48d8f951704d1fb72ffc36f18ee2e0 : SUCCESS in 5m 25s","accounts_in_message":[],"_revision_number":5},{"id":"070ad8ddac7a5034396ca83cd5e9413a2e5656fe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-05-29 09:49:43.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"d077c2d9589b8b958a606856082dd772c0f68ca0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-29 11:39:22.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/60eb2907c6ca480b919200b6d8a79753 : SUCCESS in 12m 34s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/54b80fac815345d3ba6ef399a3b30e9c : FAILURE in 59m 10s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/bfa88512df22435a9231fdb14cf655a0 : FAILURE in 1h 02m 01s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/c49ab07a3063468ba8caf5e1793634d2 : FAILURE in 1h 18m 44s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f1a168947d0c464988d2ff49157572dc : FAILURE in 1h 01m 26s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/b6a7fd1a6932461da77cb1bf09613425 : SUCCESS in 1h 00m 48s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aa178235f1474b409fd0cb551a95351f : SUCCESS in 54m 40s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/d9395015946a415785d9df6d48b82ddd : FAILURE in 1h 07m 37s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aebcd717039045f585896805caf90bd8 : SUCCESS in 1h 46m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b79b7dfb0764d7383e0fab2e0e37aec : SUCCESS in 5m 16s","accounts_in_message":[],"_revision_number":6},{"id":"26bbb5e65e339652bea638f721aa1eace77bddf7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-05-29 11:41:15.000000000","message":"Patch Set 7: Commit message was updated.","accounts_in_message":[],"_revision_number":7},{"id":"d78fbabc95836bcbb34e2d4d8a67925b64a7f079","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-29 12:11:27.000000000","message":"Patch Set 7: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":7},{"id":"da58c739f494895b48b5d4daa1e14ac02b0a60c0","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-05-29 16:14:59.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"64d5f459dc6bf19315551cdeee01820fe0fd2fdc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-29 18:18:19.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/32dafe8434234d438929ee077d777986 : SUCCESS in 13m 10s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/a7112f337b9647a3af600b234e4b9c52 : SUCCESS in 1h 36m 33s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/ad92a86aee2649a1950e868767383873 : SUCCESS in 1h 09m 32s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/888f4d865113473db8d322384133cc23 : SUCCESS in 1h 57m 23s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f90c952534cd4dc2a9bf632a22541eaf : SUCCESS in 1h 23m 03s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/62fb594f02554e09bd4174e4fbae1a3c : SUCCESS in 1h 08m 34s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a02d480f51c84178908f885a6524eb54 : SUCCESS in 1h 05m 53s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/ef8953550f604ed89e909423005cd1fc : FAILURE in 1h 39m 41s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a331f3c9610f477da0cad6cd01362180 : SUCCESS in 1h 38m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fef50bb1bdc14f61bf3e95cb7192c0bc : SUCCESS in 5m 54s","accounts_in_message":[],"_revision_number":7},{"id":"6c0f0bba3eedf46a492885c134610b11c683e12d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-05-30 15:14:11.000000000","message":"Patch Set 8: Commit message was updated.","accounts_in_message":[],"_revision_number":8},{"id":"da2d9e81b91bd1e14cec4d64875f15e7f555b826","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-30 17:01:23.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/68ad498f71524f6cb73e29d437308fd8 : SUCCESS in 14m 54s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/e0ff447b4dda48cc9b5678b1914c0fd8 : SUCCESS in 1h 42m 53s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/7778223694344446b7ffd174f9967a16 : SUCCESS in 1h 27m 17s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/885fc575b30c40c09056f6f2bc64a67b : SUCCESS in 1h 38m 36s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4d675d9ade6e45698faf9803185e4e6f : SUCCESS in 1h 43m 00s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/ba3e3e9167314a988b9b6af8d15d73dd : SUCCESS in 1h 07m 42s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3d0d20187d084e66944865eef6018c98 : SUCCESS in 1h 03m 52s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/caf7b0c4829b4942acd605ab3a95856b : SUCCESS in 1h 40m 32s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d5bd5941c0044085b2bf176dcb71860d : FAILURE in 1h 10m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/166cb451378c40a2a6fbf0138238a8f8 : SUCCESS in 5m 30s","accounts_in_message":[],"_revision_number":8},{"id":"cc85ce5b39f64568abfaff2336efac311c02878e","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2022-05-30 21:57:03.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"179293b4b7155714bff02a0fa0fe2ef7a6d1d1f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-30 23:53:00.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/95aa7835ba8a4fdcbda7bf9c37d1185f : SUCCESS in 12m 56s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/6ca731dece4541879f38275439b0d47f : SUCCESS in 1h 43m 40s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/5f2b75d382784997b846afb26a5f42a0 : FAILURE in 1h 03m 11s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/7c96b1d33742496e88c35e3c1ec08e22 : SUCCESS in 1h 19m 46s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/090f4026d05f4304a67c61a13205c664 : SUCCESS in 1h 30m 07s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fd1bee49bba4480dad4ea6c23cca50d2 : SUCCESS in 1h 25m 10s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/684213860d2d4b028a43f052678f66a1 : SUCCESS in 1h 09m 22s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4c78f314a1cc44b0bddedd0c7dbb86dc : SUCCESS in 1h 06m 45s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/c8a7ba801f6343f59b0741f263a6130e : SUCCESS in 1h 42m 45s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d47638d5477f4d6b869b3ef74908577c : SUCCESS in 1h 46m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ba6735db58b4b398889a7b6c1df6908 : SUCCESS in 4m 57s","accounts_in_message":[],"_revision_number":8},{"id":"a73cd147380c0e093c83e345872da5970c3b37c2","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2022-05-31 00:50:42.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"0fcd601d19c98d0717391defcb559f5b90660778","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-31 04:09:28.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/96b5e786528f44aca4566487f43a78e6 : SUCCESS in 15m 23s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/ea7ca909eb7a4324931c0557e80acbff : SUCCESS in 1h 39m 37s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/299ffc7411444da685024eb84e325756 : TIMED_OUT in 3h 14m 51s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/f0b0a1e01cb64e378804d4f6a082ca6e : SUCCESS in 1h 33m 47s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/73bb741f90ec4dfebb32179e397854d7 : SUCCESS in 1h 36m 54s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c7d3668c751345c59d0d6c2477612ef3 : SUCCESS in 1h 23m 37s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/377717ec9d1341ef9464f370b77a4054 : SUCCESS in 1h 03m 51s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2d2bc17aeef749d1ac650b1a8469a287 : SUCCESS in 1h 00m 19s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/1c0069dd63854d76b6e2559098b2ec54 : SUCCESS in 1h 36m 12s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e4f4726c15294f5bb3d6d4fbbf6f57d3 : SUCCESS in 1h 27m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eec77a96b98447b8946b74f7d0e63b09 : SUCCESS in 5m 36s","accounts_in_message":[],"_revision_number":8},{"id":"8c9d23a5419d950d332cd4c4f9990f34d5731dff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-02-01 16:13:45.000000000","message":"Patch Set 9: Published edit on patch set 8.","accounts_in_message":[],"_revision_number":9},{"id":"3bf36679ae85a17fc1a9993b50b27636f7af1d13","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-02-01 16:13:50.000000000","message":"Patch Set 10: Patch Set 9 was rebased","accounts_in_message":[],"_revision_number":10},{"id":"21f81c6d6967b6a310c86a3e5f2dba5c053134cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 18:47:33.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b0e2a3a899b4affb3e77e9563efeb4d\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/85ea3f77ae1a41d59876e525fc1e266a : SUCCESS in 12m 58s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/a2da8023a82f454c9bde2458a94a9b08 : SUCCESS in 1h 13m 46s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/52abd5707d934819accaf06d6635fdd1 : SUCCESS in 1h 25m 54s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e842445191ef413f9d85bce9c4e4071a : SUCCESS in 1h 23m 33s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/11fc2324acdf45d987b8df9ef6f176cb : SUCCESS in 1h 30m 10s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5c38d8e8b3b34006a270f64fb963b6e9 : SUCCESS in 1h 14m 56s\n- openstack-ansible-upgrade-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/ca4af6cae3b64bb3b868b5dee3c29b1b : SUCCESS in 2h 11m 37s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/01390e0b369e4684ae13b6064b8cdd6b : SUCCESS in 1h 56m 09s\n- openstack-ansible-deploy-aio_distro_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/e00d4531def44bf08d2bfd186f927268 : FAILURE in 39m 12s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bbe0012061d645ef8ed689f63930e4df : SUCCESS in 1h 03m 16s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/2c10a4847aef4b168b161cb4fa5ff408 : SUCCESS in 1h 25m 00s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/8c1dc7547fa641008fb74eba05138610 : SUCCESS in 1h 32m 53s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/8f04ce3cb76c46bca9e181a33fc14b4d : SUCCESS in 1h 10m 08s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3c7b53c83a1849d398361da3f54dfdba : SUCCESS in 1h 17m 21s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d8ef83281f824e88b006bc3b6924b5a2 : SUCCESS in 1h 16m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31ac6b4441354396b5b66078a1ac4369 : SUCCESS in 4m 21s","accounts_in_message":[],"_revision_number":10},{"id":"ad08aee50c9f5d47cd72fdf37003a26c4be21ae6","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-02-07 08:52:55.000000000","message":"Patch Set 10: Code-Review+2","accounts_in_message":[],"_revision_number":10},{"id":"eca230e6dc23fe3b2f5a47c98eadc568db28c11e","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2023-02-21 19:10:04.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"2ad64cdc0d92d1dbab0ec084fda27b176ae5f09d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 19:12:30.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"9912f7481c9677793e53413a6444e9f267518c53","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 20:57:59.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2a457d1bc71a4d3992e4b17617202f4f\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9c9ab465d7544555b60314e9bab57b04 : SUCCESS in 11m 56s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/9bbabdcc20424515baf0b51c1bc4c291 : SUCCESS in 1h 20m 21s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/e7a02f7a5d9c4274ae55409818f7fa41 : SUCCESS in 1h 30m 02s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e3540501d7e349bdab804b5c916f3f16 : SUCCESS in 1h 31m 25s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/019b55255b374c84b5948c3132dc9f8a : SUCCESS in 1h 25m 10s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/1d24861640fe4e7abebde21fd5d5349a : SUCCESS in 1h 28m 35s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/70d988a05eb54787bf344e58d2f76940 : SUCCESS in 1h 37m 18s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/e5a4231293884b4bb5c53734cef55a4a : SUCCESS in 1h 15m 09s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0deed6754c6342e09d6ae2ceba4bf740 : SUCCESS in 1h 19m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a97ec164cd1479ebbce644ad61fd4f6 : SUCCESS in 4m 53s","accounts_in_message":[],"_revision_number":10},{"id":"de47d11461c45c91f1326027ab9816afd40fe59f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 20:57:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":10},{"id":"2880ca119f8cff33de7c370a859cbeae8c373f9f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 20:59:59.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e23dbb29215a48f4a87db8b4b7ea5f7c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a314c74f8a514a6a8382d0bfc13a2823 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"a020ff87cde136a5c507b2cdc719d8c1dd85824d","revisions":{"21469c61daa7cee7bae95167f3ba5f47ae413874":{"kind":"REWORK","_number":1,"created":"2022-05-28 08:03:17.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/14/843714/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/14/843714/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/14/843714/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/14/843714/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/14/843714/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/14/843714/1"}}},"commit":{"parents":[{"commit":"a0c419c9b87961b5ea8cbd27f657075a46ce03a3","subject":"Merge \"Fix certificate installation for keystone\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/a0c419c9b87961b5ea8cbd27f657075a46ce03a3"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-05-28 08:03:25.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-05-28 08:03:25.000000000","tz":60},"subject":"DNM - test multiple keystone instances for os_keystone tests","message":"DNM - test multiple keystone instances for os_keystone tests\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/843713\nChange-Id: Ia454d95a48dff1fa1856137df74a548d9c7d8a11\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/21469c61daa7cee7bae95167f3ba5f47ae413874"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/21469c61daa7cee7bae95167f3ba5f47ae413874"}]},"branch":"refs/heads/master"},"b4d15a3b792cd4c855ccc86ac59b72a3a2f22e47":{"kind":"REWORK","_number":2,"created":"2022-05-28 08:05:50.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/14/843714/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/14/843714/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/14/843714/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/14/843714/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/14/843714/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/14/843714/2"}}},"commit":{"parents":[{"commit":"a0c419c9b87961b5ea8cbd27f657075a46ce03a3","subject":"Merge \"Fix certificate installation for keystone\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/a0c419c9b87961b5ea8cbd27f657075a46ce03a3"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-05-28 08:03:25.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2022-05-28 08:05:50.000000000","tz":0},"subject":"DNM - test multiple keystone instances for os_keystone tests","message":"DNM - test multiple keystone instances for os_keystone tests\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/843713\nChange-Id: Ia454d95a48dff1fa1856137df74a548d9c7d8a11\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/b4d15a3b792cd4c855ccc86ac59b72a3a2f22e47"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/b4d15a3b792cd4c855ccc86ac59b72a3a2f22e47"}]},"branch":"refs/heads/master"},"a6bba9e73e3c6df9b808ed87446db4e2e40b4aab":{"kind":"REWORK","_number":3,"created":"2022-05-28 08:35:23.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/14/843714/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/14/843714/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/14/843714/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/14/843714/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/14/843714/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/14/843714/3"}}},"commit":{"parents":[{"commit":"a0c419c9b87961b5ea8cbd27f657075a46ce03a3","subject":"Merge \"Fix certificate installation for keystone\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/a0c419c9b87961b5ea8cbd27f657075a46ce03a3"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-05-28 08:03:25.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2022-05-28 08:35:23.000000000","tz":0},"subject":"DNM - test multiple keystone instances for os_keystone tests","message":"DNM - test multiple keystone instances for os_keystone tests\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/843713\nChange-Id: Ia454d95a48dff1fa1856137df74a548d9c7d8a11\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/a6bba9e73e3c6df9b808ed87446db4e2e40b4aab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/a6bba9e73e3c6df9b808ed87446db4e2e40b4aab"}]},"branch":"refs/heads/master"},"44d44ee5a959f32e84cc72efa9742e11719cd4f6":{"kind":"REWORK","_number":4,"created":"2022-05-28 08:36:37.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/14/843714/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/14/843714/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/14/843714/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/14/843714/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/14/843714/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/14/843714/4"}}},"commit":{"parents":[{"commit":"a0c419c9b87961b5ea8cbd27f657075a46ce03a3","subject":"Merge \"Fix certificate installation for keystone\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/a0c419c9b87961b5ea8cbd27f657075a46ce03a3"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-05-28 08:03:25.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2022-05-28 08:36:37.000000000","tz":0},"subject":"DNM - test multiple keystone instances for os_keystone tests","message":"DNM - test multiple keystone instances for os_keystone tests\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/843713\nChange-Id: Ia454d95a48dff1fa1856137df74a548d9c7d8a11\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/44d44ee5a959f32e84cc72efa9742e11719cd4f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/44d44ee5a959f32e84cc72efa9742e11719cd4f6"}]},"branch":"refs/heads/master"},"0f25e777a296379a7ae31bf13b212b745db9c9d6":{"kind":"REWORK","_number":5,"created":"2022-05-28 08:39:00.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/14/843714/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/14/843714/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/14/843714/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/14/843714/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/14/843714/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/14/843714/5"}}},"commit":{"parents":[{"commit":"a0c419c9b87961b5ea8cbd27f657075a46ce03a3","subject":"Merge \"Fix certificate installation for keystone\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/a0c419c9b87961b5ea8cbd27f657075a46ce03a3"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-05-28 08:03:25.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2022-05-28 08:39:00.000000000","tz":0},"subject":"DNM - test multiple keystone instances for os_keystone tests","message":"DNM - test multiple keystone instances for os_keystone tests\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/843713\nChange-Id: Ia454d95a48dff1fa1856137df74a548d9c7d8a11\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/0f25e777a296379a7ae31bf13b212b745db9c9d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/0f25e777a296379a7ae31bf13b212b745db9c9d6"}]},"branch":"refs/heads/master"},"51503cbe5449559da7e054bb31706a540d09b253":{"kind":"REWORK","_number":6,"created":"2022-05-29 09:49:43.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/14/843714/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/14/843714/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/14/843714/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/14/843714/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/14/843714/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/14/843714/6"}}},"commit":{"parents":[{"commit":"a0c419c9b87961b5ea8cbd27f657075a46ce03a3","subject":"Merge \"Fix certificate installation for keystone\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/a0c419c9b87961b5ea8cbd27f657075a46ce03a3"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-05-28 08:03:25.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2022-05-29 09:49:40.000000000","tz":0},"subject":"DNM - test multiple keystone instances for os_keystone tests","message":"DNM - test multiple keystone instances for os_keystone tests\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/843713\nChange-Id: Ia454d95a48dff1fa1856137df74a548d9c7d8a11\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/51503cbe5449559da7e054bb31706a540d09b253"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/51503cbe5449559da7e054bb31706a540d09b253"}]},"branch":"refs/heads/master"},"38189e053e9a02306291a9be0f6dfeae4a7280a5":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2022-05-29 11:41:15.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/14/843714/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/14/843714/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/14/843714/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/14/843714/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/14/843714/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/14/843714/7"}}},"commit":{"parents":[{"commit":"a0c419c9b87961b5ea8cbd27f657075a46ce03a3","subject":"Merge \"Fix certificate installation for keystone\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/a0c419c9b87961b5ea8cbd27f657075a46ce03a3"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-05-28 08:03:25.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-05-29 11:41:15.000000000","tz":0},"subject":"DNM - test multiple keystone instances for os_keystone tests","message":"DNM - test multiple keystone instances for os_keystone tests\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/843713\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/843740\nChange-Id: Ia454d95a48dff1fa1856137df74a548d9c7d8a11\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/38189e053e9a02306291a9be0f6dfeae4a7280a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/38189e053e9a02306291a9be0f6dfeae4a7280a5"}]},"branch":"refs/heads/master","description":"Edit commit message"},"99b165b5526e5a0a8a38f8193c5c61f2e37a5d1e":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2022-05-30 15:14:11.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/14/843714/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/14/843714/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/14/843714/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/14/843714/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/14/843714/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/14/843714/8"}}},"commit":{"parents":[{"commit":"a0c419c9b87961b5ea8cbd27f657075a46ce03a3","subject":"Merge \"Fix certificate installation for keystone\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/a0c419c9b87961b5ea8cbd27f657075a46ce03a3"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-05-28 08:03:25.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-05-30 15:14:11.000000000","tz":0},"subject":"DNM - test multiple keystone instances for os_keystone tests","message":"DNM - test multiple keystone instances for os_keystone tests\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/843713\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/843740\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-repo_server/+/843931\nChange-Id: Ia454d95a48dff1fa1856137df74a548d9c7d8a11\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/99b165b5526e5a0a8a38f8193c5c61f2e37a5d1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/99b165b5526e5a0a8a38f8193c5c61f2e37a5d1e"}]},"branch":"refs/heads/master","description":"Edit commit message"},"5e43e7ecefaab0b0938483c6eafc5c4533415aac":{"kind":"REWORK","_number":9,"created":"2023-02-01 16:13:45.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/14/843714/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/14/843714/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/14/843714/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/14/843714/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/14/843714/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/14/843714/9"}}},"commit":{"parents":[{"commit":"a0c419c9b87961b5ea8cbd27f657075a46ce03a3","subject":"Merge \"Fix certificate installation for keystone\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/a0c419c9b87961b5ea8cbd27f657075a46ce03a3"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-05-28 08:03:25.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2023-02-01 16:13:44.000000000","tz":0},"subject":"Test multiple keystone containers for os_keystone tests","message":"Test multiple keystone containers for os_keystone tests\n\nKeystone has particular ordering requirements for setting up\nmultiple instances and distributing fernet keys.\n\nRun the infra jobs for the os_keystone role as these test\nthree keystone containers simultaneously.\n\nChange-Id: Ia454d95a48dff1fa1856137df74a548d9c7d8a11\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/5e43e7ecefaab0b0938483c6eafc5c4533415aac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/5e43e7ecefaab0b0938483c6eafc5c4533415aac"}]},"branch":"refs/heads/master"},"a020ff87cde136a5c507b2cdc719d8c1dd85824d":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2023-02-01 16:13:50.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/14/843714/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/14/843714/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/14/843714/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/14/843714/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/14/843714/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/14/843714/10"}}},"commit":{"parents":[{"commit":"d83b32adca48f37b1020d23cb6a48ace15833c6b","subject":"Update master for stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/d83b32adca48f37b1020d23cb6a48ace15833c6b"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-05-28 08:03:25.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2023-02-01 16:13:50.000000000","tz":0},"subject":"Test multiple keystone containers for os_keystone tests","message":"Test multiple keystone containers for os_keystone tests\n\nKeystone has particular ordering requirements for setting up\nmultiple instances and distributing fernet keys.\n\nRun the infra jobs for the os_keystone role as these test\nthree keystone containers simultaneously.\n\nChange-Id: Ia454d95a48dff1fa1856137df74a548d9c7d8a11\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/a020ff87cde136a5c507b2cdc719d8c1dd85824d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/a020ff87cde136a5c507b2cdc719d8c1dd85824d"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
