)]}'
{"id":"openstack%2Fopenstack-ansible-os_keystone~859054","triplet_id":"openstack%2Fopenstack-ansible-os_keystone~master~I2810208301fb31eeeabf31e4b38add7f8aa3e00b","project":"openstack/openstack-ansible-os_keystone","branch":"master","topic":"osa/cache_backends","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2022-10-05 12:44:25.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2810208301fb31eeeabf31e4b38add7f8aa3e00b","subject":"Improve way of cache backend selection","status":"MERGED","created":"2022-09-23 08:49:22.000000000","updated":"2022-10-05 12:45:31.000000000","submitted":"2022-10-05 12:44:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"859054-osa/cache_backends","meta_rev_id":"d81882bd829ee8476276d0e39460c53814dd0f08","_number":859054,"virtual_id_number":859054,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-05 12:44:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"all":[{"value":2,"date":"2022-10-05 11:11:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-09-27 20:13:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"all":[{"value":1,"date":"2022-10-05 11:11:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-23 10:11:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-27 20:13:24.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2022-10-05 11:11:21.000000000","updated_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"reviewer":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"state":"REVIEWER"}],"messages":[{"id":"425f5fb11a8057678ecbf024f3853cb9da5af8f3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-09-23 08:49:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f0b275d9a34636e84b643098e3f475c84a2a5e06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-23 10:11:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9fce2e4843fb43ceb9a11d4266ae1a79\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bddadc5dc96543cf915d0ef55945c0fa : SUCCESS in 18m 46s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/5ae2b85baa7247c082fe693e31918370 : FAILURE in 36m 49s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/d727ba2f6322446b8b23dbc71c9dc528 : FAILURE in 40m 52s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2750181868e4449c8401dc221b81f524 : FAILURE in 39m 25s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ae1eeea5c5504e37a68fddf2c813093c : FAILURE in 39m 48s (non-voting)\n- openstack-ansible-upgrade-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/6f9f89a842d546289e25716a9ee2b655 : FAILURE in 1h 18m 20s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6a3d0b623b7c44c493a17d9440918306 : FAILURE in 1h 12m 11s\n- openstack-ansible-deploy-aio_distro_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/2288007b87e3434e9393c19dd28d4269 : FAILURE in 31m 32s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/40d37a1f3ac94ef4944f5b5a6ff4452f : FAILURE in 28m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eda9d73833d64e23abfb8370379f6ce2 : SUCCESS in 6m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2a18159ebff4410e84021c79decdf0b1 : SUCCESS in 4m 43s","accounts_in_message":[],"_revision_number":1},{"id":"47b3f2c940f5a73e7d381de754f7b071c9afe9c2","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-09-24 15:20:49.000000000","message":"Patch Set 1:\n\nrecheck - gates seems to be fixed now","accounts_in_message":[],"_revision_number":1},{"id":"890c34dcf77122f441b5defdee5566ab5bf614d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-24 17:21:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/591dbdf3c5904dbfb3df7477fb971184\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1bb6d2caabee4921809b664513aa629f : SUCCESS in 17m 59s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/ff80c999954945c1b5dba8c4daf8eb22 : SUCCESS in 1h 39m 22s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/753ddd24dfda40f99f95fd5170c7663a : SUCCESS in 1h 18m 51s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/93d741c22f8d4ce8a241a1a443053b5d : SUCCESS in 1h 21m 25s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d8935d57dd4e41989ac638d163076181 : SUCCESS in 1h 20m 10s (non-voting)\n- openstack-ansible-upgrade-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/6c9b26a0d2fc4dd6bba525fb218a51a2 : SUCCESS in 1h 44m 39s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b8ac0719657c4e45987e1e84d8168104 : SUCCESS in 1h 53m 49s\n- openstack-ansible-deploy-aio_distro_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/f39988f6a4f145a09b6a5dd9b1448c5e : FAILURE in 33m 33s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/df5576354ebf48c894a0ddaa9a8178e5 : FAILURE in 27m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/026cb610a9914fe1911c0d331b025541 : SUCCESS in 4m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b35a7b39eb8d429aa6f01ac006ff5cb2 : SUCCESS in 4m 03s","accounts_in_message":[],"_revision_number":1},{"id":"493a37ddf1963ed0bbfd4116bb64db07734685fb","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-09-27 20:13:24.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"208a57c46cd481fb0cc11594044377fe51042adc","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2022-10-05 11:11:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9bdfd2c0f07ee0153a447b1192581f8c29ba17ee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-05 11:13:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9dd0dbf278ce79991436d03df353cce035863d20","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-05 12:44:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ae8c60ec9d7b45cf9b7bcef04810f5be\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/584ea1afdd2a43e1ba64c93977d54e6d : SUCCESS in 14m 11s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/41aa308110084d44bc9b9b8f39c63de4 : SUCCESS in 1h 12m 57s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/b3c6ac43197e45c2a2550d3ae5ff5e84 : SUCCESS in 1h 25m 01s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c50cd1cb084f487eaf9ef118e485ce14 : SUCCESS in 1h 24m 38s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a51e89effe8f4cd0942ef92dc03cd72c : SUCCESS in 1h 30m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8ee13b647774ae3bb7fcc9d263f2509 : SUCCESS in 6m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/08f790aaa804449ab9d47238ab84c4cc : SUCCESS in 4m 30s","accounts_in_message":[],"_revision_number":1},{"id":"959c3e96b6e619208015fa83305aeba6730dc830","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-05 12:44:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d81882bd829ee8476276d0e39460c53814dd0f08","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-05 12:45:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c751ae8c342c493dbbe33999872d140a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8824866eec9476fb1237c70cbac937d : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0a24c61e3e65e126691577c9b2843a7d9f448301","revisions":{"0a24c61e3e65e126691577c9b2843a7d9f448301":{"kind":"REWORK","_number":1,"created":"2022-09-23 08:49:22.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/54/859054/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/54/859054/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/54/859054/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/54/859054/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/54/859054/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/54/859054/1"}}},"commit":{"parents":[{"commit":"3b4fc2e9aa7552fb9e78a52de4a30861099a024e","subject":"Merge \"Add the option to deploy keystone without apache\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/3b4fc2e9aa7552fb9e78a52de4a30861099a024e"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-09-23 08:49:09.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-09-23 08:49:09.000000000","tz":120},"subject":"Improve way of cache backend selection","message":"Improve way of cache backend selection\n\nAt the moment we don\u0027t provide any option rather then use memcached\nbackend. With that we also hardocde list of packages that should be\ninstalled inside virtualenv for selected backend.\n\nAdding bmemcached requirement to oslo_cache.memcache_pool [1] gives us\nopportunity to refactor this bit of deployment and allow to be more\nflexible in backend selection and requirements installation for it.\n\n[1] https://review.opendev.org/c/openstack/oslo.cache/+/854628\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/858981/4\nChange-Id: I2810208301fb31eeeabf31e4b38add7f8aa3e00b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/0a24c61e3e65e126691577c9b2843a7d9f448301"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/0a24c61e3e65e126691577c9b2843a7d9f448301"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
