)]}'
{"id":"openstack%2Fopenstack-ansible-os_keystone~867453","triplet_id":"openstack%2Fopenstack-ansible-os_keystone~master~I83fbde781bbedd6e84f2ff1b1136b4558bf1da00","project":"openstack/openstack-ansible-os_keystone","branch":"master","topic":"reno-zed","hashtags":[],"change_id":"I83fbde781bbedd6e84f2ff1b1136b4558bf1da00","subject":"Update master for stable/zed","status":"MERGED","created":"2022-12-13 13:16:00.000000000","updated":"2022-12-13 14:02:15.000000000","submitted":"2022-12-13 14:01:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"867453-reno-zed","meta_rev_id":"d976ef43b6ed926c19e407ae27e07e23c441f8b1","_number":867453,"virtual_id_number":867453,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-13 14:01:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-12-13 13:47:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-12-13 13:47:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-13 13:36:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-13 13:47:01.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"35d6df59763481cce89e11493535c039fb3ea524","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2022-12-13 13:16:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2ae679e8ff5b6b5f533357976e5faff66935e6e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-13 13:36:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a826e456208465f9f52122d54d60bee\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b0d7b7610bb84fccb4ab3881e1ce321e : SUCCESS in 14m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a88706ba14d4a3c88e43a8b7777cf23 : SUCCESS in 7m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/15fbd1a321c74d24b6c984d7179ce2ba : SUCCESS in 4m 08s","accounts_in_message":[],"_revision_number":1},{"id":"1cd0ab52514dd35b492149186b1ad9b20d8157ee","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-12-13 13:47:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e8df169b1bb151a04af52f00ffe3d9be689545b9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-13 13:49:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"71aaf6c5aaae395918c5c819e4802451d94c143a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-13 14:01:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac27546020b44b25b2f218e921d3660a\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5f8ba4e50efb42bb8acccbf31585c02c : SUCCESS in 11m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03ef008f941c413699b8e2f035b041b5 : SUCCESS in 5m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a85bc2821190461d9d64fc53ae0be50a : SUCCESS in 2m 52s","accounts_in_message":[],"_revision_number":1},{"id":"af6376e79538c79705d155aa45b4ca666fe71e8d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-13 14:01:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d976ef43b6ed926c19e407ae27e07e23c441f8b1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-13 14:02:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/919a1d397824408bb4e874a2a6165f19\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/797070dbd51b46ce80bbdf1aaffcbef8 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d83b32adca48f37b1020d23cb6a48ace15833c6b","revisions":{"d83b32adca48f37b1020d23cb6a48ace15833c6b":{"kind":"REWORK","_number":1,"created":"2022-12-13 13:16:00.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/53/867453/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/53/867453/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/53/867453/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/53/867453/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/53/867453/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/53/867453/1"}}},"commit":{"parents":[{"commit":"8017d4dd84a7b1069fff7bf6f5b3ae27c31590cd","subject":"Define venv_tag as separate task for distro","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/8017d4dd84a7b1069fff7bf6f5b3ae27c31590cd"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2022-12-13 13:15:56.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2022-12-13 13:15:56.000000000","tz":0},"subject":"Update master for stable/zed","message":"Update master for stable/zed\n\nAdd file to the reno documentation build to show release notes for\nstable/zed.\n\nUse pbr instruction to increment the minor version number\nautomatically so that master versions are higher than the versions on\nstable/zed.\n\nSem-Ver: feature\nChange-Id: I83fbde781bbedd6e84f2ff1b1136b4558bf1da00\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/d83b32adca48f37b1020d23cb6a48ace15833c6b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/d83b32adca48f37b1020d23cb6a48ace15833c6b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
