)]}'
{"id":"openstack%2Fopenstack-ansible-os_keystone~871190","triplet_id":"openstack%2Fopenstack-ansible-os_keystone~master~Id5b626ad1725899779f35fb51f56f72f88dc3157","project":"openstack/openstack-ansible-os_keystone","branch":"master","topic":"separated-haproxy-service-config","attention_set":{},"removed_from_attention_set":{"32666":{"account":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"last_update":"2023-02-06 23:26:04.000000000","reason":"\u003cGERRIT_ACCOUNT_32666\u003e replied on the change","reason_account":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2023-02-06 23:26:55.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Id5b626ad1725899779f35fb51f56f72f88dc3157","subject":"Enable separated haproxy config","status":"ABANDONED","created":"2023-01-19 21:47:16.000000000","updated":"2023-02-06 23:26:55.000000000","total_comment_count":5,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"6b99406e131f278a0e525ae6ecf1db7234596442","_number":871190,"virtual_id_number":871190,"owner":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2023-02-02 10:29:03.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2023-02-03 19:06:36.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-19 23:31:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-03 19:06:36.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"8bd57b82918d39467470f39fae45c6457b88d45f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2023-01-19 21:47:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"986b1c72136017d4dc52209364fb066591892bb6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-19 23:31:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0c65535b2681495aa7d903dafc23a682\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ce7bb4b6142641098706b323a763c5fe : SUCCESS in 11m 40s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/7e54aa0ad3de4b4ebfeccbacc7f63d7a : FAILURE in 27m 38s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/f9e84d726ff045d0a3091b1c379698bc : FAILURE in 29m 17s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f1a7d39850464c7e8ec1394fad7989a1 : FAILURE in 30m 46s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/22c14dbb8e134722b8815ab746310976 : FAILURE in 26m 52s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6ea1e934bc8a40df98bed8170e1ece25 : FAILURE in 27m 29s\n- openstack-ansible-upgrade-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/615d2aa3da6349978060e1d86bb37490 : FAILURE in 1h 16m 49s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a840170cddf24927943fd0a7596dad3f : FAILURE in 1h 10m 17s\n- openstack-ansible-deploy-aio_distro_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/ab84082b5c5a4eb3b00546294e5ff0f2 : FAILURE in 24m 28s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5be5b2ae035c4b6e99f3c5e2592121e2 : FAILURE in 22m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf27afbd9b9442598600635a6da7d0ec : SUCCESS in 5m 01s","accounts_in_message":[],"_revision_number":1},{"id":"e23c40b5bfea9c588e0dc9f5236ddf5fa40876f6","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2023-02-01 18:46:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"870891d926afba450b19d6e5960ebe88e16efc82","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 21:05:36.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a66c7f341e2b440a89f51a184ae17cf9\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b2658a1129a54074afc0c44b697bc20a : SUCCESS in 14m 45s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/d1f1cb9e322249e983b94d09815300b7 : SUCCESS in 1h 20m 56s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/677e222892b241d2b3a14bbe1a238911 : SUCCESS in 1h 22m 09s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5a3ad8070cdb446784ccda47e6ec0a22 : SUCCESS in 1h 22m 54s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3ff1226225e342dcb7a60cdadb84587a : SUCCESS in 1h 33m 48s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1d4837199ec14225a54d588e9c0b947b : SUCCESS in 1h 09m 43s\n- openstack-ansible-upgrade-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/70628f2bf6e2422c92f060159a8477a5 : FAILURE in 1h 23m 02s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b6f2ce1166db499d98d73ffcf95be7b0 : FAILURE in 2h 09m 08s\n- openstack-ansible-deploy-aio_distro_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/c6e6ad1ca5044a5d9993119974f2de7d : FAILURE in 51m 35s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0e3afa618660474b8df6a45c6758a7b9 : SUCCESS in 1h 07m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b47a0bb91f1b44d0ba564afb53782699 : SUCCESS in 5m 32s","accounts_in_message":[],"_revision_number":1},{"id":"bea6f6420b539f525b305709ae3261de700c530a","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2023-02-01 22:01:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e17542a0726bdf9c95e5e4c0f72d3f7185bb2b11","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-02 00:06:41.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cce4d2d1f74c4db9b0a2df79c4b2691a\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fa97fcb50e9942028fbb029ae9cdace5 : SUCCESS in 12m 47s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/35ac31197cde486e8dc41ba22709d37f : SUCCESS in 1h 28m 30s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/d087c5383e4a42c7b2db82f4075c234f : SUCCESS in 1h 27m 20s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/364b8b00a1a64f11b69e67f87fa7bf55 : SUCCESS in 1h 23m 51s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ee5d35efc6b84636a19f554c035f82d9 : SUCCESS in 1h 23m 58s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cd4170d4978b40368439884982e3ebe2 : FAILURE in 1h 18m 43s\n- openstack-ansible-upgrade-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/9fd130267b0643f8a07ab83c27b9788c : SUCCESS in 1h 42m 46s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fc5c8e7b1316437d9d2d1c0c3f7d7b59 : SUCCESS in 2h 00m 03s\n- openstack-ansible-deploy-aio_distro_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/4e6d1f16925343558cabbe697a18713a : FAILURE in 46m 35s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1e31dcc53f5f4844b124bd7aec1c812f : SUCCESS in 1h 04m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1cf586b572b14e478ecf1e4334866131 : SUCCESS in 5m 29s","accounts_in_message":[],"_revision_number":1},{"id":"ee35e684d5de36c53dcb34dfc3ef37f3bc844700","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2023-02-02 08:31:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"aba113e0d5abea54ff3cdcb7904e498acc4ffbcc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-02 10:29:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eae5710868f94ce9947ee91635fccd33\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b53fdade77d248ebb8a6394462fdc3aa : SUCCESS in 13m 02s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/5a27dccd7cbb453d90c0a77fb1088d33 : SUCCESS in 1h 05m 40s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/78a96d48c8d44884b82b32e8936badcb : SUCCESS in 1h 14m 36s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/49c019a13343432e8ff9510bca154984 : SUCCESS in 1h 20m 49s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/16ff31b439ba423aa5f962e422c348d7 : SUCCESS in 1h 22m 57s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/db40029b7fe340de83738cad717bd5f5 : SUCCESS in 1h 13m 39s\n- openstack-ansible-upgrade-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/874c4026f78244a882351d5bcfa14cba : SUCCESS in 1h 46m 27s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c30714a0579349bd84d59951f5dfcb11 : SUCCESS in 1h 51m 11s\n- openstack-ansible-deploy-aio_distro_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/4f0d21110ec44a669e41113fe04ed79e : FAILURE in 41m 17s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1f8de3639dc94b85bdca280bf9ddc86e : SUCCESS in 1h 03m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0c8754408f447b8817767eab372c868 : SUCCESS in 6m 08s","accounts_in_message":[],"_revision_number":1},{"id":"6353cc9d8325059170629f2eb7cf6e4a9181e07e","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-02-03 19:06:36.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5547c8857f54ff190947f209b22baa0093d0f2c3","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2023-02-06 23:26:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6b99406e131f278a0e525ae6ecf1db7234596442","tag":"autogenerated:gerrit:abandon","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2023-02-06 23:26:55.000000000","message":"Abandoned\n\nnot needed since haproxy_services definition was moved to openstack-ansible repo(inventory/group_vars/\u003cgroup_name\u003e.yml)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"45fb346f1b111f5ab5e57740bd5423fc9ccaa955","revisions":{"45fb346f1b111f5ab5e57740bd5423fc9ccaa955":{"kind":"REWORK","_number":1,"created":"2023-01-19 21:47:16.000000000","uploader":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"ref":"refs/changes/90/871190/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/90/871190/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/90/871190/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/90/871190/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/90/871190/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/90/871190/1"}}},"commit":{"parents":[{"commit":"d83b32adca48f37b1020d23cb6a48ace15833c6b","subject":"Update master for stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/d83b32adca48f37b1020d23cb6a48ace15833c6b"}]}],"author":{"name":"Damian Dabrowski","email":"damian.dabrowski@cleura.com","date":"2023-01-19 21:46:50.000000000","tz":60},"committer":{"name":"Damian Dabrowski","email":"damian.dabrowski@cleura.com","date":"2023-01-19 21:46:50.000000000","tz":60},"subject":"Enable separated haproxy config","message":"Enable separated haproxy config\n\nDefine default haproxy service(s) required for openstack deployment.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/871189\nChange-Id: Id5b626ad1725899779f35fb51f56f72f88dc3157\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/45fb346f1b111f5ab5e57740bd5423fc9ccaa955"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/45fb346f1b111f5ab5e57740bd5423fc9ccaa955"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
