)]}'
{"id":"openstack%2Fopenstack-ansible-os_keystone~879379","triplet_id":"openstack%2Fopenstack-ansible-os_keystone~master~Ibf8607a4cf62ab518a09d64b1054ff7fbc580000","project":"openstack/openstack-ansible-os_keystone","branch":"master","topic":"tls-backend","attention_set":{},"removed_from_attention_set":{"32666":{"account":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"last_update":"2023-04-20 18:46:50.000000000","reason":"Change was submitted"},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2023-04-14 13:24:02.000000000","reason":"\u003cGERRIT_ACCOUNT_25023\u003e replied on the change","reason_account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}}},"hashtags":[],"change_id":"Ibf8607a4cf62ab518a09d64b1054ff7fbc580000","subject":"Rename keystone_ssl to keystone_backend_ssl","status":"MERGED","created":"2023-04-03 20:35:23.000000000","updated":"2023-04-20 18:47:49.000000000","submitted":"2023-04-20 18:46:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"879379-tls-backend","meta_rev_id":"3a88cdd5006e4502f3e7557af4f2fdbf5d3ac638","_number":879379,"virtual_id_number":879379,"owner":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-20 18:46:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2023-04-18 08:25:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-04-14 13:24:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2023-04-18 08:25:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-03 23:46:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-04 17:45:55.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2023-04-18 08:25:17.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"}],"messages":[{"id":"1f631dbe858ba8e1de701412a5da4d3982851318","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2023-04-03 20:35:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5aa971b330b35b22a360d4847fa3ef35de09dceb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-03 23:46:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e1e9c1fad2b141baaeb3ba01db0dd9cb\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0fab4c663cc14670bc3344ce08fe86b9 : SUCCESS in 11m 35s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/76e410b5b9f6467384776ee1a483f17f : SUCCESS in 1h 49m 38s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/0cbc2e1b824c4a3f8e18f2cc8c894b83 : SUCCESS in 1h 38m 05s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/73bbe2e6c51b4a5fba47e4a081329c89 : SUCCESS in 1h 28m 50s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0914f26010574c14b87fbbb1475f5947 : SUCCESS in 1h 29m 23s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6ec79e715f134cb4ab4e66ae079cdd26 : SUCCESS in 1h 12m 38s\n- openstack-ansible-upgrade-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/c90ad2b7862d4bc796ec101bfe35ffcb : TIMED_OUT in 3h 04m 24s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c209d6c19c3847a68c0ac4adecf44ab0 : SUCCESS in 2h 09m 49s\n- openstack-ansible-deploy-aio_distro_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/1035c1d35d8344e098ce6f28b4b98571 : FAILURE in 24m 00s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cce023ea0ae24d4e906aa781d7149b0a : FAILURE in 20m 12s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/72af5b80d41341f1905c9530558d63f0 : FAILURE in 37m 09s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/71c7f91fa8c74370b9c79fdbb78ba4ee : SUCCESS in 1h 13m 11s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/9210c8c57e964f04a8c969d61f94eced : SUCCESS in 1h 14m 44s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/caec3d652ba84eaf8e1c31f45fa41072 : SUCCESS in 1h 11m 31s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4684138b11904d4b93bfdcff2bf7e00a : SUCCESS in 1h 22m 34s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cdb758be749741b3a2733d3d82c89bec : SUCCESS in 1h 16m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4824a0448214312ab375199ed02e0ff : SUCCESS in 5m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7b402c0d9e9b48918cd5173bf647397a : SUCCESS in 3m 09s","accounts_in_message":[],"_revision_number":1},{"id":"a85d56a5911845ef4a53d870489a1a99a5992fec","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-04-04 17:45:55.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"60b94f1df516dfc3562ee0d4c95ce98e6cbd3b97","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2023-04-07 23:38:26.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"601b193a91d393d1105284663afd9465184a1063","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2023-04-07 23:40:43.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"99d03db426a143a011aac26492763ed2f80ea2ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-08 02:09:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/225af54487eb42b0bd99eddec7353d83\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9d63c175d04b4170a68d4479c98fc4a7 : SUCCESS in 11m 15s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/23727d567e154f769ebc7764dc636ad6 : SUCCESS in 1h 08m 10s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/ca25b3e7f68b448babdec1e22acaa1b7 : SUCCESS in 1h 19m 53s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/cf27a3f7b5104e6bbe2e9686c534acad : SUCCESS in 1h 36m 15s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/340d76428fd94eda98ffcf07d0aa1150 : SUCCESS in 1h 26m 17s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2bab9ed3c4cc4f8f9921dbd25b0336bb : SUCCESS in 1h 18m 39s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/45432fe081da4f8a8a1a82d06ad0c405 : SUCCESS in 2h 16m 50s\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d13f1b0d4aa5425d8a0b7be68500991d : SUCCESS in 2h 23m 40s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/faa8e28ce46141f9a9b19d660c309ed5 : SUCCESS in 1h 05m 19s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/40e466d17e4c490a8de09a4b5ef34f91 : SUCCESS in 57m 41s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/ee404ab27326492e931b6c9b5758e6f6 : FAILURE in 18m 55s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9304a6eccfa0435daeeb000d6a9fed86 : SUCCESS in 1h 27m 44s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/4c15a23791284ee7bb5a2efe68c14cbb : SUCCESS in 1h 00m 30s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3333fbd293e54775b3364e6a4f8a712a : SUCCESS in 1h 15m 46s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/354eb1dc591d4934a4cbbaaffe1435b8 : SUCCESS in 1h 15m 34s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fc419cebf11c461fb16a214abb807719 : SUCCESS in 1h 14m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fba073d871d48dc999824d8977208ff : SUCCESS in 4m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/887596cbe6f6429492f556042b160b84 : SUCCESS in 2m 50s","accounts_in_message":[],"_revision_number":2},{"id":"1732b0f74747c22730c88a129073d5eda441a9cb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2023-04-08 12:53:10.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"d1aba8c6813374170e00bf3d78fa1a6593fcfd4d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-08 15:24:19.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/69f57a0478a74ab190f2de5a9d9cad0d\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3c8b195e416f4503ad0f75d0689781c4 : SUCCESS in 11m 05s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/6f167d905c7e43f796cd60f569a81841 : SUCCESS in 1h 22m 59s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/3e885b9f04ca4158beb2834dc292676c : SUCCESS in 1h 04m 55s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0b0678168f8f452b91ff5fcac5c771b5 : SUCCESS in 1h 35m 04s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e1fa3c37b4054d69b3e1f9ab40a85a5f : SUCCESS in 1h 29m 53s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0d7608f704384e6a96322df2666658a6 : SUCCESS in 1h 27m 42s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/904dae5a3c9247a2a08b76752c2fd66c : SUCCESS in 2h 24m 00s\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b6792feba778485b92894e246ef71e9f : SUCCESS in 2h 04m 36s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/bfdee9fb68ff4c36b2147f8af81adf7d : SUCCESS in 1h 04m 22s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7236040f280548b28ca1c1cec6567dd2 : SUCCESS in 1h 45m 03s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/93bccd311e9a4a2d9e8028e30b5f441f : FAILURE in 18m 38s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/039a4c1479da4fc0815379da4b51ddf3 : SUCCESS in 1h 28m 23s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/dfa1f295a4374e31a9d3ebe71e9c2185 : SUCCESS in 1h 17m 55s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/77002bde8f7344e1b2d3e297917ac866 : SUCCESS in 1h 09m 45s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6664e8ae7fbf4006a7a5317f5900821f : SUCCESS in 1h 21m 08s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f439e2f306494a2cbd5ae3506be9856b : FAILURE in 33m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/447412df2b6549a59c2e375de43da459 : SUCCESS in 4m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3ec1ef25676046ce821e42a5008b0b22 : SUCCESS in 3m 07s","accounts_in_message":[],"_revision_number":3},{"id":"3520c09509b7dc8232b69b34aeaef927de2bed52","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2023-04-08 16:24:16.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6c8133068d0cfb67d3dabd72387f773385c14d97","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-08 18:51:18.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f65953446e684f799d5a0f1c42aa3597\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d7e4695f485140539fadc3f5c44e1b8e : SUCCESS in 9m 13s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/24b5d9a60cd44db88ed05fad7fc90a84 : SUCCESS in 1h 19m 51s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/49fbd655c5424b2292e9d8317608be7e : SUCCESS in 1h 11m 12s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d25b7871462c4ef3ab45832ba93cfb2c : SUCCESS in 1h 46m 54s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ca27f3f4c18d4edd844eea3cd4acc07e : SUCCESS in 1h 19m 27s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e06949b90ed04dfdb96257379dc3df71 : SUCCESS in 1h 25m 27s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b3599860ef5b4c7cbca9ec6764823d2a : SUCCESS in 2h 10m 32s\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7fb229d32ca4461283b27721328529a6 : SUCCESS in 1h 35m 54s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f44640699a2d4133b08d3643a9b30487 : SUCCESS in 1h 03m 19s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e2374de8e62b47b8a2438d4f5ebf3a3a : SUCCESS in 51m 47s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/a9095f55c9c04ba8acea724bc462b919 : FAILURE in 18m 59s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e692471b9ea349f4b37f55ac4d0b68ca : SUCCESS in 1h 26m 12s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/7c1628ee04c04895a02aa30e37c5168a : SUCCESS in 57m 43s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a7159fc3e221436abe1fae5407804fda : SUCCESS in 1h 11m 10s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/40f16f9792f44a12ae94f3809796244b : SUCCESS in 1h 17m 22s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1c1e177164a84310b212370dabd1c2cb : SUCCESS in 1h 09m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d13a774866fe431bbe740c0b9d372ba3 : SUCCESS in 5m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2f0448290ff7449da340b5b312fa4723 : SUCCESS in 2m 46s","accounts_in_message":[],"_revision_number":3},{"id":"6f761a2cb5325bcc6e9b9461d4c5735940747392","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-04-14 13:24:02.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"1a216b194f2cd68eef34d444c0b13433ab46d5d1","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-04-18 08:25:17.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"2e525799d1ee7c5e546a8dad129f0c156bbba48d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-18 08:27:21.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"87cd7dc6e1a34f5c1e240999b563dd368831cfd4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-18 10:04:54.000000000","message":"Patch Set 3: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/879914 is needed.","accounts_in_message":[],"_revision_number":3},{"id":"7bef945d25f316b353f59628843f39e71815d4cc","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-04-19 08:46:03.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"322e3479346a143e15dcd4ab92fc17ea0a4a0983","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-19 11:21:59.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/67b68fb8a1c94b629f9bc8328ed29c52\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4843c26a7ff44557bd4294713ab3f1cb : SUCCESS in 11m 16s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/3d42410c6f49406fb77173231a9f0de8 : SUCCESS in 1h 23m 05s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/2fdea16d5b1c433f9deb055b2641b112 : SUCCESS in 1h 20m 09s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7119339d4b0c4d309433ef539bf8a1ef : SUCCESS in 1h 33m 41s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3d920db769a9449ea1c1533cd6b1f626 : SUCCESS in 1h 33m 05s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/699a05fffa854bd78f8aa012c3026190 : SUCCESS in 1h 27m 58s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ce4c0c29023b4ad1a33f2e499c874198 : SUCCESS in 2h 16m 52s\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/32efdf08f0c8483a9ae51a9e76ca4f41 : SUCCESS in 2h 18m 26s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3feda4bf8c1849b8aec1327b078771df : FAILURE in 27m 15s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/acc448b0774f4550990ba89c65d38bfd : FAILURE in 28m 43s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/aac3bf42c73a4d14963f8280a59f295e : FAILURE in 40m 17s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/86880d6a5250432180aa6383121ddcae : SUCCESS in 1h 24m 47s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/d70cf95ce5624279a4c9633af0a45095 : SUCCESS in 1h 17m 42s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/975172dcb2a44cdf89e59ca52880341d : SUCCESS in 1h 12m 11s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f57cbac6770a41acad49704d72239304 : SUCCESS in 1h 18m 20s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3213ffbe8428411194a1e3e01519257a : SUCCESS in 1h 11m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a352bda9d474bbe906da46c0550430c : SUCCESS in 6m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fa04cb2b59a5476888aba3592416f562 : SUCCESS in 2m 49s","accounts_in_message":[],"_revision_number":3},{"id":"439d0d1decf1a70832161def0e2bde8b1b8121f2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-19 11:24:04.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"382157eed7102bf7b49e488f87c0dfc25e7aaaeb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-19 12:58:06.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4e9dbe98dbc84b599364097e8b3dd134\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fbbc75d6f02b43b1b6a8a5b8aa10964f : SUCCESS in 16m 34s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/7539fc1c6ace4ce08d87022311e0462d : SUCCESS in 1h 26m 42s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7a482867f89c4515a2458dc79a5d4de1 : SUCCESS in 1h 32m 11s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8b05945cb6834d6e9a64909226ad95db : SUCCESS in 1h 32m 43s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9fa8ab6c27e64f8e92d21b2da56a612c : SUCCESS in 1h 24m 36s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/333a09d768534086a3e6c78efda7faa2 : SUCCESS in 1h 16m 54s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/610c071469e34006befb39162fca114b : FAILURE in 1h 13m 35s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/50d3d3a3b2bd4d42bd8b5804154e7358 : SUCCESS in 1h 21m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2bcce7683d944813ade525d75e938b9b : SUCCESS in 4m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3b12346d52784c65a6d1b2ca1a843e66 : SUCCESS in 3m 39s","accounts_in_message":[],"_revision_number":3},{"id":"117ca44741008ec642f0f4c50ac5c857aa34f243","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-04-20 09:31:39.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"739a01606b8108614f31f5b1677c0d7e4591adf7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 11:39:14.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f7bad37cabc245acb0fdb4e6d9272735\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bca4aa900e3e49fc859bce398c216ad2 : SUCCESS in 14m 08s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/9082a349ba4546d49b8ff3235291b8aa : SUCCESS in 1h 46m 51s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/56599f0a813a497eb76437eba4020e53 : SUCCESS in 1h 35m 27s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2e2e7d94bd3a45768c8312bb0b32f813 : SUCCESS in 1h 39m 29s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e9b387195cb04f038ab5345e3432e65a : SUCCESS in 1h 41m 14s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2711ccfc6c8b40e38180f637022ae405 : SUCCESS in 1h 35m 21s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/fa03613e93814cefacfdeeb539963e0c : SUCCESS in 1h 58m 54s\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/089756842348456b914fcaf9fca20a78 : FAILURE in 2h 01m 14s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/746ad41bf0ba4d708ddfce84d84763a3 : FAILURE in 34m 17s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fecc99eeb3624fa49f66c0cf5bdb277e : FAILURE in 28m 42s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/065502582426452dbd7a7eaaecd4df82 : FAILURE in 41m 00s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/cdce4a7b7577414bb64c0ca08eff7986 : SUCCESS in 1h 20m 54s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/d2114f0a55ee40c88cf0a5845f6eba6a : SUCCESS in 1h 11m 51s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c980df26552647c69e188d166d71159b : SUCCESS in 1h 11m 15s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7f27da5ff2ea48939932c5c338ddd09e : SUCCESS in 1h 12m 42s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b01ffe9685a2420bb6b3dc6648b3fc7a : FAILURE in 1h 15m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c23a95aad8b4b849273250a14c341a6 : SUCCESS in 3m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/278118990bc34fd3a248485e13368a17 : SUCCESS in 2m 47s","accounts_in_message":[],"_revision_number":3},{"id":"be0f6ddd204b31d77fd935efad8f58d802e4e1ad","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-04-20 14:17:46.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"dec4914ca393889f6a4fe7253bfccd18004a3ae2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 16:57:54.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc1f8cef7d7b4138a216606f123dbd28\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0436e2ef9fbe417697b0ba05c32a998e : SUCCESS in 13m 57s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/99683246765e4b20b22774f7506dd7e4 : SUCCESS in 1h 22m 06s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/80d026f4675d40eaa811f3f75123f249 : SUCCESS in 1h 43m 06s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/38c10d5b765b461290e98210624b88d2 : SUCCESS in 1h 44m 03s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bafad2a9319f4449814583f5ae70f2f1 : SUCCESS in 1h 32m 18s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8a44d41b2f60439996ed60cb8938f699 : SUCCESS in 1h 38m 49s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c73a9f8c100c41e884deba70f1e38a01 : SUCCESS in 2h 22m 56s\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/602ef738e3c148699fbee9695b399077 : SUCCESS in 1h 54m 08s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f62775ab02be4a5c9b4bf111bc8c5917 : FAILURE in 27m 43s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b4c2af842f534d7db7ef4d6bf1028fe6 : FAILURE in 28m 52s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/65e031bf3e52458c89acd4d575741e2c : FAILURE in 52m 12s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9109990f32da4b4ba6944ba5568706f0 : SUCCESS in 1h 41m 50s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/a1662fc526a442ceab8c655936d33b2b : SUCCESS in 1h 06m 32s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cf1bfe0bf13949f88616d2f8fe991135 : SUCCESS in 1h 30m 58s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a2392b0621f14c55b9543fac1194e978 : SUCCESS in 1h 21m 21s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8389e5e787c8487aa6bbc1dfe44fd6b9 : SUCCESS in 1h 25m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bcd5771c024649929a4f7fd9a7f5a5c1 : SUCCESS in 4m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/71bb2f4526b84cd1b67fc5c35022ab67 : SUCCESS in 3m 02s","accounts_in_message":[],"_revision_number":3},{"id":"33bd84cf41ea466cf27212c2ab6d1246324aebf8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 17:00:04.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"75d27b19cf21ffd797e79d1166032b058996d0d7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 18:46:50.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/76d92bf0e1bb49c68505f6251d50ad03\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a08b6e941c5745009f38538aae4b3e21 : SUCCESS in 11m 20s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/ff771cdd49734bc6b19d7e8678277af5 : SUCCESS in 1h 13m 08s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1d419a722a5b417bbdb2cad0845beb49 : SUCCESS in 1h 41m 32s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d0433cb8c47148dbb3d8219347e1ec78 : SUCCESS in 1h 37m 48s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f18ea0591d0d4656b4ecd4bdc9337eff : SUCCESS in 1h 39m 37s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/b4a6b99b3c3f42b39e93934f80bcac2b : SUCCESS in 1h 16m 42s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a319739448154e31a6477e55c1aad9a4 : SUCCESS in 1h 10m 48s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/57a3d58a6b4f4be395313f82ad2fad0a : SUCCESS in 1h 20m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5e3504e473648129df56370e866983e : SUCCESS in 4m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/53a48bf2c37e49f3b49f5efb7ad4b502 : SUCCESS in 3m 26s","accounts_in_message":[],"_revision_number":3},{"id":"7d5a927d46bd06aedfcc3aebc4f8c05e78f8b612","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 18:46:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"3a88cdd5006e4502f3e7557af4f2fdbf5d3ac638","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 18:47:49.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a2f9a60cc5a44e4b899f9c3de2899bb\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e8fa6b0987a4237a89275d56b2ace61 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"6661a9dab73b35c4b72bc4c0b4c6c3097ae0dc68","revisions":{"937b0578c37c6e922aed8073d3252b57c6f140f0":{"kind":"REWORK","_number":1,"created":"2023-04-03 20:35:23.000000000","uploader":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"ref":"refs/changes/79/879379/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/79/879379/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/79/879379/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/79/879379/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/79/879379/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/79/879379/1"}}},"commit":{"parents":[{"commit":"fad35159429d2fd45ebbb6c6bc7ef0b6e74502be","subject":"Define CA cert when needed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/fad35159429d2fd45ebbb6c6bc7ef0b6e74502be"}]}],"author":{"name":"Damian Dabrowski","email":"damian.dabrowski@cleura.com","date":"2023-03-30 22:05:40.000000000","tz":120},"committer":{"name":"Damian Dabrowski","email":"damian.dabrowski@cleura.com","date":"2023-03-30 22:05:40.000000000","tz":120},"subject":"Rename keystone_ssl to keystone_backend_https","message":"Rename keystone_ssl to keystone_backend_https\n\nTo standarize variable name across roles, this change renames\n`keystone_ssl` to `keystone_backend_https`.\nAll other roles use `\u003cservice\u003e_backend_https` format.\nIt also better describes what it does. With `keystone_ssl` it\u0027s unclear\nwhether it is about frontend or backend.\n\nBackward compatibility will not be implemented because securing haproxy\ntraffic to its backends with TLS is currently not supported by OSA so\nit is hard to leverage `keystone_ssl` variable anyway.\n\nChange-Id: Ibf8607a4cf62ab518a09d64b1054ff7fbc580000\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/937b0578c37c6e922aed8073d3252b57c6f140f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/937b0578c37c6e922aed8073d3252b57c6f140f0"}]},"branch":"refs/heads/master"},"1510f1187e5a73523d1400ad1e80b8b34c5fbf54":{"kind":"REWORK","_number":2,"created":"2023-04-07 23:38:26.000000000","uploader":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"ref":"refs/changes/79/879379/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/79/879379/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/79/879379/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/79/879379/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/79/879379/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/79/879379/2"}}},"commit":{"parents":[{"commit":"339dfb2e0e12c9d9f1a034539852ef6a9d144b4a","subject":"Use chain cert file for apache","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/339dfb2e0e12c9d9f1a034539852ef6a9d144b4a"}]}],"author":{"name":"Damian Dabrowski","email":"damian.dabrowski@cleura.com","date":"2023-03-30 22:05:40.000000000","tz":120},"committer":{"name":"Damian Dabrowski","email":"damian.dabrowski@cleura.com","date":"2023-04-07 23:38:15.000000000","tz":120},"subject":"Rename keystone_ssl to keystone_backend_ssl","message":"Rename keystone_ssl to keystone_backend_ssl\n\nTo standarize variable name across roles, this change renames\n`keystone_ssl` to `keystone_backend_ssl`.\nAll other roles use `\u003cservice\u003e_backend_ssl` format.\nIt also better describes what it does. With `keystone_ssl` it\u0027s unclear\nwhether it is about frontend or backend.\n\nBackward compatibility will not be implemented because securing haproxy\ntraffic to its backends with TLS is currently not supported by OSA so\nit is hard to leverage `keystone_ssl` variable anyway.\n\nChange-Id: Ibf8607a4cf62ab518a09d64b1054ff7fbc580000\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/1510f1187e5a73523d1400ad1e80b8b34c5fbf54"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/1510f1187e5a73523d1400ad1e80b8b34c5fbf54"}]},"branch":"refs/heads/master"},"6661a9dab73b35c4b72bc4c0b4c6c3097ae0dc68":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-04-08 12:53:10.000000000","uploader":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"ref":"refs/changes/79/879379/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/79/879379/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/79/879379/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/79/879379/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/79/879379/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/79/879379/3"}}},"commit":{"parents":[{"commit":"931695475cbb261678e4571c2523bf6bd4c2bef5","subject":"Use chain cert file for apache","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/931695475cbb261678e4571c2523bf6bd4c2bef5"}]}],"author":{"name":"Damian Dabrowski","email":"damian.dabrowski@cleura.com","date":"2023-03-30 22:05:40.000000000","tz":120},"committer":{"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","date":"2023-04-08 12:53:10.000000000","tz":0},"subject":"Rename keystone_ssl to keystone_backend_ssl","message":"Rename keystone_ssl to keystone_backend_ssl\n\nTo standarize variable name across roles, this change renames\n`keystone_ssl` to `keystone_backend_ssl`.\nAll other roles use `\u003cservice\u003e_backend_ssl` format.\nIt also better describes what it does. With `keystone_ssl` it\u0027s unclear\nwhether it is about frontend or backend.\n\nBackward compatibility will not be implemented because securing haproxy\ntraffic to its backends with TLS is currently not supported by OSA so\nit is hard to leverage `keystone_ssl` variable anyway.\n\nChange-Id: Ibf8607a4cf62ab518a09d64b1054ff7fbc580000\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/6661a9dab73b35c4b72bc4c0b4c6c3097ae0dc68"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/6661a9dab73b35c4b72bc4c0b4c6c3097ae0dc68"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
