)]}'
{"id":"openstack%2Fopenstack-ansible-os_keystone~888585","triplet_id":"openstack%2Fopenstack-ansible-os_keystone~master~Id92330b3c709201a74612c8353cefa75778eac0c","project":"openstack/openstack-ansible-os_keystone","branch":"master","topic":"osa/core-2.15","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2023-08-14 10:45:47.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id92330b3c709201a74612c8353cefa75778eac0c","subject":"Fix linters and metadata","status":"MERGED","created":"2023-07-14 15:19:36.000000000","updated":"2023-08-14 10:46:45.000000000","submitted":"2023-08-14 10:45:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"888585-osa/core-2.15","meta_rev_id":"e6c5b805306467683a99eb5419052d46eb15d123","_number":888585,"virtual_id_number":888585,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-14 10:45:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"all":[{"value":2,"date":"2023-08-08 20:51:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-08-14 08:44:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-08-14 08:44:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-14 17:28:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-08 20:51:04.000000000","updated_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"reviewer":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"state":"REVIEWER"},{"updated":"2023-08-14 08:44:04.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"e3944f373ab622c45b8be80735115d08eafc27d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-07-14 15:19:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"17776bbc5f87d3f28f3d60950a056b10b6775051","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-14 17:28:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/884363ef465c4ac69c9add5bb814ef8f\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1ef0a6ab165b417abe4452c1fe9bc027 : SUCCESS in 15m 05s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/cbbe836dc66248ceb4038e7e2795a4b8 : FAILURE in 44m 14s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/af7760ac002f4b41bc0c9e9533827abe : FAILURE in 44m 05s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/94e71e2e5b184c1b912574ffadd72d0c : FAILURE in 48m 47s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/55f4fd29b76a422aa1198d1b08caaf65 : FAILURE in 39m 46s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1211f2524bac45728d1d923e934e8bfa : FAILURE in 45m 22s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/851e3d7d1ce84738ae77d6e625aaa819 : FAILURE in 48m 45s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3799943738794beeb173c29cc3ae00ad : FAILURE in 45m 32s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9e806f684dc14e44a7700a0f2a7379a9 : FAILURE in 1h 08m 13s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a938d79d6a3e44feb5f8b09e1280f0a6 : SUCCESS in 2h 05m 46s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c28859a514d24785a015e4bb9cafa3b7 : FAILURE in 36m 32s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/abad61ef8d654f90b50522cf8ba578fa : FAILURE in 28m 47s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/0e87034882b64dd28e1fed0103670207 : FAILURE in 38m 03s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e59fb1635f2e4fe483a73abffe9f0c79 : FAILURE in 1h 23m 29s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d022fb7b93374b63a2de31bf3fa39831 : FAILURE in 1h 30m 12s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/6b985a10d8284357ae634f59f4fea913 : FAILURE in 1h 02m 06s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/307144b54928463dad821d5d472de80c : FAILURE in 1h 03m 19s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/100e5613da4249339d37649aa7e28a32 : FAILURE in 1h 06m 42s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8a8efcaface4482d8ec771f5ec59be4c : FAILURE in 1h 04m 57s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1af3770806d141ad8b2c0b22732a3395 : FAILURE in 1h 10m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/150e53227ea845d19daf8fdf855eca19 : SUCCESS in 4m 45s","accounts_in_message":[],"_revision_number":1},{"id":"a6307945098f3562dd582f241c093813693839d0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-07-14 18:45:01.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"d8a4625074239ec8cbe0c6881417fbcdd1da6f59","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-14 21:58:32.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8dd2e1828a624cf48f6d82d73fc76357\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4b2a14170a2a4819af502031c4391fba : SUCCESS in 11m 53s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/3222e31156b34528a1590b9e33e3565f : SUCCESS in 1h 28m 36s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/398bb4205bba4f93b7d9609068498784 : SUCCESS in 1h 28m 34s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/65053807e9ee433ab1694feafedf6a69 : FAILURE in 1h 07m 53s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f8d6cba8bf254748b2bc6e36f9f802bd : SUCCESS in 1h 35m 35s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/238d1eeb9c8b430db6f2adf1902bba98 : SUCCESS in 1h 23m 41s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/46f1b36748124074839acb6674b62f90 : FAILURE in 1h 17m 33s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/78676ed5cf6541bdb26e08e6002888c0 : SUCCESS in 1h 32m 44s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4017b2305f574229bdeb07805b484cd6 : FAILURE in 57m 29s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/96e5a2aec46d4ed6818909dbf9539435 : SUCCESS in 2h 19m 57s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f10edc2c771949b19733c54f807fd26b : FAILURE in 47m 42s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/10259685abec4642833b311ca3983869 : SUCCESS in 57m 25s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/0cdd3d6311b74613b5fbb9ffbbcc71c8 : FAILURE in 41m 03s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/817881843bc547d38b0224fb7973cc1d : SUCCESS in 1h 28m 38s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6f1688d29d104224bffe8c1ada20ba3a : SUCCESS in 1h 25m 48s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/4ce186964df34511bb86c5427b80f610 : SUCCESS in 59m 55s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2b11928fa34c44dabfe9a843a50d1102 : SUCCESS in 1h 10m 31s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8a6ba24a7aa8442cb9af8ef270048c8e : SUCCESS in 1h 08m 07s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d75a9fc8b9a64f79af4406d3093cc3c1 : SUCCESS in 1h 15m 24s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3edf315671e945c6bbb8957ca5b08e01 : TIMED_OUT in 3h 08m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/400f388d100647b484a6059dcc46901f : SUCCESS in 4m 41s","accounts_in_message":[],"_revision_number":2},{"id":"8de784314f2e28d6ef7f87d5dff22d823248ad95","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-07-17 07:37:26.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b98d163e8075c7011be2d84f72a12100a476eba7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-17 09:44:38.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a243d1f1ce0d4f5a9a13e6f7cbe21840\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/16d8941437514ef0b35fde7cb4f65f98 : SUCCESS in 11m 34s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/9d2ceecbd4a7443cbfd88d35d8b3a63e : SUCCESS in 1h 10m 55s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/26490f5f9b8145e58d8ea9a97e3ac8d1 : SUCCESS in 1h 37m 39s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1519a02c264044f790cc0b3ef1f48127 : FAILURE in 1h 11m 19s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ab3173d8c434408ca5463d84ff9a68d8 : SUCCESS in 1h 44m 24s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/40d9774bfa124495a2773b0cd09bc3ab : SUCCESS in 1h 36m 28s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/de08d0cf83d54d1aa1f23dcab2559536 : FAILURE in 1h 17m 56s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/73d89c86f4f844b9a31c70b2a55615dd : SUCCESS in 1h 39m 22s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/419cce60d5224047a5a3e1683b2e96fd : FAILURE in 1h 09m 29s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2c11fc6bdefd4f7e8c2cb2674e852525 : SUCCESS in 2h 04m 30s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/37f09fab56614279a11c2c0cb8fb3640 : FAILURE in 50m 16s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4b672ce2fd6e418b8100ffaeed7d47b0 : SUCCESS in 59m 34s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/17204164c1e14658935b9972dd323e59 : FAILURE in 42m 58s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/35c13daa11b048ff9cf4abaca59c0ada : SUCCESS in 1h 25m 37s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f8bf96fc59b541039c376f12462f3b4e : SUCCESS in 1h 44m 35s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/0a9269ed31b846c08d8bf361c1caee72 : SUCCESS in 1h 03m 47s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f7cf3e4ad9c544ebb54dfd5c6f1e9081 : SUCCESS in 1h 22m 45s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/26fd7b72710343679937f49f44c48a7d : SUCCESS in 1h 19m 27s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8d3b28f4aad247819dc2a4eadad73500 : SUCCESS in 1h 18m 50s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a63a1ab95eaf437b98c1610e01ef85c9 : SUCCESS in 1h 01m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c94baeb10ab849a48cd445ad04a40c5d : SUCCESS in 4m 46s","accounts_in_message":[],"_revision_number":2},{"id":"efde3762b689fea513f7a07b2bd6ceb8f9b7c223","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-07-19 07:18:31.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1a51d4c486675b0071514a60bd0b2b0df56ff77c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-19 10:31:36.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2db4a793e7124aef81586287a10a2af7\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/76fb06dffe91440da223ce6f2187deba : SUCCESS in 10m 36s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/050aca1325db4bc0aa8efbe4e68b52e5 : SUCCESS in 1h 29m 34s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/f2595968025148b3888f8fc4d7134c39 : SUCCESS in 1h 50m 21s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b8390b87b27a43c29369e2e2a600f390 : SUCCESS in 1h 40m 51s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7f3d3fa400c54c9d9522c772e73c3267 : SUCCESS in 1h 44m 57s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/34a4a4af6f5b4f7a86fcaa2f62206765 : SUCCESS in 1h 26m 39s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3d7376120e924742b77e0defef881bc9 : SUCCESS in 1h 49m 14s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/60098c9b39954102b37e947d5821ce2b : SUCCESS in 1h 54m 12s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a6c8a7d99e9e4f2981d74242595d04a7 : SUCCESS in 2h 31m 42s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/42ccf3833a5e4bba99b8b39e4bcf5809 : SUCCESS in 1h 55m 04s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9438f6a76a4f464fae04f68e0af29b58 : SUCCESS in 47m 21s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/be09cb8fde93428f9242b9f8ade5d767 : SUCCESS in 1h 12m 33s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/f9f96eb927824b0da1e2a14963fce12f : FAILURE in 44m 05s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/496b6cad14a343338dd32274d14acaa6 : SUCCESS in 1h 38m 30s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a4161d3eeafd494fb570498563bb2c43 : SUCCESS in 1h 40m 26s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/5de9e5857c1043a8ab88fbcf50eb1da2 : SUCCESS in 1h 19m 20s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/41b2c00a845a4c808505dd2a691c322f : SUCCESS in 1h 18m 21s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/21189dc041e142c5b4fa0e5a6e2762de : SUCCESS in 1h 21m 43s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3b7d68c399c34c719f4e2d6fbf31a46a : FAILURE in 34m 10s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fe7224567a334a6f898e29343eb44039 : SUCCESS in 1h 22m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93de7595b06c49d499342ead7ed5dc88 : SUCCESS in 5m 06s","accounts_in_message":[],"_revision_number":2},{"id":"d5af7f450a4ec85d28c3429606c1f1473e3e4470","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-07-19 11:39:29.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"979d5e8908b6ba7a862d92497aa2f4f8f20dc367","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-19 14:49:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/01026085a3a64b4fa2c4abed0247fafc\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/aec862713c5a4000aea21e561742e8b2 : SUCCESS in 16m 53s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/688a5fb2191b4ce49e5181f2cfa2ef08 : SUCCESS in 1h 20m 38s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/5c29b244250d4acc84078aca6a75539e : SUCCESS in 1h 35m 37s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/cdd0cbcfa9194e2b87d3cff5bf87db25 : SUCCESS in 1h 45m 59s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fe92b3f0b12a461ea4ba5c97664539c7 : SUCCESS in 1h 28m 57s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/05fadb41b51a40dd92e1bae94283bf37 : SUCCESS in 1h 37m 18s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4867c5f51e4b4687a323b632c413c46e : SUCCESS in 1h 30m 11s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1beeb6e04a6640c1b52c7b91bf66a9ce : SUCCESS in 3h 01m 38s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/63363ec4cb914d38b6c58d085102d501 : SUCCESS in 2h 48m 30s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a51bb724a9f64fa0ad06a5e6981ce463 : SUCCESS in 1h 10m 07s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d4416bc161e74ec4953b97091d7a08f0 : SUCCESS in 1h 03m 11s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/8c69d1d151be466e8180957d79a39c25 : FAILURE in 45m 04s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9a0552d5cf2541009142c049c9e0207e : SUCCESS in 1h 30m 05s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/44f62c2938b745f99e08fe5f7b7d353c : SUCCESS in 1h 34m 51s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/57402cf279ca48f1890c6130a66a9c03 : SUCCESS in 1h 09m 00s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/febcdf35ce684c279a6ee5ecc83a4be5 : SUCCESS in 1h 10m 52s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e0ffb1c042654bc096a6ab7883ffec22 : SUCCESS in 1h 23m 20s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/693118e79698472480c147c15e9f30f8 : SUCCESS in 1h 27m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d151ab786744766ab0bbc7b5aefc24e : SUCCESS in 4m 22s","accounts_in_message":[],"_revision_number":2},{"id":"5748a5bcdac042dde432ce029fe5d243bc4538a2","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2023-08-08 20:51:04.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"557b663ae57a3f6eab68b64b389e220d3a97836a","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-08-14 08:44:04.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"3493c9587400ccb5b4f8ff4fd18aaf695a436adf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-14 08:46:17.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"121633fd03ec09a44e2a4a460fcb95725a87dcbb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-14 10:45:47.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/23a1f8d410c94679b87f10eea2936a6f\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cbc7942c4868486ba65eb12890cfc8cc : SUCCESS in 13m 24s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/b781254a32ec46a6a67f852d4ff61613 : SUCCESS in 1h 30m 16s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0448caae942c4398a4897fc2233cde3c : SUCCESS in 1h 26m 19s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fe01e378df264625b36c3c27782a3218 : SUCCESS in 1h 23m 43s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/45569be75fec40b4b851e9d3ac0825b7 : SUCCESS in 1h 30m 14s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8b03ea82ce7b465ba26bb26500207d71 : SUCCESS in 1h 49m 11s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5123fcd9d9a748ce80a6df927dc63888 : SUCCESS in 1h 20m 34s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/bd477835af9d4e65b337d5d73b42ba97 : SUCCESS in 1h 30m 55s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/6a2c40eefe014569acf132ca2f9b1f5b : SUCCESS in 1h 13m 28s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0e660c2f90c3473f93e08e1059db9017 : SUCCESS in 1h 07m 27s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/18f7bd12b43d4b8daa390d6c12c71646 : SUCCESS in 1h 17m 48s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ecbea6434f2648309023fba48a36a59a : SUCCESS in 1h 07m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bfff05056722475fa1f21142dfaef04d : SUCCESS in 5m 24s","accounts_in_message":[],"_revision_number":2},{"id":"c5dd0bf1206aec420fd7e886d8a50b1add0240de","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-14 10:45:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"e6c5b805306467683a99eb5419052d46eb15d123","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-14 10:46:45.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/65797234ac6f4b688a6f24742f4f0f4e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9865c6f538024ea5b77bcb9d0d953288 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"eea1a4853f542de0745c3eff91462cd9b8b82872","revisions":{"857ae2734d44b1a3d85cf3959fbf143250ce92e4":{"kind":"REWORK","_number":1,"created":"2023-07-14 15:19:36.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/85/888585/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/85/888585/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/85/888585/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/85/888585/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/85/888585/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/85/888585/1"}}},"commit":{"parents":[{"commit":"b73bcd9981255dfa6eb39fb8ef25a5f852b00b17","subject":"Fix SSL logic in keystone-httpd.conf.j2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/b73bcd9981255dfa6eb39fb8ef25a5f852b00b17"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-07-14 15:19:27.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-07-14 15:19:27.000000000","tz":120},"subject":"Fix linters and metadata","message":"Fix linters and metadata\n\nWith update of ansible-lint to version \u003e\u003d6.0.0 a lot of new\nlinters were added, that enabled by default. In order to comply\nwith linter rules we\u0027re applying changes to the role.\n\nWith that we also update metdata to reflect current state.\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-role-systemd_service/+/888223\nChange-Id: Id92330b3c709201a74612c8353cefa75778eac0c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/857ae2734d44b1a3d85cf3959fbf143250ce92e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/857ae2734d44b1a3d85cf3959fbf143250ce92e4"}]},"branch":"refs/heads/master"},"eea1a4853f542de0745c3eff91462cd9b8b82872":{"kind":"REWORK","_number":2,"created":"2023-07-14 18:45:01.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/85/888585/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/85/888585/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/85/888585/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/85/888585/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/85/888585/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/85/888585/2"}}},"commit":{"parents":[{"commit":"b73bcd9981255dfa6eb39fb8ef25a5f852b00b17","subject":"Fix SSL logic in keystone-httpd.conf.j2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/b73bcd9981255dfa6eb39fb8ef25a5f852b00b17"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-07-14 15:19:27.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-07-14 18:44:53.000000000","tz":120},"subject":"Fix linters and metadata","message":"Fix linters and metadata\n\nWith update of ansible-lint to version \u003e\u003d6.0.0 a lot of new\nlinters were added, that enabled by default. In order to comply\nwith linter rules we\u0027re applying changes to the role.\n\nWith that we also update metdata to reflect current state.\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-role-systemd_service/+/888223\nChange-Id: Id92330b3c709201a74612c8353cefa75778eac0c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/eea1a4853f542de0745c3eff91462cd9b8b82872"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/eea1a4853f542de0745c3eff91462cd9b8b82872"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
