)]}'
{"id":"openstack%2Fopenstack-ansible-os_keystone~897806","triplet_id":"openstack%2Fopenstack-ansible-os_keystone~stable%2F2023.1~Ic9d37a8463d137508d20de20b10af806a223f852","project":"openstack/openstack-ansible-os_keystone","branch":"stable/2023.1","attention_set":{},"removed_from_attention_set":{"31542":{"account":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"last_update":"2023-10-11 14:40:25.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic9d37a8463d137508d20de20b10af806a223f852","subject":"oidc: fix recognition of x forwarded headers from v2.4.11","status":"MERGED","created":"2023-10-10 13:01:58.000000000","updated":"2023-10-11 14:41:27.000000000","submitted":"2023-10-11 14:40:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"897806","cherry_pick_of_change":894616,"cherry_pick_of_patch_set":3,"meta_rev_id":"c68666af3ddf705594d25e0b4728607db388c301","_number":897806,"virtual_id_number":897806,"owner":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-11 14:40:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"recommended":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"all":[{"value":1,"date":"2023-10-11 08:26:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":2,"date":"2023-10-10 13:06:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-10-11 07:34:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-10-11 07:34:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-10 13:06:14.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2023-10-10 15:44:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-11 07:34:22.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2023-10-11 08:26:54.000000000","updated_by":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"reviewer":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"state":"REVIEWER"}],"messages":[{"id":"156b476b34335c89798455a6a086b257a0c9adc8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2023-10-10 13:01:58.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"25ccd552b3f8e25b4934e90350085b3e3fac9214","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-10-10 13:06:14.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5197b928e704ebca31c4bb6050bd99c9c8fe2569","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-10 15:44:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3829a30419234760a875935e6e04b292\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9013fd173fe94f96851d6d12fa7d553b : SUCCESS in 16m 25s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/aef860bed3e8453c9f51b31928a49a11 : SUCCESS in 1h 18m 05s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/72ecf978fdd44bbc9b1f5c84ec89a648 : SUCCESS in 1h 43m 39s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6cc9e007b45341afafaa27f4c8140a03 : SUCCESS in 1h 31m 19s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3ac24101a1a843c39052f1312b5efbda : SUCCESS in 1h 43m 28s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b2f439149ac443f886a2c11bce87dec4 : SUCCESS in 1h 40m 45s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0034a68433444f70a9b499ccc8e284f1 : SUCCESS in 1h 47m 55s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bc3cd023ff9c4cb7aecf4f1d0d7a6cb6 : SUCCESS in 1h 38m 07s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ade29a264caf402b8f8020ea3d3f036c : SUCCESS in 2h 33m 29s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5f547d9495aa48888456b1f2cec92081 : SUCCESS in 2h 09m 31s\n- openstack-ansible-upgrade_yoga-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5c4a7d709949435fb46c52b93fcbd0ac : SUCCESS in 1h 54m 08s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/49827702a22d4d4aa1a7f2ee57f4ba3a : SUCCESS in 1h 10m 12s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/78f0e275b5ef4fe8b96d4483803bfcc2 : SUCCESS in 1h 04m 10s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/9549a18dce4447d2b60b262ceeddc446 : SUCCESS in 1h 41m 55s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/db6d573938bb46d9b9bf5cfd3a59dc9a : SUCCESS in 1h 42m 45s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/dfed045408914f29b5504a2913fe915b : SUCCESS in 1h 41m 12s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/a73a9aaa92f344b3bbe4950351137cc7 : SUCCESS in 1h 07m 25s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f2f29b4418614506bd7b0f6609bbdb59 : SUCCESS in 1h 19m 35s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7d9791b4f3124ab9844b1f7755a34b11 : SUCCESS in 1h 15m 49s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/11da4f88f7074c44bbe1fde84fa5b023 : SUCCESS in 1h 16m 16s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/49de9797642f433d80d042f7059b17b7 : SUCCESS in 1h 19m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66a1bcb7ca444d8f9164f21c3a12178b : SUCCESS in 4m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/293a3b54f2214dc5ad37c5e607eb7556 : SUCCESS in 2m 52s","accounts_in_message":[],"_revision_number":1},{"id":"627d6818fc00e7d8f56b62b70ac085ee9a235960","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-10-11 07:34:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"911b6599972241c5eeef3c507a9ddd216f1eedc2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 07:36:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"11848e7fef56cc48c259cc2cabd548860ece137e","author":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"date":"2023-10-11 08:26:54.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"6d2164eae53a33752af8e9a90d94ef227dd0f338","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 09:26:54.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7be6a232e0fa40518087ab7693510a2b\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/51bc31801f3346f39d16fdbc0116e5a9 : SUCCESS in 10m 42s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/de392b963b4b4751a2ace916c22ea5b5 : SUCCESS in 1h 32m 56s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/339a942dc95d4b8a92444b9c0153f96a : SUCCESS in 1h 44m 33s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bbd8b4cff1724c01820f54dbaa50a2d4 : SUCCESS in 1h 32m 24s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/21c09d8df07f4a2385d6c4db80145144 : SUCCESS in 1h 34m 47s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7f12609d8e7b4519a12d8c78dab4b66b : SUCCESS in 1h 23m 19s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2eaef16749864fc4a8f51db8a3762894 : FAILURE in 45m 42s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9c327fab16334dfdaf1f058883469c76 : SUCCESS in 1h 40m 24s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/33a206c47fae467396f1e6d0179f186d : SUCCESS in 1h 10m 29s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d79aa074d81c4a8da4322f2481af09f4 : SUCCESS in 1h 26m 06s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b672931545fc412b9e85cc3eaed726c3 : SUCCESS in 1h 14m 29s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/29630e3e5b57496eb3275101b9fa8027 : SUCCESS in 1h 27m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7cc143794e2a4ac4ae3b295b49ec4b9d : SUCCESS in 4m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5a7359491a184bd5ac2d4f5cd29f47b4 : SUCCESS in 3m 15s","accounts_in_message":[],"_revision_number":1},{"id":"d805b137c671a56baf81af752f14e5ef42c5e2e5","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-10-11 10:10:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3cd2b55b41091df6f52d843a61ad566042a67fc1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 12:43:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9d27544da5a40f2bd22fe2b48611f67\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2b7f31afa62a48d48735db5449819abe : SUCCESS in 12m 06s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/047a80d4a1e84c64b72e71f995a93252 : SUCCESS in 1h 36m 16s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/50ac30f44e1b4a2487904734d8dedf41 : SUCCESS in 1h 43m 48s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c14dddde6aa5483fafe3b14aff36f4e1 : SUCCESS in 1h 36m 59s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/696d3c3d5c884bb2aec3efd44e4da84d : SUCCESS in 1h 32m 14s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/578ae3a3ab3f41f993ef0c3fab48e995 : SUCCESS in 1h 28m 21s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d9de23b4a97145b48358b2f46133ef16 : SUCCESS in 1h 42m 11s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d019c59b88f9495d8c6172c2feaa6c9e : SUCCESS in 1h 31m 07s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/8f041a2259e24507a49fb1f1664c2861 : SUCCESS in 2h 18m 04s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3074202262f34c23be0934f995767faa : SUCCESS in 2h 15m 15s\n- openstack-ansible-upgrade_yoga-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3e0dc102d6444e6298350afaa1873ca0 : SUCCESS in 2h 24m 07s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/bfda5889de3d4fb78f9f44fabb177952 : SUCCESS in 1h 02m 27s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4e214620ceea4be2aeaf92e8b20491a9 : SUCCESS in 1h 06m 28s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/ea957390eab540a9857978689374025e : SUCCESS in 1h 24m 44s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2a8a55da30794d5db194abd349ee3c14 : SUCCESS in 1h 27m 14s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/986d49f9b25e414c89197e3a8c123712 : SUCCESS in 1h 44m 56s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/af530b607ba1458fbe4e1fae1e052af5 : SUCCESS in 1h 07m 45s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/87e3b99df71040489be60fc62878faa2 : SUCCESS in 1h 21m 33s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fa19b684032f42fbbe3bed8ee250a433 : SUCCESS in 1h 17m 59s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b62250b3f4014bf393d0f3c29c7485f4 : SUCCESS in 1h 14m 49s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d675101dc26c4d2e9de9349693355cdb : SUCCESS in 1h 18m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f76f7a614de5463692a218b8ec54593e : SUCCESS in 4m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/723ddec1f1d94c99b3e30fd0ed6a98bf : SUCCESS in 3m 24s","accounts_in_message":[],"_revision_number":1},{"id":"910ad5e0ede0c88f38c2c05f2bf544cc87fd1a51","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 12:45:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"39771625a33f166e046ba41591a95ccee220200b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 14:40:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/71bacd3ab0894b3fa18369ce4588503c\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fd314f8650a34c4d99c5c6e838f1460e : SUCCESS in 11m 33s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/871f477513184fe5b1899a73edfbb37a : SUCCESS in 1h 16m 25s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/83e9b64574784163a765c45ef6877e90 : SUCCESS in 1h 32m 57s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/419b86e9ae8e455e97b6e8280bbb95d8 : SUCCESS in 1h 45m 46s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/72b1000f0f4d412ca6066364e75117cd : SUCCESS in 1h 54m 56s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/563ebfb6a380427fa95cec8bb7f19439 : SUCCESS in 1h 40m 43s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6adb7b4eca9c409cbd7ef862705af45b : SUCCESS in 1h 27m 14s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/bfeda32b49904d5cad6d3ac20c0af870 : SUCCESS in 1h 42m 01s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/a27850a48d3d42a2ad7c4b3e1afed207 : SUCCESS in 1h 14m 34s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bff12496f75e4d0fb2c8ae5ace662fd6 : SUCCESS in 1h 15m 29s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cb78bab0d6614512a2038aa962724620 : SUCCESS in 1h 25m 37s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a8f7782b7a714dd08788272997f583f6 : SUCCESS in 1h 27m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/052d341ec5b3427492bf401337024b20 : SUCCESS in 5m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/17885e25664541efb4012c351c281589 : SUCCESS in 3m 14s","accounts_in_message":[],"_revision_number":1},{"id":"cee1724cd9c7659923a414ef665496a4d7c54eb7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 14:40:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c68666af3ddf705594d25e0b4728607db388c301","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 14:41:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9f33300fb2cf4aab886566e33e769599\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9453e181aa48476dbca8c38db66bc204 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ce6ade57bdf6633fdef8d22827c30be9c029720f","revisions":{"ce6ade57bdf6633fdef8d22827c30be9c029720f":{"kind":"REWORK","_number":1,"created":"2023-10-10 13:01:58.000000000","uploader":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"ref":"refs/changes/06/897806/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/06/897806/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/06/897806/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/06/897806/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/06/897806/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/06/897806/1"}}},"commit":{"parents":[{"commit":"f805a3ae5fd0f25321ca9ad7e8c5c4d13f1edc22","subject":"Install libldap-common for keystone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/f805a3ae5fd0f25321ca9ad7e8c5c4d13f1edc22"}]}],"author":{"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","date":"2023-09-12 07:51:13.000000000","tz":60},"committer":{"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","date":"2023-10-10 13:01:58.000000000","tz":0},"subject":"oidc: fix recognition of x forwarded headers from v2.4.11","message":"oidc: fix recognition of x forwarded headers from v2.4.11\n\nThe Apache mod_auth_openidc requires explicit configuration in\norder to read the X-Forwarded-Proto from the reverse proxy as\nof version v2.4.11 which comes in from Ubuntu Jammy.\n\nEventually this will need to become the default and the\nvariable added in this patch can be removed.\n\nChange-Id: Ic9d37a8463d137508d20de20b10af806a223f852\n(cherry picked from commit b54478e7e1e82bde1eddcab63126e9b2d423f68f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/ce6ade57bdf6633fdef8d22827c30be9c029720f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/ce6ade57bdf6633fdef8d22827c30be9c029720f"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
