)]}'
{"id":"openstack%2Fopenstack-ansible-os_keystone~916878","triplet_id":"openstack%2Fopenstack-ansible-os_keystone~master~I92b95acc5ec35468cafe041b277ef9fb3c21c2e4","project":"openstack/openstack-ansible-os_keystone","branch":"master","topic":"osa/messaging_improvements","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2024-05-09 08:20:54.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I92b95acc5ec35468cafe041b277ef9fb3c21c2e4","subject":"Add variable to globally control notifications enablement and disable RPC","status":"MERGED","created":"2024-04-24 09:47:23.000000000","updated":"2024-05-09 08:21:56.000000000","submitted":"2024-05-09 08:20:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"916878-osa/messaging_improvements","meta_rev_id":"f884faccbacf44a50fd094147414e1fb96bacb6b","_number":916878,"virtual_id_number":916878,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-09 08:20:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2024-05-07 13:04:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-05-09 06:33:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-05-09 06:33:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-24 12:27:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-07 13:04:26.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2024-05-09 06:33:25.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"d070b47e3c689309957b8a1b2f323039d1971160","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-04-24 09:47:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2f256e53129d05d50d2080530f4ba2328e87f8ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-24 12:27:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/14192bdcec66465dbe5e8bd062fac4fb\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a70074708616422eb4ccc230bd5ed60a : SUCCESS in 14m 32s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/7dca041b0a0144df9443f183dced7df8 : SUCCESS in 1h 04m 56s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/1043a8ca93d0439d822f6eb3ceb59599 : SUCCESS in 1h 18m 23s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/0a5f5377ffcd4a5a9a30eca2e18ec23d : SUCCESS in 1h 26m 19s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/98e9147674b640b5beba9d711f83cf9f : SUCCESS in 1h 25m 01s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/20e538eb84bd4765b82458771e055720 : SUCCESS in 1h 12m 23s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/95bd153c05dd4d349bc508f128b34b81 : SUCCESS in 1h 40m 18s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2be5bc63eadd47628a4175cd91b9cccc : SUCCESS in 1h 19m 49s\n- openstack-ansible-deploy-aio_metal_quorum-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c46ba28d346a41bbb73a4b58ba7baae2 : SUCCESS in 1h 12m 14s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/8d02b7b123d14ad89b530ac40e25351e : SUCCESS in 1h 41m 02s\n- openstack-ansible-upgrade_2023.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6ccd1e953e654747b4f2af7a109ffb9c : SUCCESS in 2h 29m 36s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/867a52453dc64d2bbf0baa1ca8726fd2 : SUCCESS in 1h 25m 14s\n- openstack-ansible-upgrade_2023.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/203c03df791d46b4be6bc052eed6fdbe : SUCCESS in 2h 05m 55s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1248e8e3d9844a3d8111139f0f646a5f : SUCCESS in 1h 59m 07s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a54202721c114f2daa18a0acc5a05cf3 : SUCCESS in 1h 52m 48s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ca41b45859624b72bd1043ad77e144aa : SUCCESS in 1h 05m 21s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c92554db4e354099b95a5755684a083e : SUCCESS in 1h 00m 24s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/64b6152926284339b5c98d7f85a23893 : SUCCESS in 1h 31m 08s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a22c041c13ad4838a41540b9c0882d53 : SUCCESS in 1h 27m 12s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b073dc48a64144bd81ffe26561a21c4a : SUCCESS in 1h 28m 07s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/a4e771c1961b4b81b94c8c9dc4589c3c : SUCCESS in 1h 09m 00s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/8c332ac882ad4a538543d028c3f1dc5a : SUCCESS in 1h 10m 57s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ca212a99fbed43d2915a357f5d75278c : SUCCESS in 1h 18m 12s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/798b6903bda646cb9ce70ea0a6fab03c : SUCCESS in 1h 17m 45s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b4dff945b7f741018361a48df08fd704 : SUCCESS in 1h 13m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aca0f771b7124fa68499885e30384555 : SUCCESS in 4m 20s","accounts_in_message":[],"_revision_number":1},{"id":"8a4dee4969c3bd7e930e5a67cbc3d5859da44404","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2024-05-07 13:04:26.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d3ba4820962b557fb43e3318aa27575c19aac626","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2024-05-09 06:33:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c156a8f0b107f090a8321a9ae9004c1b03489c9e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-09 06:35:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c42edf49af9d4697f25ad7178bf3bf49c73acbdc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-09 08:20:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f9be4d1436644f268a7e66dc662fb09e\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0cf46147ee564050bdf2f5b4818df05a : SUCCESS in 15m 31s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/8a84b69dfd8a44bc9197c492bf1784c3 : SUCCESS in 1h 03m 18s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/95f67c1e8150417ba9b47b3f71d0ffdc : SUCCESS in 1h 20m 14s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/27066868dd474f6b882148347b76883c : SUCCESS in 1h 22m 30s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/59eef54f471e4ee5896fbe849b319d07 : SUCCESS in 1h 20m 11s\n- openstack-ansible-deploy-aio_metal_quorum-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7e739eedf24f4c25b9d2b92b877dae47 : SUCCESS in 1h 20m 11s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/05ef087fcac14d6196e614b4e23e5a36 : SUCCESS in 1h 19m 31s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9be6e57243a04cd9b821a38e2d43ea70 : SUCCESS in 1h 19m 31s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/8918d37ee2bf47abb2255be1dd368a5b : SUCCESS in 1h 20m 16s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2a679321f0a04d5ba7a651c89ef2b4b8 : SUCCESS in 1h 34m 58s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/c4ebfa21bdfa4d4594bb6bb34455c4ed : SUCCESS in 1h 24m 02s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/ae8e9a8b04e640c0bb74c3dfb761d627 : SUCCESS in 1h 21m 46s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7ec62963b36d45588902f780142f6746 : SUCCESS in 1h 17m 23s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3304cd1827464fa0a53f2e5a86081f9d : SUCCESS in 1h 27m 16s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9dc74950eff345ac907f2dfbee9661c9 : SUCCESS in 1h 14m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1ec01ca153746bfa3e213b94e4efb54 : SUCCESS in 4m 04s","accounts_in_message":[],"_revision_number":1},{"id":"bf341d59d94fc5c3048de18688894975e249f7a5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-09 08:20:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f884faccbacf44a50fd094147414e1fb96bacb6b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-09 08:21:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9a42c7b633e4c4db4153efd0ed52e49\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e1fe656820242b092dd1a40f3136b04 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1f4d83ef844b741c9e2f565849ba440232e33ea2","revisions":{"1f4d83ef844b741c9e2f565849ba440232e33ea2":{"kind":"REWORK","_number":1,"created":"2024-04-24 09:47:23.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/78/916878/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/78/916878/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/78/916878/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/78/916878/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/78/916878/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/78/916878/1"}}},"commit":{"parents":[{"commit":"5044687dfee21ca631f2a8daad054c3411a28327","subject":"Add service policies defenition","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/5044687dfee21ca631f2a8daad054c3411a28327"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-04-24 09:47:12.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-04-24 09:47:12.000000000","tz":120},"subject":"Add variable to globally control notifications enablement and disable RPC","message":"Add variable to globally control notifications enablement and disable RPC\n\nIn order to be able to globally enable notification reporting for all services,\nwithout an need to have ceilometer deployed or bunch of overrides for each\nservice, we add `oslomsg_notify_enabled` variable that aims to control\nbehaviour of enabled notifications.\n\nPresence of ceilometer is still respected by default and being referenced.\n\nPotential usecase are various billing panels that do rely on notifications\nbut do not require presence of Ceilometer.\n\nThis change also disables RPC communication for Glance since there\u0027s\nno signs of RPC usage in Glance code. RabbitMQ seems to be used solely\nfor notifications\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-plugins/+/914144\nChange-Id: I92b95acc5ec35468cafe041b277ef9fb3c21c2e4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/1f4d83ef844b741c9e2f565849ba440232e33ea2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/1f4d83ef844b741c9e2f565849ba440232e33ea2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
