)]}'
{"id":"openstack%2Fopenstack-ansible-os_keystone~919563","triplet_id":"openstack%2Fopenstack-ansible-os_keystone~master~I5d9505ed4b385673cc719e9821ee2e1c2ba5c754","project":"openstack/openstack-ansible-os_keystone","branch":"master","attention_set":{},"removed_from_attention_set":{"31542":{"account":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"last_update":"2024-09-30 11:44:30.000000000","reason":"Change was submitted"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2024-09-30 11:44:30.000000000","reason":"Change was submitted"},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2024-09-30 09:55:22.000000000","reason":"\u003cGERRIT_ACCOUNT_25023\u003e replied on the change","reason_account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}}},"hashtags":[],"change_id":"I5d9505ed4b385673cc719e9821ee2e1c2ba5c754","subject":"Change example to contain domain name instead of UUID","status":"MERGED","created":"2024-05-14 11:41:39.000000000","updated":"2024-09-30 11:45:27.000000000","submitted":"2024-09-30 11:44:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"919563","meta_rev_id":"53a9760efeec6b65291c314ab7b6a54b4517c6f5","_number":919563,"virtual_id_number":919563,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-30 11:44:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":2,"date":"2024-09-27 10:04:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-09-30 09:55:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-09-30 09:55:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":1,"date":"2024-09-26 06:42:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-14 14:48:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-15 09:21:59.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2024-09-25 07:08:45.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2024-09-27 10:04:39.000000000","updated_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"reviewer":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"state":"REVIEWER"}],"messages":[{"id":"c636c197ac2965a48c48901c4148df1021eb5af0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-05-14 11:41:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"207286bec3385fd80905106aa1a24e54f5b712c5","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-05-14 11:41:54.000000000","message":"Patch Set 1: Backport-Candidate+1","accounts_in_message":[],"_revision_number":1},{"id":"685c5e368523c47d801826adcb21d48403b0b4a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-14 14:48:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/512f23bf6ab24564bf81a3747a1b37b3\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4a07731bec474e4da26e13488f7fc819 : SUCCESS in 13m 36s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/8f39240e92684a07a80cd74bef900015 : SUCCESS in 1h 07m 44s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/2a700b0ff433460db85be5f7092c01a7 : SUCCESS in 1h 18m 22s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/7ba20394fefd4bf0b39886c69b28cf10 : SUCCESS in 1h 19m 01s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5d25a963068048819e469152fac9a845 : SUCCESS in 1h 26m 45s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5c4cf6c9aa83430da960a2cc8a00ce52 : SUCCESS in 1h 11m 58s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3b6bb40c8a514440ba0380dd578ebe1e : SUCCESS in 1h 22m 04s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/274b7548599b49e6be1895cf038f6051 : SUCCESS in 1h 11m 33s\n- openstack-ansible-deploy-aio_metal_quorum-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d38d67f9ec2f43d3ad9d53233f184232 : SUCCESS in 1h 23m 39s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0900950596c7499ea3cc346469b63a26 : SUCCESS in 2h 14m 22s\n- openstack-ansible-upgrade_2023.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/06cacc6c03f7473086e7875f2d452cee : SUCCESS in 2h 59m 39s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/85c5e6a3ac064ea29890005ea13ca8c1 : SUCCESS in 1h 46m 51s\n- openstack-ansible-upgrade_2023.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/01abeb0714624bd2882c07725863b216 : SUCCESS in 2h 23m 41s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/32611811448e4001b7e913dc88634b6b : SUCCESS in 2h 03m 49s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/96a32366ca3745caa481d9716c6bc649 : SUCCESS in 1h 51m 02s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e6aab8a3ba5e44fe99a47d9d0451fcb9 : SUCCESS in 1h 08m 35s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f4f827fe2a92447eab8cd5b7910b1d19 : SUCCESS in 1h 07m 03s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/6fa4e3c459a54f5cb899635fb69cb7f3 : SUCCESS in 1h 22m 32s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/534d201b46814b9b8ab67e215edb9351 : SUCCESS in 1h 31m 06s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d5c1a52ba3494fc29a4add2e3442b9f7 : SUCCESS in 1h 31m 37s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/ebae78e0b8d84d4b8959c4cab4a328b4 : SUCCESS in 58m 24s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/0498fbc2e5af439588b693400980bdb3 : SUCCESS in 1h 23m 50s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bf1f38fc525047e7bc566b4fc8234dde : SUCCESS in 1h 12m 24s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/969a888e51bc4ad393238415b5a868da : SUCCESS in 1h 15m 00s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/64b4a09ff09841e88b8656d9897fb0ec : SUCCESS in 1h 12m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4218af16504941d3b61b25ce75508924 : SUCCESS in 4m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2615c04e511a46bb911d1619f29ae09b : SUCCESS in 2m 55s","accounts_in_message":[],"_revision_number":1},{"id":"255fc3fda10a97607e008ba105b346cdb706d441","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2024-05-15 09:21:59.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4be07ee29d9303fbc6eacde4022d9e5127042988","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-09-24 11:20:31.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR **is:MAX** OR is:MIN\")\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"64563cb918ace62d0231578d065cf0cf5429d6ed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-09-24 11:21:22.000000000","message":"Patch Set 3: Commit message was updated.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"**changekind:NO_CODE_CHANGE** OR changekind:TRIVIAL_REBASE OR **is:MAX** OR is:MIN\")\n","accounts_in_message":[],"_revision_number":3},{"id":"fb5d270764beb1dbad14b6aba2fcd0b10eaacfd0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-24 13:50:46.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9bccd1e450284fd4a70b7ad02f320316\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fd99f9ccd47a4124a381b4d4eb655342 : SUCCESS in 6m 03s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/6b449c12ee4343b48147f84f4fa7d518 : SUCCESS in 1h 03m 14s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/d60d525cb63445bcba731fcf32fd052a : SUCCESS in 1h 14m 13s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0785b43de53f48668258690a494e73c3 : SUCCESS in 53m 22s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/695d461eea5b4c748fbfc89c145d0c1f : SUCCESS in 48m 32s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dc93b0325d874182af3bd42617b83c69 : SUCCESS in 51m 08s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f3f158bbb2584000bddb6ab6de810428 : SUCCESS in 1h 17m 42s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ebe1ccc19ddc40c8838f78b050b658cd : SUCCESS in 49m 18s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/24258303fb9b487d8cfb39ab73ba3b42 : SUCCESS in 1h 13m 18s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/36bf28c2e2914b2f9f4d183f8e7b5df9 : SUCCESS in 1h 18m 08s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-noble https://zuul.opendev.org/t/openstack/build/282cb07011f6451d8066429ff6a18679 : SUCCESS in 1h 25m 17s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/20925cbd5bf2488e9c3a8d6af5bccba6 : SUCCESS in 2h 02m 05s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/77fdcd549e63442cabec5eaf3bba7c26 : SUCCESS in 1h 43m 29s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7de5d89c333049599322c8cccddc4bc2 : SUCCESS in 2h 10m 31s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/35f3e7b790ec42e5992b83087efdc47b : SUCCESS in 1h 46m 02s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/354d42aa8d7f4e36a84eb8ac66681cb9 : SUCCESS in 54m 10s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/54f1063621354ab8a85a315a31b0bdbe : SUCCESS in 57m 03s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/d80b86c2e91e492fb99a4111bec2e31a : SUCCESS in 1h 22m 25s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c9b6b719041644cb9721b902f1f4603f : SUCCESS in 1h 13m 30s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/407c064bf04e431aa58965101259f6b4 : SUCCESS in 1h 16m 36s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/14dc227284044ecf9dca70bed0c06643 : SUCCESS in 1h 03m 51s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d86c500c7d64493aa6106c8a168386ab : SUCCESS in 1h 06m 41s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/819a199aec494cdfba1d9dbd1755aac5 : SUCCESS in 1h 11m 01s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/37b4bc459eba4017912e98973db0ec29 : SUCCESS in 1h 16m 51s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d0fc71d7b4fc4df49b2fba01a1d67149 : SUCCESS in 53m 13s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e79ef28a3f1f4e37b6c776c385dc99c5 : SUCCESS in 1h 10m 51s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/37836c72e4e741cf969b8052171ab504 : SUCCESS in 52m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a933beef06c4ef4934ddb91621489ff : SUCCESS in 4m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6f0de2db332c4e64b7cc0aca9089625a : SUCCESS in 2m 52s","accounts_in_message":[],"_revision_number":3},{"id":"14ed24611d7ed5ce473f2955adf8419c80967455","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2024-09-24 17:07:22.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"ef5632909a1eea059201337407d328c72fed5185","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2024-09-25 07:08:45.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"a9dbca3bb41979d443465b1400e5836046630e3e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-25 07:10:51.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"e6d1b7395bf0a287542f908886c385aa1c9d3d78","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-25 09:08:08.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cda7596f5e6c4786bff3e17f3ad84e6f\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6591de9181a045268829cdb64eb5df58 : SUCCESS in 6m 13s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/d3f0aab17c7d43a2ba4a990f129632c4 : SUCCESS in 1h 19m 41s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a2fe3bf346124d858a645f214ad906c7 : SUCCESS in 1h 21m 56s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cb6e27928f094675a818926553eacde8 : SUCCESS in 1h 12m 50s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b45ddf7dea3d4def85c7bb731bf43b9a : SUCCESS in 1h 22m 41s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f865e525bba84b1eaf3141266caf6b84 : SUCCESS in 1h 40m 18s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-noble https://zuul.opendev.org/t/openstack/build/56982acb3c7b4a84ab1448241ea871e6 : FAILURE in 16m 34s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/22383c44606a470a94dbdf63a21ba01f : SUCCESS in 1h 02m 29s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f408335996f6466ba94526339510469b : SUCCESS in 1h 23m 54s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/85df2084adef4befbeb427e226edd593 : SUCCESS in 1h 19m 27s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/531ec8d7bb1c4411b67567e473b1f73e : SUCCESS in 1h 18m 03s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4e3c5cd7d03b4f5987d8654c64e521a8 : SUCCESS in 1h 22m 11s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/b18dada1965b4ffd97bb41cf12aacefb : SUCCESS in 1h 12m 57s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/394dfe8d65834ae3925b9975cc4ad6ca : SUCCESS in 1h 06m 55s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2e2907c1d36a44b8a73eb269ab99b2e7 : SUCCESS in 1h 14m 36s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f6dd819710b64ff19cfe7d67cfd0d8cc : SUCCESS in 1h 15m 05s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/00b41d5ec21e4d089d4040a3e4810d09 : SUCCESS in 1h 18m 52s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e9f572c27626440b990846c0a27f1d90 : SUCCESS in 1h 19m 20s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c20fa885e48f4d5cbdd8a77e03a8661c : SUCCESS in 1h 12m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec85d40230e5462aa64fab21e46b07a3 : SUCCESS in 3m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c8032ea6f9244a33aeb6b45a741ef856 : SUCCESS in 2m 33s","accounts_in_message":[],"_revision_number":3},{"id":"dad6fc3610d2a72ae3c7e28801e1bb88c2614be8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2024-09-25 15:49:15.000000000","message":"Patch Set 4: Commit message was updated.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"**changekind:NO_CODE_CHANGE** OR changekind:TRIVIAL_REBASE OR **is:MAX** OR is:MIN\")\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-2\n* Workflow+1\n","accounts_in_message":[],"_revision_number":4},{"id":"14b4b1ab14cecfee585364957ebe271637bdb2aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-25 18:07:28.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5bb2813bc7594b6c899f01739f3dcf60\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/108262a80b0f47c6a0848b3f50406662 : SUCCESS in 7m 47s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/417c8d2a35634a9094c553e4ef8965e9 : SUCCESS in 1h 11m 58s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/91e92596ce97409280586061f5fdfee6 : SUCCESS in 1h 08m 41s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c15f297dd5da47e7a664b3fdaed2bf1a : SUCCESS in 1h 16m 53s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6ab0c94f3ebc400682728e01fa5d1f7e : SUCCESS in 1h 19m 42s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/171d1c5f7e3d4963a93ac68b8911daba : SUCCESS in 1h 21m 35s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9c6b0737121043bcbe7d86a3bb46b181 : SUCCESS in 1h 15m 47s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6272775b5a1047e08d656fbd2a4280dd : SUCCESS in 1h 08m 35s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/850e730c4f124f5e83456596a99bdd4d : SUCCESS in 1h 09m 23s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5ace618c7c0b4a36aa137016538e80bd : SUCCESS in 47m 24s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a46bd22d8df74f87af902a80118f551e : SUCCESS in 1h 13m 06s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9c8b0c82af24461a852cfb8cc6a1e206 : FAILURE in 19m 12s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/684ac95cd6b64ae68be69fd058ef2b64 : SUCCESS in 1h 39m 21s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/828ddf177b0f4c07a7b6e84a1918847a : SUCCESS in 2h 09m 10s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f948f5562bfa481b8b384b48186c49c1 : SUCCESS in 1h 53m 55s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4615f4100cb24f839e5740628fe74a5a : SUCCESS in 56m 31s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a267e0436fa842c9a6da9ba219e97bb1 : SUCCESS in 57m 24s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/f3075c12bb3848ee825931c28e78b005 : SUCCESS in 1h 17m 45s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/aced99c8748f451eba47989abf8ba110 : SUCCESS in 1h 10m 31s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/78a5528bdddb4a1abaa54ecd73aca2e1 : SUCCESS in 1h 26m 08s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/4a976a285c464d7cae1747e3d970dcc3 : SUCCESS in 1h 08m 28s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8669d2bc380b489aae6819c3e5921673 : SUCCESS in 1h 10m 51s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3a8caf419417469e8f4ee2c53a7f6e9d : SUCCESS in 1h 14m 34s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/768ef1f14dea400eb6edd6c95b2e3ea0 : SUCCESS in 1h 04m 12s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0b34ec617e9d405fa608abf7cee0b509 : SUCCESS in 49m 06s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2f2db9b825414030a3be9f350908c480 : SUCCESS in 1h 16m 16s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/78a9748e86b144f390ea68891fe286eb : SUCCESS in 1h 07m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f15b049c558412a8522609caa1b4d86 : SUCCESS in 5m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/07f8a651019e45809c8148113b4f3b64 : SUCCESS in 2m 31s","accounts_in_message":[],"_revision_number":4},{"id":"795670d95dda7dab815f1e4e16112252736192ce","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-09-26 06:42:05.000000000","message":"Patch Set 4: Backport-Candidate+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b019d92f0fcbd9ee789460ca31f1896454b35ba4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-26 08:50:04.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5e03dfa1542044b28378f316e248d452\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1a2684454709405f830e3781ef0058aa : SUCCESS in 7m 45s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/9f5d8beeb1244d929bad8a699cc9517e : FAILURE in 1h 13m 51s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/571e38d3705e40a3832fb1bdbe6c905c : SUCCESS in 1h 16m 43s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/977170476f7a4162bd7581fd4df53f10 : SUCCESS in 1h 21m 02s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3d5b734c86754f84aacdb636a324741b : SUCCESS in 1h 01m 51s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/201ba7a4a98d43dc83fd946400a76dd2 : SUCCESS in 1h 10m 12s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/acf09854cc794bdfb5d1a9c1a1226919 : SUCCESS in 1h 18m 42s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e7580e12b8b043f1a2ea8541ce295321 : SUCCESS in 45m 15s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f4aa12a60ed1486ebad172c61035d8bf : SUCCESS in 1h 15m 46s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1a021881c5134f14ae5e5820d4cd26d6 : FAILURE in 56m 31s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-noble https://zuul.opendev.org/t/openstack/build/479a94bd9c0741d091ff0ea437d4de6c : FAILURE in 59m 36s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a83d738631224cb5a251f15e4ed3d7c9 : SUCCESS in 1h 37m 07s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1bd26fdb4acf478fb392dd859b4e0d64 : SUCCESS in 1h 46m 57s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3270ba34f4884803a5c93d9cb9651ce6 : SUCCESS in 1h 11m 14s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c25cfc12824d43ed93da92655ce022da : SUCCESS in 1h 41m 05s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9af48f3112d34aeba46d1b59e577ca92 : SUCCESS in 51m 18s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ad9272cd5cf5478c9f31aa3164c42ef6 : SUCCESS in 36m 03s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/80205733852347178e246ac72555405f : SUCCESS in 1h 18m 04s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5ca5bb4ee1a44dfba0996e34554fa1af : SUCCESS in 1h 20m 31s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/fbd5844c54644bfd8ecb5fa4bc1c77e6 : SUCCESS in 1h 25m 24s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/fd97409721b94e7893d0fd6120d98539 : SUCCESS in 1h 16m 17s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cef01454768d47578c5f1a000d1aa071 : SUCCESS in 1h 18m 47s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7eb77260c9014199bd7bc15731a0ca6b : SUCCESS in 1h 18m 54s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/06b0891dbfdf435f812a4bb10065cfad : SUCCESS in 1h 11m 08s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e8571a84e0e443aaba06b90c394cf2de : SUCCESS in 1h 25m 14s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/00a200d0ea614eb4b944d69ad53a8f99 : SUCCESS in 1h 12m 47s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/600cc87d666f46feade3086d1db4e0b1 : SUCCESS in 1h 02m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/914505173ef54b008d560848c781c76e : SUCCESS in 4m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/df600999416949ea837e83dacd897bdb : SUCCESS in 2m 29s","accounts_in_message":[],"_revision_number":4},{"id":"394a5c8c32cccf045fd4f0030da33198f157b083","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-09-27 07:24:30.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"4f7d00096701436ac03ef080384538639c33e8f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-27 09:52:19.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5650b4539a044be18429d5f06b215db3\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2301b317cbb1457fa731e59ac87d6fff : SUCCESS in 6m 48s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/572d687cd78e4b52907866c8ea120e04 : SUCCESS in 1h 07m 52s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/f4147aa5ec374adfb17afe4980c334fa : SUCCESS in 1h 11m 59s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/012a72a907ff497a981fe134f268ca31 : SUCCESS in 1h 23m 43s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/463f37208eb94d24b9572a9ad667e703 : SUCCESS in 1h 07m 57s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0852d7b314284f0fb2b818bd66830e62 : SUCCESS in 1h 19m 02s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c3eeb5e8ad3a4821a4e65c7134de7183 : SUCCESS in 1h 24m 55s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4ee7d6b0a1194d4d9d7a1b527163496d : SUCCESS in 1h 11m 16s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a97c31248eb349e28ee755c8e36c1ccf : SUCCESS in 1h 19m 04s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/82f985cfac384a3cb098c8060658efbd : SUCCESS in 1h 19m 11s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0c2e9857e0254894a9c6d96c384093ef : SUCCESS in 1h 21m 31s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f051864a1b394d6eaefb57261a39ea13 : SUCCESS in 1h 51m 03s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1a8e79711c83462282b3b3a43535de31 : SUCCESS in 1h 35m 23s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ae44ffa8328f479aa1512d8c65895096 : SUCCESS in 2h 09m 04s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d261ba94cfca4fe7b1822394f412a501 : SUCCESS in 1h 56m 02s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e26abea6180f4847a79d8de42a7e49a9 : SUCCESS in 41m 07s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/86f29b4bb4694e70a7eb4f4d0e1fccb3 : SUCCESS in 59m 51s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/d9e0c522d4ce4107858492279dcbb466 : FAILURE in 1h 15m 19s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/425706fd94dc457bb99030b407e4fcb1 : SUCCESS in 57m 21s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2ad7f3ba9792477699a7b3a3cb08a43c : SUCCESS in 1h 17m 19s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/df942ff7b3404e7294dc3ecd406dbf53 : SUCCESS in 1h 10m 55s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2da6d6384e7b431fb61ad55947818f65 : SUCCESS in 50m 00s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/56c5d32329814510b5936904b02d8969 : SUCCESS in 50m 19s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/88dcf10a2ba94405beaa3eb8c7d49cb8 : SUCCESS in 1h 15m 56s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/52cd1ac404064bd9ae6ad9b2895daa1c : SUCCESS in 1h 08m 45s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/509d6bf75c984308bcdc90ee5fb28ccb : SUCCESS in 1h 49m 06s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/59196bc9d82e45a380f3113397d83a5e : SUCCESS in 1h 09m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/417e726542454b9f990b837a3e1c14a3 : SUCCESS in 4m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7110e09e3407483980a801a3f23879ab : SUCCESS in 2m 59s","accounts_in_message":[],"_revision_number":4},{"id":"a3d249c91f624c0f9a8822c41deff97d7242c8fa","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2024-09-27 10:04:39.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"f99987b80a1d370d88320f6df9e3d62bc9cb560a","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2024-09-30 09:55:22.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"a5b1747c7561856be74ad87f1f65c077cdc4bba8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-30 09:57:23.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"ea234e879e33d530befb6f0187b8e2b3cbcf5df7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-30 11:44:30.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/68b322579326463d80e702cdfa7efeff\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b4c1cecfb8c64edda385cd4b0b35ea91 : SUCCESS in 7m 18s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/20d1c4182610490bbc2c9a0ab2a2af15 : SUCCESS in 1h 09m 05s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2125d9a9a3c447afb7932a113a4c7530 : SUCCESS in 1h 20m 50s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e6c209e2df5b436c9ee93cd8b5aea903 : SUCCESS in 1h 16m 48s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f700412763ef40c08c176186a04ae8a5 : SUCCESS in 1h 05m 16s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6679edcc0a384c5e94cd68d19b5124e5 : SUCCESS in 1h 13m 13s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-noble https://zuul.opendev.org/t/openstack/build/390d6e9d3397425387909c7cb35de7da : SUCCESS in 1h 24m 02s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d55da1b994094be68e7687bc821b8d05 : SUCCESS in 1h 22m 44s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e7e4deb1d8b74f4bab1fe4408cccbccd : SUCCESS in 1h 17m 58s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/86cb0757dfe243ddbb4faa4a7d1d3825 : SUCCESS in 1h 18m 35s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/bf5ecbc58e1040a2a565ac09f1618d5f : SUCCESS in 1h 27m 02s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2bbcbcfc8c6d4aed9a52e4216a55b9c6 : SUCCESS in 1h 16m 55s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/671c1912be0c4ba18bfc32953ef74538 : SUCCESS in 45m 55s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7bf0f0c4cc214d2ba82f18b5ce5d9fe0 : SUCCESS in 1h 09m 39s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c82c2eccf2854332967a3297704c320a : SUCCESS in 1h 03m 27s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/de2f887f3fd041d69edee409383b5b9e : SUCCESS in 1h 12m 05s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/62070a647531404196bfb54e61818b8e : SUCCESS in 1h 12m 00s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e1b48e6403d9402d99a19332daef9016 : SUCCESS in 1h 19m 21s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/39509518a2f34921a4f81f24655301aa : SUCCESS in 1h 14m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e9a5bff24294300aae82d98c3484f05 : SUCCESS in 4m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/54fd49c91e0b4d54940b5753810ac190 : SUCCESS in 3m 44s","accounts_in_message":[],"_revision_number":4},{"id":"edc0f3aacd5422e6da845e35860ca2f6a3657bdc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-30 11:44:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"53a9760efeec6b65291c314ab7b6a54b4517c6f5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-30 11:45:27.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad4cf11db1ba46c9b879fc77903a85e3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1044087671ae48e8a2514276ab2a6754 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"c3aa3c36233b79b3391e95a4ce100bde3e667391","revisions":{"a3933286ee20c108d2f885ec9bfc6400038b073e":{"kind":"REWORK","_number":1,"created":"2024-05-14 11:41:39.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/63/919563/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/63/919563/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/63/919563/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/63/919563/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/63/919563/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/63/919563/1"}}},"commit":{"parents":[{"commit":"98e41b6eee98f0a3b7965f83a911604e67ef9415","subject":"Add qos_prefetch_count to variables","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/98e41b6eee98f0a3b7965f83a911604e67ef9415"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-05-14 11:41:30.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-05-14 11:41:30.000000000","tz":120},"subject":"Change example to contain domain name instead of UUID","message":"Change example to contain domain name instead of UUID\n\nRelated-Bug: #2065680\nCloses-Bug: #2064718\nChange-Id: I5d9505ed4b385673cc719e9821ee2e1c2ba5c754\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/a3933286ee20c108d2f885ec9bfc6400038b073e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/a3933286ee20c108d2f885ec9bfc6400038b073e"}]},"branch":"refs/heads/master"},"c74fb0ff002637d01f92a7a2b3766f7ac156b36c":{"kind":"REWORK","_number":2,"created":"2024-09-24 11:20:31.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/63/919563/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/63/919563/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/63/919563/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/63/919563/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/63/919563/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/63/919563/2"}}},"commit":{"parents":[{"commit":"98e41b6eee98f0a3b7965f83a911604e67ef9415","subject":"Add qos_prefetch_count to variables","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/98e41b6eee98f0a3b7965f83a911604e67ef9415"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-05-14 11:41:30.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-09-24 11:20:31.000000000","tz":0},"subject":"Change example to contain domain name instead of UUID","message":"Change example to contain domain name instead of UUID\n\nRelated-Bug: #2065680\nCloses-Bug: #2064718\nChange-Id: I5d9505ed4b385673cc719e9821ee2e1c2ba5c754\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/c74fb0ff002637d01f92a7a2b3766f7ac156b36c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/c74fb0ff002637d01f92a7a2b3766f7ac156b36c"}]},"branch":"refs/heads/master"},"c8fafa504400af40374aa1f6e5ec2cda419329c2":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2024-09-24 11:21:22.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/63/919563/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/63/919563/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/63/919563/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/63/919563/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/63/919563/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/63/919563/3"}}},"commit":{"parents":[{"commit":"98e41b6eee98f0a3b7965f83a911604e67ef9415","subject":"Add qos_prefetch_count to variables","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/98e41b6eee98f0a3b7965f83a911604e67ef9415"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-05-14 11:41:30.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-09-24 11:21:22.000000000","tz":0},"subject":"Change example to contain domain name instead of UUID","message":"Change example to contain domain name instead of UUID\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-openstack_hosts/+/930272\nRelated-Bug: #2065680\nCloses-Bug: #2064718\nChange-Id: I5d9505ed4b385673cc719e9821ee2e1c2ba5c754\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/c8fafa504400af40374aa1f6e5ec2cda419329c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/c8fafa504400af40374aa1f6e5ec2cda419329c2"}]},"branch":"refs/heads/master","description":"Edit commit message"},"c3aa3c36233b79b3391e95a4ce100bde3e667391":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2024-09-25 15:49:15.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/63/919563/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/63/919563/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/63/919563/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/63/919563/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/63/919563/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/63/919563/4"}}},"commit":{"parents":[{"commit":"98e41b6eee98f0a3b7965f83a911604e67ef9415","subject":"Add qos_prefetch_count to variables","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/98e41b6eee98f0a3b7965f83a911604e67ef9415"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-05-14 11:41:30.000000000","tz":120},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2024-09-25 15:49:15.000000000","tz":0},"subject":"Change example to contain domain name instead of UUID","message":"Change example to contain domain name instead of UUID\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-openstack_hosts/+/930272\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-rabbitmq_server/+/930446\nRelated-Bug: #2065680\nCloses-Bug: #2064718\nChange-Id: I5d9505ed4b385673cc719e9821ee2e1c2ba5c754\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/c3aa3c36233b79b3391e95a4ce100bde3e667391"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/c3aa3c36233b79b3391e95a4ce100bde3e667391"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
