)]}'
{"id":"openstack%2Fopenstack-ansible-os_keystone~921845","triplet_id":"openstack%2Fopenstack-ansible-os_keystone~stable%2F2024.1~I2e801fade862608382aa0adb868eff3f44ed938b","project":"openstack/openstack-ansible-os_keystone","branch":"stable/2024.1","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2024-06-13 14:59:42.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}},"hashtags":[],"change_id":"I2e801fade862608382aa0adb868eff3f44ed938b","subject":"Define oslo_messaging_rabbit section if either RPC or Notifications are enabled","status":"MERGED","created":"2024-06-12 11:19:14.000000000","updated":"2024-06-13 19:23:19.000000000","submitted":"2024-06-13 19:22:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"921845","cherry_pick_of_change":921796,"cherry_pick_of_patch_set":1,"meta_rev_id":"8c83907d2ad38492d95d62149110e949114251e6","_number":921845,"virtual_id_number":921845,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-06-13 19:22:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"all":[{"value":2,"date":"2024-06-12 11:56:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":2,"date":"2024-06-12 11:19:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"all":[{"value":1,"date":"2024-06-12 11:56:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-06-12 11:56:00.000000000","updated_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"reviewer":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"state":"REVIEWER"},{"updated":"2024-06-12 13:53:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"951c06371b71c576205c5828ad517c28ab574f00","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-06-12 11:19:14.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"6c5133beca840b65f529a41b3318950a89e84dba","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-06-12 11:19:30.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e85e19b8a1f9d7d1c12871ca32c6274974b69064","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2024-06-12 11:56:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"16a1a24eba6e094d6d08463ecca03f7512b66896","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-12 13:53:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/daccbcb62ca64ce6b84d454a7eb98314\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c49b717ec6694a7596a10a2358681103 : SUCCESS in 11m 32s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/5cf756302bfb4d57a7b10fab8b0ef539 : SUCCESS in 1h 21m 04s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/299b16e9f345491ba84def522a5957f1 : SUCCESS in 1h 21m 30s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/8113c36cdd3c49faa3d5da708532aadb : SUCCESS in 1h 27m 26s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/993a0b1c8ec340958eba63cec23b23f0 : SUCCESS in 1h 29m 03s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8bb623a2d57742128df47d3a4f9ffca9 : SUCCESS in 1h 19m 52s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e63ba58eb7694650852fc9d26a0f3e17 : SUCCESS in 1h 27m 01s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9b54f2ab40ef46c08812ed7cc5718b43 : SUCCESS in 1h 33m 19s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5c5d0ae9e06d40dfbf3ecc1d73237aa4 : SUCCESS in 1h 26m 00s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f4c357b810a44bc2981c137bcf679890 : SUCCESS in 2h 12m 55s\n- openstack-ansible-upgrade_2023.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/85faf1ee1a4942b7bdeb1c921a951890 : SUCCESS in 2h 27m 15s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/696fb419005b42a78e64d48081d39061 : SUCCESS in 1h 55m 25s\n- openstack-ansible-upgrade_2023.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6ce11bd8f41242a4b2a89b1ce0260fc8 : SUCCESS in 2h 19m 59s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/db109daf58d54f7aaec031ec16707d48 : SUCCESS in 2h 26m 11s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f9e67ddca174497784fd94fa513e91b0 : SUCCESS in 2h 01m 20s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/003c4239e5e14bf392f0280d9ee50efc : SUCCESS in 1h 10m 52s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c0393b954d6345f79461b1c2e4012b55 : SUCCESS in 1h 07m 21s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/cb5db19df8674bc0a98ce5118bb194c7 : SUCCESS in 1h 21m 48s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/13dfa8a8d8404371ac31ef537413659d : SUCCESS in 1h 28m 57s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/67ddbe8294c341669cf2a0e4e5b1b43b : SUCCESS in 1h 29m 18s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/9bc54f258c0f40cfbafa10485732214a : SUCCESS in 1h 14m 01s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/6d5c46e04c4948beb1a9080eb13b29ac : SUCCESS in 1h 25m 09s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b8480d55636649969b6cddcb05c997c7 : SUCCESS in 1h 14m 44s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a1e39759f8eb4db8b228572c0c80c6e8 : SUCCESS in 1h 25m 16s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0332c4be90164c069c7e6d806ef721d9 : SUCCESS in 1h 14m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/576f25f5a8a3437798fffef2dedde61a : SUCCESS in 4m 40s","accounts_in_message":[],"_revision_number":1},{"id":"bdf9c242b6e245babd7085abb1b5290961289d4f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-12 13:56:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a1aba56cbfd99c84071b4c5c1c0f353a9db71df1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-12 15:28:13.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3bf5bc850f324d65a1c6498b07f5c331\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f514bcdd36b444e2836cce5455e2f4e9 : SUCCESS in 12m 48s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/79427bc2e70a4ac08fe4479d4d92ed0d : SUCCESS in 1h 15m 07s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/01cd2968c7d549c7bfbd81caff79d73c : SUCCESS in 1h 18m 57s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6454fc19a7cb44ceb13a537f4e7e49ed : SUCCESS in 1h 18m 47s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1cb9d548f1104554be3f042ec1e15c47 : SUCCESS in 1h 13m 07s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/15aa7c5481634df8adebc4196ca81fca : SUCCESS in 1h 21m 00s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f0f59d07ace44f1098f433e01d5b6b59 : SUCCESS in 1h 25m 34s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0c1986b3c7a94ed38ca3b06b6a6112c7 : SUCCESS in 1h 26m 22s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/eaf06c30ecce488ba8e844bdae8e9ec6 : SUCCESS in 1h 26m 46s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/514938be404d4c9eb223e85d058418c9 : SUCCESS in 1h 26m 28s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/eff5153dd1924c2abe0c57229c66b485 : FAILURE in 1h 07m 48s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/d291adb4a8fd42798f582a45911f53ff : SUCCESS in 1h 13m 06s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/80de4365363144f984a7f182f8b4e797 : SUCCESS in 1h 15m 43s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/124a47ba4bf34b2e896474440d6cd7a5 : FAILURE in 1h 08m 28s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/66b83e8e4a324e7887c16665fb70380f : SUCCESS in 1h 17m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/daab3b35e8d445689dd84cddbfe692a2 : SUCCESS in 4m 51s","accounts_in_message":[],"_revision_number":1},{"id":"dbcb1ce2e28b2735ba1f9bb6ae10452a5942ae7f","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-06-12 15:29:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c3cc7b0a7742353601335ba3ed3f9ddc9b91057b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-12 18:09:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd0cb8d122054504891c62282b5eafb5\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5b574f6be0a04fb5a6163a164420ffae : SUCCESS in 12m 11s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/d2a1f7a0a078477f9a7e71b3e4f6cff0 : SUCCESS in 1h 14m 24s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/32945142f0e74e3abb5e607bea22c8f9 : SUCCESS in 1h 26m 54s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/3a6896d0b2ff42f0ad7f2b6be5a52ccd : SUCCESS in 1h 26m 17s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7def3d550c5b48ffbb4918eeda0ff946 : SUCCESS in 1h 31m 35s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f8035c1bcbfa43dab0ded9d83f618009 : SUCCESS in 1h 23m 20s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/bd8d6971fa434caf8e16cd72ab6b32f3 : SUCCESS in 1h 37m 06s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/afe3fc725ec848258938a6cb9f6685bb : SUCCESS in 1h 22m 33s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f8bbbc5164d4424d92227afa1b348b18 : SUCCESS in 1h 19m 27s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a1f2614251b64378a6b9a2c5ebc74a0e : SUCCESS in 2h 05m 09s\n- openstack-ansible-upgrade_2023.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e6f6ef558d82457b94bd39b9babb03ec : SUCCESS in 2h 31m 29s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/66b8fde2cd134557acd7f333d35d380b : SUCCESS in 1h 55m 28s\n- openstack-ansible-upgrade_2023.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b2a48b7cd2234bb5b3ea792ff2d57d76 : SUCCESS in 2h 10m 22s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/aa6a7a209e2d41b0b36885f7d65b506e : SUCCESS in 2h 19m 08s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/48421061d28a41dd9d444aa755d51f52 : SUCCESS in 1h 48m 23s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/8aaa034020f7400bac0aa2f69f8bbc88 : SUCCESS in 1h 01m 58s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dc97806df3334416abc4c9991bfb6b0e : SUCCESS in 55m 34s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/5fc5859da88b4087b0c572defda45f1e : SUCCESS in 1h 27m 24s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f411ca49271644c294e768071473e99d : SUCCESS in 1h 27m 10s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4f48af17640f40d5b36ea9879e3f5693 : SUCCESS in 1h 26m 07s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/a817ced61799475b9ac07d8f3f1256e0 : SUCCESS in 1h 09m 37s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/885fbb9774d64f818ac81a30dec5dad9 : SUCCESS in 1h 17m 17s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6cbad86e24754654898ecc4dc4eac21e : SUCCESS in 1h 07m 58s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/51a87ceb3b464f13ba5c512146107bdd : SUCCESS in 1h 08m 29s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d393e2af4fb749929aa574bd80d88b24 : SUCCESS in 1h 12m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d787eedaea4a403e9ec33f4fd3f2f287 : SUCCESS in 4m 13s","accounts_in_message":[],"_revision_number":1},{"id":"2161964384a55806af00d93e8f0475c32465d850","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-12 18:11:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1cdca51f7f785c68b1a7658b421de01a0e0dee7f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-12 19:54:31.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/45a69630640349078a6d3bb06a019a64\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bc8da6343f5f45a6a26b845896e729c5 : SUCCESS in 10m 26s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/74be6f8c72a745caa2125fcdc16ba1a9 : SUCCESS in 1h 22m 40s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/dc6cb470ad3b40baafc496fcddbc367f : SUCCESS in 1h 15m 29s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1e402a0c8e6a4e499b2dc128323e9ff9 : SUCCESS in 1h 30m 42s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/23e8b6dbac0f49058e937f16b61a59c1 : SUCCESS in 1h 22m 07s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c5472f9e28774bb5b22fea451d95da9e : SUCCESS in 1h 20m 29s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/785f3e6eb19849b0ab3ec142ab97cb6f : SUCCESS in 1h 22m 00s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a9589f62474b421399027546c50eef7c : SUCCESS in 1h 19m 08s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/48e28eae0c894696b869dd49e267ecc3 : FAILURE in 1h 32m 28s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/27924cd490f4473b8ac4aff27e337dd1 : SUCCESS in 1h 29m 07s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/a623a9b7ecb74b5f9d33d019e8447354 : SUCCESS in 1h 10m 43s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/3bc064df5fe84a81bfab5a13006ed691 : SUCCESS in 1h 22m 06s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/33914df14bbf42e1b50abb5ea7a570dd : SUCCESS in 1h 12m 06s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/77c49b773fa74f8aa07a082364d1012f : SUCCESS in 1h 16m 16s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/95d768e11a28477086cc560310ead2eb : SUCCESS in 1h 26m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51bfec5775894394a3d473c7baf1833e : FAILURE in 3m 16s","accounts_in_message":[],"_revision_number":1},{"id":"6b858a367a83fe8d9b40159feaf24a9e57748da0","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-06-13 08:49:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8430e012e2b39acc447e1b62519657b790227498","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-13 11:12:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/139880055d0148ef83a90993e28ecdc0\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4b7922debdc84f1e808698ab7db8d109 : SUCCESS in 13m 09s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/76592ccafe3a4bc581501eee8f54e29f : SUCCESS in 1h 11m 13s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/5731dcc204ac4f6a9bcc2cfdb408e731 : SUCCESS in 1h 22m 39s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/c479a71ae64a46e78bce9690cec0dfe2 : SUCCESS in 1h 17m 26s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b89fd7130f954534af9ad9e6a2476a0b : SUCCESS in 1h 26m 22s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d9c0d26c394a4c8fb34e246e26921d44 : SUCCESS in 1h 29m 41s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/886076af925f42a0b4da92a64be04ad9 : SUCCESS in 1h 28m 53s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f23e6636437343429d9c550bed07ab98 : SUCCESS in 1h 29m 51s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7c79ee82cc06453799db38fabc55ebc7 : SUCCESS in 1h 11m 41s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3f84d79a1c5949deb46acc041b7195c6 : SUCCESS in 1h 53m 54s\n- openstack-ansible-upgrade_2023.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/eef5a64bf30f4cd28ebd53241558ed64 : SUCCESS in 2h 15m 03s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/08ca254d48bb4500955f5bc69eb2ebff : SUCCESS in 2h 04m 00s\n- openstack-ansible-upgrade_2023.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2e1e1e5aac69447ea6cf18c217eb2efd : SUCCESS in 2h 02m 30s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5047a277afa545b6ad543d93569dac6a : SUCCESS in 2h 01m 34s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8c9c87cc0d6848059e840a8c84899e6f : SUCCESS in 2h 03m 26s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/944cfdb5eb9940c0ac06fcda8ae6ff32 : SUCCESS in 1h 02m 13s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4af85ca16c2948788bcab0b4d028a981 : SUCCESS in 56m 28s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/ea631c69eca74e029954af518e15bf73 : SUCCESS in 1h 17m 48s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0b9b14f709c6439a98edd39737c60764 : SUCCESS in 1h 33m 55s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/411cdbfd48514cf8bd092779e214ab52 : SUCCESS in 1h 32m 19s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/24839af5f34248c1b141c9d7d698edc0 : SUCCESS in 1h 17m 34s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/cafd513139144360af59eeab38a77f0a : SUCCESS in 1h 11m 26s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a43377a9065c4247a077b80cc91ef103 : SUCCESS in 1h 11m 01s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c02a66009e904ff2ba585fb548a5a9e5 : SUCCESS in 1h 28m 53s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ff7009124782452d8b743108e83cb0e7 : SUCCESS in 1h 16m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad0e164e68d940868d15db64d057e9ff : SUCCESS in 4m 28s","accounts_in_message":[],"_revision_number":1},{"id":"4d2c8ca7a8d581100b8873acc0f039ed74ad34f2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-13 11:14:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"84f9e9a7daca9b552bf02021eb92f5151e703a6b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-13 12:57:06.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5346d35ef99e42e19493cecb30b5048a\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f9687bda200742a6856db15d29e17abe : SUCCESS in 12m 11s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/ad77ffaf5d3e4b879a03782a8b5b0aee : SUCCESS in 1h 29m 38s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/32bb11e591db4ac68741b3416ea03912 : FAILURE in 1h 20m 22s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c36f6bfd5e3845129c36d2827b5fa45b : SUCCESS in 1h 29m 20s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/70ad00634aac45fd995edaa6434f6197 : SUCCESS in 1h 08m 54s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9a9dba1edc044b01b812dcda963119d8 : SUCCESS in 1h 29m 40s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a5f5589b3e054ae096acab0009bbc168 : SUCCESS in 1h 23m 04s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a8044e1b947b46f5aff8e15010a8caf6 : SUCCESS in 1h 25m 51s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/84ce2e341c1d46aa9dad3002912597f1 : SUCCESS in 1h 27m 18s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f8b3abefc0ea4076b88b804e5159ce62 : SUCCESS in 1h 39m 15s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/c2c7803fa8b44cc6867ec7f8c256a66f : SUCCESS in 1h 08m 15s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/1f78b38d2aa04b0f8fbca1a991de701f : SUCCESS in 1h 15m 22s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/080f6031b7a848c28f1521e8fd1ddea6 : SUCCESS in 1h 18m 28s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/addeb9a8cdcd458bbc920e58c808311f : FAILURE in 1h 30m 39s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/246ad266c5be4386923ba2a4d6ba8203 : SUCCESS in 1h 21m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84385eb49f5940dd9ca4b11f2c4eef07 : SUCCESS in 4m 19s","accounts_in_message":[],"_revision_number":1},{"id":"b2476b028e7d861c86c0c734856252aedc0f9f7c","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-06-13 14:59:42.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4083930627686403e5c2e7ea95b86c7474e6f6dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-13 17:37:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/950b00cf2771461ba594230ef9cca798\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c5f64b5d2d634a659850c20aba5f68bc : SUCCESS in 10m 40s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/6d09601a04e04dd4b379861adb3af5b8 : SUCCESS in 1h 16m 30s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/f961ad240422440e885a0bd145501d6c : SUCCESS in 1h 23m 38s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/0b65c53c67964586a9f9bdd39b0d4413 : SUCCESS in 1h 18m 43s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5133cc2387c64f05a134a6c14eaed9ad : SUCCESS in 1h 26m 55s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ddb5d6803330445693246000948464e1 : SUCCESS in 1h 24m 20s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2cfb9b5267d4420da597f4a3dfedbbc5 : SUCCESS in 1h 41m 34s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/873bccac51b54793b6e62782b413eae2 : SUCCESS in 1h 15m 39s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/724c67ec356c4e6c8b03a8b792adc214 : SUCCESS in 1h 22m 50s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ea90fafa3bc04c00b01c54f374cac985 : SUCCESS in 1h 45m 53s\n- openstack-ansible-upgrade_2023.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5dbe0f8aabe44fc78c51bd6c87a9050e : SUCCESS in 2h 30m 33s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/86571c451a424afa87c7cfa6ca424e2e : SUCCESS in 2h 05m 04s\n- openstack-ansible-upgrade_2023.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fae9fb82d37546d8b829894ed4bc86bb : SUCCESS in 2h 17m 29s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0f7796087c7948dc8a5291888058f817 : SUCCESS in 2h 06m 47s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4642c8f6080b4cc48b3758a213267a1c : SUCCESS in 1h 42m 24s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5b438cf30aa84bc6a2b690104d066477 : SUCCESS in 1h 04m 29s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/433c2931726c465c8667fc7f73b988de : SUCCESS in 59m 38s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/a453df595ec14616b89194e9f5a7dd94 : SUCCESS in 1h 23m 58s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/956b90641b924bf798c1d7ccc82ea01a : SUCCESS in 1h 15m 44s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ec7359b74a0c41668b65c6ec91d21a77 : SUCCESS in 1h 30m 35s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/45305d3e5116457f9ee82b74b132eba9 : SUCCESS in 1h 11m 42s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/ffa59431bff149d884fbd3cb4484e8b2 : SUCCESS in 1h 18m 07s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/99c4fb834a9b438896c25812db4c959c : SUCCESS in 1h 03m 38s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/16f4d2ca2b324e45b6cfc89a7f02390e : SUCCESS in 1h 11m 49s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fc4baac5fcba48ba98aa147104de1ac0 : SUCCESS in 1h 00m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba24ded7f46644bea30e97d03a650bdd : SUCCESS in 5m 14s","accounts_in_message":[],"_revision_number":1},{"id":"da73f2ec10d7f9cc54d32cf44ab9454abd341638","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-13 17:39:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9a23178497b92680b9b103e4ce745ef6076ada47","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-13 19:22:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8f36bd40e0464678a52dfb3bf9af7c5f\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a430ae5355544cb185789e1f14240bb7 : SUCCESS in 18m 03s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/fe869f50924e49f09aa67d239bdb37b2 : SUCCESS in 1h 20m 52s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/940dd381800148b4a66089a71cab050f : SUCCESS in 1h 22m 16s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/28711bd7da664f1d9ba585e037b15312 : SUCCESS in 1h 35m 48s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a374f9b0b0c44d9e89502f3cfe19752c : SUCCESS in 1h 30m 20s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6be387bb84854ece80c9d00e5ef622c5 : SUCCESS in 1h 21m 27s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e254db4dc1934448a1cbfd581bf22c61 : SUCCESS in 1h 28m 39s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/60c47dd2aa0b41519039ca95b74251b9 : SUCCESS in 1h 16m 16s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/04f56ecc43074486bc931791006c4bf7 : SUCCESS in 1h 32m 47s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/eee5fca2272e43c694061e85898fe1bc : SUCCESS in 1h 27m 18s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/68cfe25b195346139fdfe458251c70f8 : SUCCESS in 1h 13m 35s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/4cc4e29c6bc148c7b8e6dc346dd52346 : SUCCESS in 1h 20m 31s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e0306542ed6f4b17b44243e2c9699d15 : SUCCESS in 1h 16m 40s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3df525fe1bf641cc8a6d4a84aa7a8111 : SUCCESS in 1h 16m 23s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2e35e27cb7fd48f5a00c628c5c75f119 : SUCCESS in 1h 16m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/661f7e2f9d634e039d845588310d7436 : SUCCESS in 5m 49s","accounts_in_message":[],"_revision_number":1},{"id":"f6fa52e0cca04a9fdbbf83fe267ffda9f7139f20","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-13 19:22:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"8c83907d2ad38492d95d62149110e949114251e6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-13 19:23:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6f6af784c46a4c4286ab3b67abea6162\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7dd1b43022224ce4becf4b9e1b11dc42 : SUCCESS in 35s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c3b51d76a1ea2cfac39756200b2714636a4ae645","revisions":{"c3b51d76a1ea2cfac39756200b2714636a4ae645":{"kind":"REWORK","_number":1,"created":"2024-06-12 11:19:14.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/45/921845/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/45/921845/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/45/921845/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/45/921845/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/45/921845/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/45/921845/1"}}},"commit":{"parents":[{"commit":"34dfc99559724f3deba959dc9c27d277d6d1a65d","subject":"Update TOX_CONSTRAINTS_FILE for stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/34dfc99559724f3deba959dc9c27d277d6d1a65d"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-06-11 19:36:59.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-06-12 11:19:14.000000000","tz":0},"subject":"Define oslo_messaging_rabbit section if either RPC or Notifications are enabled","message":"Define oslo_messaging_rabbit section if either RPC or Notifications are enabled\n\nAt the moment Keystone config would not contain rabbitmq tunning for\nnotifications, while this still might be preffered.\n\nWe adjust template to contain oslo_messaging_rabbit if either RPC or\nNotifications are enabled.\n\nChange-Id: I2e801fade862608382aa0adb868eff3f44ed938b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/c3b51d76a1ea2cfac39756200b2714636a4ae645"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/c3b51d76a1ea2cfac39756200b2714636a4ae645"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
