)]}'
{"id":"openstack%2Fopenstack-ansible-os_keystone~922640","triplet_id":"openstack%2Fopenstack-ansible-os_keystone~stable%2F2023.1~I4cfef16841f95328d4ae7e4666f5a8fac053440b","project":"openstack/openstack-ansible-os_keystone","branch":"stable/2023.1","attention_set":{},"removed_from_attention_set":{"14552":{"account":{"_account_id":14552,"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","username":"bjoern"},"last_update":"2024-07-11 09:40:16.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4cfef16841f95328d4ae7e4666f5a8fac053440b","subject":"Remove X-Forwarded-Proto header in apache","status":"MERGED","created":"2024-06-24 15:52:13.000000000","updated":"2024-07-11 09:41:13.000000000","submitted":"2024-07-11 09:40:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"922640","meta_rev_id":"3564fb4156e0c1271acd1352ae6aa6acc90ad553","_number":922640,"virtual_id_number":922640,"owner":{"_account_id":14552,"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","username":"bjoern"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-11 09:40:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2024-07-11 07:50:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":2,"date":"2024-06-24 16:01:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2024-07-11 07:50:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-06-24 16:01:01.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2024-06-24 17:47:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-11 07:50:28.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"b418a33606a2e6cb73876f2766c8746112da2163","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14552,"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","username":"bjoern"},"date":"2024-06-24 15:52:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7001396e56575fdeabfb1ea3f9a54e382e38f7a2","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-06-24 16:01:01.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8f7235ecbeb9926b9290852db56cfd6252d2c77a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-24 17:47:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ce62560044364016836258565e183e30\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/65239db7913f40a9993108e86ca95804 : SUCCESS in 10m 52s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/40b909a83d184e1ead216cc7241d539a : SUCCESS in 1h 28m 46s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/3e7b8b5dd2184834acced13ac1233553 : SUCCESS in 1h 43m 14s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c4853758b28d44009c1a69e151f27805 : SUCCESS in 1h 46m 34s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/21d83cde73b14e7d8e14ed8cc0a7b2c8 : SUCCESS in 1h 45m 39s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/720ce0f206034da582d24734bb4efdc9 : SUCCESS in 1h 33m 14s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/901e2a7975d54a59b952fbb6da8fdae3 : SUCCESS in 1h 42m 21s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7c9da60e96ce465e96ef37ebcd1e4826 : SUCCESS in 1h 25m 26s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/bb91a9e751004aec88939cc46e8155e5 : FAILURE in 25m 09s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6aeb4c4e3a9647049c666e7021ed7a50 : FAILURE in 16m 40s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c9fed168ef35403fa2987f55f01ef329 : SUCCESS in 1h 04m 46s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/68cd868ad8bc4be9a4a2c14a5eebc881 : SUCCESS in 59m 42s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/595110af4bbf4997b9abb1cea4f07456 : SUCCESS in 1h 10m 51s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4b604105f70c46b99558592af68e493a : SUCCESS in 1h 16m 02s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a3999212a670436b90b3c42d35422ed6 : SUCCESS in 1h 36m 40s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/ac76286e73d44b8fb9b7b7ed22000faf : SUCCESS in 1h 13m 23s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/40fd18fe779f41cc89fb620e0d4fd576 : SUCCESS in 1h 16m 20s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b76786e3173d4e1882b5d1771f18c29b : SUCCESS in 1h 07m 51s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/706f560a6cb148c397c5213c064eddfa : SUCCESS in 1h 10m 42s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7c9aeee336554d3d899d070fbc87b50f : SUCCESS in 1h 18m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0fcd8a9f11f4545ab0a53c4b3b06442 : SUCCESS in 4m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/106794d152dc4ddcb2cfe90661712558 : SUCCESS in 3m 32s","accounts_in_message":[],"_revision_number":1},{"id":"d85caf0bf7c80e4844424cb3d2e920a9a19f0bb3","author":{"_account_id":14552,"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","username":"bjoern"},"date":"2024-06-25 19:17:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"69b0a3cb40727b9c1e4e5d6caf22c827d8ce7c98","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-25 21:09:11.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7fe01c9006cd400985c7b45fac8bdcc4\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e5c65c33bcd145188471e8eb2a941118 : SUCCESS in 10m 24s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/776de5cc7575455888ac2e140409d4f9 : SUCCESS in 1h 23m 48s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/48a4d8056dc74272a8e2c3fcdc07ac40 : SUCCESS in 1h 37m 22s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9f2a9d9742a4406eabbac6e1ada5e57f : SUCCESS in 1h 34m 50s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ca7aba01df504ebc87ca45063f95b5bf : SUCCESS in 1h 43m 07s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f6f1fb82a42b4d78aeaf9e65c21d90ba : SUCCESS in 1h 32m 36s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4199ad4d81b14c65a86ef5187519cbbf : SUCCESS in 1h 34m 38s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0e77ba5eab444384ba1c6f3245e7c7c8 : SUCCESS in 1h 26m 03s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/decc2d601a5f4496804e96380f8f309c : FAILURE in 19m 41s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8f07ce6798a04dfdae693f43bf28bebb : FAILURE in 17m 15s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ba1559e4d44d44e99f29e07f4c3a1c7d : SUCCESS in 51m 39s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/591ca786173f4ad4a02db67e774beb40 : SUCCESS in 1h 02m 07s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/5e2b213cc86f45f29f1bf2f355e762cf : SUCCESS in 1h 29m 41s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/bb5a682b781a4fd49e19e16c795d90ec : SUCCESS in 1h 26m 34s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7c8b5792a813496bbe034c9259c2b02d : SUCCESS in 1h 36m 53s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/70c0b711a57e400bbf94237e90438525 : SUCCESS in 1h 13m 21s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/410bcb2d4edc400ba1881cb8346456be : SUCCESS in 1h 11m 26s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9a0b3c881513416ba2c4448857c262c4 : SUCCESS in 1h 17m 01s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8e9bd89fd6b04ef2aa25145c228af04a : SUCCESS in 1h 14m 28s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1478b4a1583248df97c545720b5ae73f : SUCCESS in 1h 18m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/113c5c8d12a445f5a5f050269ee0e9e4 : SUCCESS in 3m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/70ff5412ebab406db9da9b699c73da58 : SUCCESS in 3m 45s","accounts_in_message":[],"_revision_number":1},{"id":"2ed9afcfc10fe07abad59ebaa32e36bacb72bc0b","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-07-10 19:38:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"aa8d8a6e08a2e5c45c371dd08589b83b8fa7ff0b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-11 01:36:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b0c6190c0f3e410cbfc52be4159522f1\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7d67f9664acb4a989377d07d26284dd4 : SUCCESS in 10m 34s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/8b43f9982aa243078ba6f2267944cc4d : SUCCESS in 1h 25m 36s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/fbd5b210c0c4421bbc6c1ecc5c90ec3a : SUCCESS in 1h 35m 32s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d1d5be0db8b14f4095ea520e284ca654 : SUCCESS in 1h 43m 00s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f4e4a2e978b843ec80f43cbf95322b40 : SUCCESS in 1h 29m 08s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/377ca3e4d3764bb69d83713de765815d : SUCCESS in 1h 22m 18s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ee0f583cff91450fb69723fe8d205c2d : SUCCESS in 1h 42m 20s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1cc07e6a092545c5af2a79824ce19752 : SUCCESS in 1h 24m 13s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/358a9b30f0b2456681364eda808bbd06 : SUCCESS in 2h 20m 12s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4ad23201954f40e4b5925a7a570352c7 : SUCCESS in 1h 49m 13s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5d857861869445daaf7154d4373c2667 : SUCCESS in 1h 09m 20s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c9224b90e00a4370aaebd9a513a098db : SUCCESS in 1h 03m 46s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/1ab99b04f17f40f6987328a729937837 : SUCCESS in 1h 30m 21s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/54fb1a6e012142968e88be374b5c0f30 : SUCCESS in 1h 32m 05s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4843aae17d45469c96985d3c805b3897 : SUCCESS in 1h 30m 35s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/81afc3ef08e94eb2a7f892fe5b278235 : SUCCESS in 1h 08m 32s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2aa4afab4ba74968ab4c2d0eca024d59 : SUCCESS in 1h 22m 26s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1ac37f0b99234b29b7c295ca1e075df5 : SUCCESS in 1h 16m 08s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/47463279651f4e9ca6e9b7aeec2fc247 : SUCCESS in 1h 15m 30s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3f46aa428794464d8287567282ed4fcd : SUCCESS in 1h 24m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dbf4b459a4994a81a9b3ff04a89b69a8 : SUCCESS in 5m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/36ab329b32a64a85a42058013506ad35 : SUCCESS in 3m 25s","accounts_in_message":[],"_revision_number":1},{"id":"6dbb3843e86ebdf6cc04f603a71b7ed0d62864d4","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2024-07-11 07:50:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"29452c07551f6f78e2337181972028853dd8e80d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-11 07:52:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7d524a9278c0e3e6991240e6f430a68027214235","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-11 09:40:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e2edfc1b6a554fbcb60e50b0b40cdeaf\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bc26394ed25e451bbe3a122c5c5333f7 : SUCCESS in 10m 23s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/cf0ea783707141c48d782aaf1ea5323c : SUCCESS in 1h 35m 36s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/17eeef7ce9e043678e6334353a99b70e : SUCCESS in 1h 33m 26s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1b351e42551d47659c906dba8059e73d : SUCCESS in 1h 19m 55s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0278eb9ad3a348a2a1cdd77e98078ef9 : SUCCESS in 1h 35m 12s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a6a64aabf85c4798955b9ff9105f75e5 : SUCCESS in 1h 32m 32s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a0c794a93f054328b7b89ea175dedbd2 : SUCCESS in 1h 28m 43s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b1a9980c10f6487f954b02bb24df548b : SUCCESS in 1h 36m 58s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/9ae4b39dbff2424aaf740bd55abe59d3 : SUCCESS in 1h 08m 43s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ccb80e9fb2744aea8931317ee413b81c : SUCCESS in 1h 17m 42s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e7f19e334c234816bc012c2ef307de0d : SUCCESS in 1h 10m 11s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2a75422f71d2456fa532ee705b6a4786 : SUCCESS in 1h 17m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c4b6d496d524ab29f6d230fa4e9ed5f : SUCCESS in 5m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a50f12671ac446988d8c447cad1726a3 : SUCCESS in 3m 46s","accounts_in_message":[],"_revision_number":1},{"id":"2cb722b6aabe0425f69b0629d800682176f001b4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-11 09:40:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"3564fb4156e0c1271acd1352ae6aa6acc90ad553","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-11 09:41:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec24545d7a8a43feb5defaa544f5d0c2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f446e7dd67f4766a2ed6f15915bf4ed : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a45472ac4fbd8eb4a3230296b820f494650ef722","revisions":{"a45472ac4fbd8eb4a3230296b820f494650ef722":{"kind":"REWORK","_number":1,"created":"2024-06-24 15:52:13.000000000","uploader":{"_account_id":14552,"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","username":"bjoern"},"ref":"refs/changes/40/922640/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/40/922640/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/40/922640/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/40/922640/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/40/922640/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/40/922640/1"}}},"commit":{"parents":[{"commit":"d93e2bb46a802653dea66f955ad37613307575d9","subject":"Fix permissions for SSH private key for Ubuntu distro installations","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/d93e2bb46a802653dea66f955ad37613307575d9"}]}],"author":{"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","date":"2024-06-04 15:49:50.000000000","tz":-300},"committer":{"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","date":"2024-06-24 15:51:59.000000000","tz":-300},"subject":"Remove X-Forwarded-Proto header in apache","message":"Remove X-Forwarded-Proto header in apache\n\nTher X-Forward-Proto header should be added on the external loadbalancer only,\notherwise admin, internal endpoint requests to keystone may suddenly flip to\nHTTPS while HTTP is configured in default.\nThis also affects request to local keystone API calls for monitoring etc.\n\nCloses-Bug: #2068039\nChange-Id: I4cfef16841f95328d4ae7e4666f5a8fac053440b\n(cherry picked from commit e8d0f0db5f623f3a2ebbacca6b237f16f7d27202)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/a45472ac4fbd8eb4a3230296b820f494650ef722"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/a45472ac4fbd8eb4a3230296b820f494650ef722"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
