)]}'
{"id":"openstack%2Fopenstack-ansible-os_keystone~924719","triplet_id":"openstack%2Fopenstack-ansible-os_keystone~master~I192b8213f1fd90e28d15691e233f2fd671429a94","project":"openstack/openstack-ansible-os_keystone","branch":"master","topic":"osa/noble","attention_set":{},"removed_from_attention_set":{"31542":{"account":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"last_update":"2024-08-05 20:48:50.000000000","reason":"Change was submitted"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2024-08-05 20:48:50.000000000","reason":"Change was submitted"},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2024-08-05 19:12:25.000000000","reason":"\u003cGERRIT_ACCOUNT_25023\u003e replied on the change","reason_account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}}},"hashtags":[],"change_id":"I192b8213f1fd90e28d15691e233f2fd671429a94","subject":"Use platform dependant sshd service name in restart handler","status":"MERGED","created":"2024-07-23 11:01:56.000000000","updated":"2024-08-05 20:49:43.000000000","submitted":"2024-08-05 20:48:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"924719-osa/noble","meta_rev_id":"1d51aa6dbe21425826c019deb2690f122ad25852","_number":924719,"virtual_id_number":924719,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-05 20:48:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-08-05 19:12:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-08-05 19:12:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-23 13:42:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-24 09:24:36.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2024-07-24 12:29:36.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"cd1395f0a8dd43ab12d452ee15b1b9cf3f4f36a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2024-07-23 11:01:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"50151ca9bf89bb5c356c54e08ce3f0ea06ecfdca","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2024-07-23 11:02:37.000000000","message":"Topic set to osa/noble","accounts_in_message":[],"_revision_number":1},{"id":"3e7fb60dd779bc013d93fc50b9be9a69d4bdc426","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-23 13:42:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb1b364e13f947a48a701478e93f6c1b\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/57970b1883834e41a1f051e804ac00dd : SUCCESS in 17m 23s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/895ea90a2c184c1d8158aa4f462c0606 : SUCCESS in 1h 39m 11s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/f89e7f5002e64110aa7a390fddf5e6ec : SUCCESS in 1h 37m 42s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a710ed3fabf34f3bafa311370b466159 : SUCCESS in 1h 35m 45s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9c7df7988d1447f7b43e3d7b351aa7be : SUCCESS in 1h 28m 23s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5d4bde3a119c47ccbf25037e4a444d82 : SUCCESS in 1h 31m 15s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cba6bb9114164d91938a5180196fed8c : SUCCESS in 1h 29m 45s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/38684cb38c7f4a8cb55fe155a9b9bf62 : SUCCESS in 1h 26m 03s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/bf36532dd93f43c786be796880c29cee : SUCCESS in 2h 01m 02s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/82e39bfa76614022bf717a2d11516330 : SUCCESS in 2h 03m 53s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/bda9f7bf52ee49f8a858167fe2cd3439 : SUCCESS in 2h 06m 25s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/520cfb50825c48449472134758ab3103 : SUCCESS in 1h 43m 29s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c421dc8a73a24c059a622af9885168f4 : SUCCESS in 1h 02m 49s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/14bedee163c34bcc8e7ff3b6d7f5a21f : SUCCESS in 1h 00m 58s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/f1e6ebf2f44e44ec8294e730884b7819 : SUCCESS in 1h 26m 28s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2076fa1ce9274a8ba4e4cd274b47e608 : SUCCESS in 1h 18m 28s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/8ef9088984934de79e567d3a0ac6a3c2 : SUCCESS in 1h 21m 25s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/274618b7df614ad6bc995860a02fcce7 : SUCCESS in 1h 13m 52s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c96937f94280474d9411e85c14459f08 : SUCCESS in 1h 23m 50s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9333fe48319544e6adf1ee3eb3fa3cce : SUCCESS in 1h 15m 02s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2d9fae1116f841ecae9c80127a774858 : SUCCESS in 1h 14m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d25801a2a19945b5aac7d66ad44baea1 : SUCCESS in 4m 20s","accounts_in_message":[],"_revision_number":1},{"id":"f437b852b8242da89c426b348aba6c633997e955","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-07-24 09:24:36.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4f6aaf616e0908a02b924201365fa8e9f0b44ab6","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2024-07-24 12:29:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"34ba13bb720705b4bf7721bd70a5dc606836f991","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-24 12:31:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"70665509f791c9f71729a2d2c4ed899748b7301a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-24 14:06:57.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c0d355221982445f8fda9d5194ff4c95\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f7040e6dd83a4ec08ef163a29791e6f6 : SUCCESS in 11m 20s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/22bfa3cc65664216be3f08fa9ef731a1 : SUCCESS in 1h 18m 46s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/bb1f31e8be9d429f8d56dbb34c6b58d4 : SUCCESS in 1h 19m 40s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/80a4ea4f6691478989adc5db6ed9c6dd : SUCCESS in 1h 18m 15s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3a3e8a35ad1344a39764e9b90888a899 : SUCCESS in 1h 25m 50s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3f7bd72aea3f4f38b7eec4c8987a66ad : SUCCESS in 1h 28m 43s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/41921ef8a3194387af2419fc3b684589 : FAILURE in 1h 05m 11s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c302482f14284eac885620da4d2cde0f : SUCCESS in 1h 26m 39s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3936cb17394040beba40b87ed9dc4a9c : SUCCESS in 1h 25m 16s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/ab55eb854419467a8958f73b78df0e5e : SUCCESS in 1h 16m 01s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d2e2a754d33e46d5a24ad32c30c582ea : SUCCESS in 1h 15m 31s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/53140bee56e846c987fd77e9c7c95b82 : SUCCESS in 1h 18m 17s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0d1732a95f86418fa621cb00f95b14c9 : SUCCESS in 1h 08m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3f81f5821f84b70980eb8a69fee333d : SUCCESS in 4m 36s","accounts_in_message":[],"_revision_number":1},{"id":"b362e5343baf1181dae110571919786ce0be80b4","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-07-24 14:13:37.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7091d3f146d7177bc4b751a42b6c99c53b60a1fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-24 16:44:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c60c2d84380d40cd872b8c30ae6ec10f\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3a0a29b04e824b6185f58d319b6ef581 : SUCCESS in 13m 02s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/633691e492514c77aa379cc1ca0d4027 : SUCCESS in 1h 14m 08s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/bc22f8e1c5fa4d069715182993a0b3f8 : SUCCESS in 1h 11m 14s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7ebe8459994340baa9e8a36af9c3192f : SUCCESS in 1h 22m 19s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/60503128388744a2b76b23be72e9c99d : SUCCESS in 1h 08m 38s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d9ca6d2f7b5646bea3c66d20d0948e8f : FAILURE in 1h 52m 51s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dd3fc114bbf7417da17510974d7b8a10 : SUCCESS in 1h 19m 25s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9ad7392d0ef7461f835f91d71cd8e8fd : SUCCESS in 1h 05m 15s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6a2c7cd626344d4aa2c13bd52a7767e0 : SUCCESS in 2h 05m 57s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9eefdd4406a34ff78fe59ebe05edd826 : SUCCESS in 1h 29m 58s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d9d036f0623f4c579d6ea05628e605b5 : SUCCESS in 2h 18m 32s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/961f4cf5063e430dbafa7c6b2c9f1bcc : SUCCESS in 2h 04m 12s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5ef54b3ad9194fd8b88ab4dae7f0c943 : SUCCESS in 1h 10m 23s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a169018b3a8444728bca8b9fb8029b0c : SUCCESS in 1h 06m 34s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/b0b7c145c0214bfdb98e8ce2c260af0c : SUCCESS in 1h 13m 46s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/38e5eccbc319450d915bbd450b18d625 : SUCCESS in 1h 18m 22s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/012072c1bd3247a59126d693cd8bc586 : SUCCESS in 1h 11m 04s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/6d9276ef5f35481a9d9f82d9f4952335 : SUCCESS in 1h 26m 36s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/01ce06264ed7441aa1c7567c8167f916 : SUCCESS in 1h 06m 21s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9c5d8842efa34c239934ca4e048637bf : SUCCESS in 1h 05m 50s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9e7b51c68c214e16a3acc5256350d6bd : SUCCESS in 1h 06m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea6bc87a3e034773b1b8326bb73a4771 : SUCCESS in 4m 34s","accounts_in_message":[],"_revision_number":1},{"id":"ca66c853ae953721ed3794fedea85b31f8ac5f64","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-07-29 08:31:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"688e73b4a8a9a5cf433af0f0906cac43ae34e4f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-29 10:49:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/637d450b8891424498e0996176fc0e34\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/415dedf5759e4c908af7eadede4b4987 : SUCCESS in 10m 38s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/2fbed380696949a791fabaae5b80d434 : SUCCESS in 1h 20m 24s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/97d777df6f6b4dad9ce32567b5af0204 : SUCCESS in 1h 26m 40s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3e24de9b1b834330b48e53c1e224841c : SUCCESS in 1h 20m 00s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dda653cffbd64814866156ddf96bf9bc : SUCCESS in 1h 20m 11s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7914aa958828417f9eca7c78dd12001a : SUCCESS in 1h 27m 51s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cb45d3514ab248e9bb1f18f3671269d2 : SUCCESS in 1h 13m 53s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8240b79c1572436e8195c5bc302197ed : SUCCESS in 1h 23m 53s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/06d20041ce4841edbf6609d5756e24dc : SUCCESS in 2h 01m 24s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/249ffdf7c5d646939cd7e3373daf965e : SUCCESS in 1h 36m 04s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/dfb4f31e49354200a0b3f07e0994c21a : SUCCESS in 2h 08m 06s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8b3c472d8dc04d43af9c2b59f68bfcec : SUCCESS in 2h 00m 14s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2e0fc6b0702d4738b6825929e244085a : SUCCESS in 57m 46s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ec85390c510647b297fcfeb35cbe4d31 : SUCCESS in 49m 20s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/b7ebed02c35047b290102e8a231ef7e1 : SUCCESS in 1h 15m 29s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/707dd0f1754142d39cd3e91fa29fa59a : SUCCESS in 1h 20m 16s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a4aae50020584584ab5540914042e1ba : SUCCESS in 1h 12m 13s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/98ecc7456fcc43f9b510652d5b861df3 : SUCCESS in 1h 15m 49s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/65a9ae37ba14433796a2f815b81f6d44 : SUCCESS in 1h 09m 57s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/de33fe304c334846870f64ed18d5611f : SUCCESS in 1h 09m 30s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ecada6e0f4df45b7af6a6c4415cab2a5 : SUCCESS in 1h 16m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fecd6d34ccd24b9992de970f837db3db : SUCCESS in 3m 48s","accounts_in_message":[],"_revision_number":1},{"id":"853c012e08e4ea8f5eadff12b77e54624e07cc34","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-29 10:51:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"802a94d2481408af231dc35b2f71c1b3218530fe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-29 12:25:04.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/36c1e17a2ba44b3da76e9ce3c21edcb0\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3fdfd14714f542b9afd521a7cafca33e : SUCCESS in 10m 35s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/3de6843b7bc8460fa700bbcb2ac6f2d6 : SUCCESS in 1h 15m 01s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6e7ac85323414488b792986437575bca : SUCCESS in 1h 20m 11s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/75bee754456a4bb4a190ee651c2903f2 : SUCCESS in 1h 20m 41s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f8a58b6c40fc46768d724923999db965 : SUCCESS in 1h 26m 32s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/8cf8e0ce5080424692cf60f2726d7882 : FAILURE in 23m 44s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/27410734de014baa892e8368b19267f6 : SUCCESS in 1h 18m 24s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/32a0b528d58a49fd80c77a8ae22f1ae1 : SUCCESS in 1h 26m 11s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/840ae16136554c488bfadbff1c630803 : SUCCESS in 1h 32m 26s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/fd260a518f6043caa5f9e3926bd3632b : SUCCESS in 1h 15m 12s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/16693e924d7447f39e89e91adac5dced : SUCCESS in 1h 14m 26s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8d1fef8b1cdc49d4a4aa324d404890ce : SUCCESS in 1h 18m 52s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/11839a006e484e17bfb05a3413de1cfc : SUCCESS in 1h 08m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/170959c5ab5441ef9b89d65ea98da48e : SUCCESS in 7m 38s","accounts_in_message":[],"_revision_number":1},{"id":"736da579960578ea99d04073afd255e258084756","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-07-31 11:05:17.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"aba64484c295ce50182da9580f847c89796cdc5a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-31 13:35:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/193e9af6caf44154b2ea0a2882a81439\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b6bd63b16aa74dd2a692d642c1683810 : SUCCESS in 12m 19s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/693cd5d1f52b4907a4e799f2d7da9022 : SUCCESS in 1h 20m 55s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/8f4b3125ad1e4790b2e8eeb3afc0c129 : SUCCESS in 1h 18m 02s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f13624d7bfef437ab6ee80ab7dcc139f : SUCCESS in 1h 20m 55s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f41da4b56cfd470689f11d105a16dd67 : SUCCESS in 1h 12m 20s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1cb1156e5e874484af5543deb8d69629 : FAILURE in 26m 45s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/015b9446d4034896b8577782daec0189 : SUCCESS in 1h 19m 21s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/be68ca5a3d104e68b8aed014c024faa6 : SUCCESS in 1h 14m 48s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4c444ea416bd4384af59c36240a1299a : SUCCESS in 2h 21m 51s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c51bef41ec004549afdf666dcad81c5e : SUCCESS in 1h 51m 40s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0297ec08017343f181841307f51a6cce : SUCCESS in 1h 59m 25s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/48a676f4d11649ea918536cb634ee870 : SUCCESS in 1h 47m 21s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/83166d8a595549f093170814ad683c5e : SUCCESS in 1h 00m 54s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e5827ee6d9144e62819961c2a202b10c : SUCCESS in 1h 00m 52s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/dcf8e1c02e184925960343d984393ac7 : SUCCESS in 1h 19m 31s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f75b376c98a34f6086920423ea295e89 : SUCCESS in 1h 21m 31s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/92c6a20b4e954e4a935e3ae699ac72c3 : SUCCESS in 1h 20m 37s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/a999c97fcba649b8bd3a8b2ba966ad78 : SUCCESS in 1h 11m 07s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8396b8eb1a7040aaa607369dac88ddae : SUCCESS in 1h 09m 50s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/23e3612916034fb1860f00b6b5139603 : SUCCESS in 1h 08m 55s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ee1ed00d83684b5da6f2e0cfea482a40 : SUCCESS in 1h 09m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65525be86f3f46399cd27d803c6ca43f : SUCCESS in 4m 10s","accounts_in_message":[],"_revision_number":1},{"id":"2c5e5b128f7ce0c62368b983a6618686f6482dfd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2024-08-05 11:20:35.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":2},{"id":"fb2e5756bf3bfc8807c1d6d24fc30f7980aec4f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-05 13:44:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a759c0d7fcd84fde9adbc1e92f4dacd7\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3d4dd61d2b204ffc81963de3d0f446c3 : SUCCESS in 11m 11s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/7d206f14c96f46e3ba876024cbdd3329 : SUCCESS in 1h 23m 09s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/319f3156d86a436baa189f4a6a7295bf : SUCCESS in 1h 15m 05s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5895562bc7d348f0b9b865b083b0ff7e : SUCCESS in 1h 29m 21s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e12d05e2bd1a48ce982393bec87d2768 : SUCCESS in 1h 08m 31s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/39b274bffa364045be75ea7f6340c142 : SUCCESS in 1h 32m 17s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/41bcf856cba24c3aa3ccf73e2bae1800 : SUCCESS in 1h 12m 23s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5f121bfcd3544c30a3e8477124571ab2 : SUCCESS in 1h 11m 23s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ee9f8e30505e42a8aa42a86492e5af2c : SUCCESS in 1h 58m 55s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2c6a0c29b1114662b7721a34c61c3141 : SUCCESS in 1h 47m 00s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6b8689e43c56405184c39b831e5dc345 : SUCCESS in 2h 15m 38s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/834445785e6547e89f827c49023438d1 : SUCCESS in 1h 37m 19s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/07ae12feb08744e59db78c544da712df : SUCCESS in 1h 01m 02s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f89acc7062904e64af6d47808b8d0257 : SUCCESS in 57m 47s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/fd02ef97b9de4a558acc07dab32fe54d : SUCCESS in 1h 23m 14s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6a35c57ae3454b50a49a3f485aca91e7 : SUCCESS in 1h 20m 38s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5bbad0ea9bbf420d9f2fe5bdb8a8b710 : SUCCESS in 1h 28m 15s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/18685d92dc5b4754be40d4360d286dbb : SUCCESS in 1h 08m 06s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/83dbaacc497647258d014ff7d9137b44 : SUCCESS in 1h 08m 42s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/64d8b80cc3df4a67bbeb2ea868fd12c3 : SUCCESS in 1h 04m 15s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a6da7d49d8c34ebc93f636fee54d374b : SUCCESS in 1h 16m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a59165c62dc2414f86022bef923ab834 : SUCCESS in 5m 46s","accounts_in_message":[],"_revision_number":2},{"id":"c61cd582d52375c3b5d0ef6e4c84f042112224ae","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2024-08-05 19:12:25.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"88a591b534594f57ed017df93fb102789142188f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-05 19:14:32.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"072a9dedd742dd6dff45923ec9a3651ff0b79426","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-05 20:48:50.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd8de01bd4374877b11202c0c37d7c2e\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f2ecfaf6b6fe4a2c8b5a8c5daf436757 : SUCCESS in 13m 00s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/87e216777f784ca1add7319ba555f1dc : SUCCESS in 1h 21m 42s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2c5ab39f3230442f86480658b19faa54 : SUCCESS in 1h 14m 42s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d8d6e9a6c23f4d9ca73497dfa6e439b3 : SUCCESS in 1h 08m 04s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f5087095d2364279abd7e32d14d65aab : SUCCESS in 1h 14m 01s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a1dde47982484335b5629808acd93d2d : SUCCESS in 1h 28m 47s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b909879a4a524e0b83df025fbbe8559c : SUCCESS in 1h 19m 53s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3676aeb60fc44287905143c218d3fe7a : SUCCESS in 1h 27m 16s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4cd45aa8ee814d30ba3e90b6a8057cb6 : SUCCESS in 1h 26m 01s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/44d4705936d84aa5a276770d48d3e40e : SUCCESS in 1h 12m 26s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f9a221cf2eb84158a1f86bcd49dca169 : SUCCESS in 1h 13m 36s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/141a55ee4cf24c39806c1450fa51cf8f : SUCCESS in 1h 11m 55s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/177b15706f3649569d8f90643a62fa2d : SUCCESS in 1h 15m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e00d5878322439a9cead167edb399c9 : SUCCESS in 4m 59s","accounts_in_message":[],"_revision_number":2},{"id":"8cc77e0d1fcb8ace4ceb92a913978e640893b136","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-05 20:48:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"1d51aa6dbe21425826c019deb2690f122ad25852","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-05 20:49:43.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8c8cf750a71421eb556e732e4e6fc76\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7122036eb43846f6b9f9b09bb5be288d : SUCCESS in 36s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d1c1fad704350cd821821ed14f468cf3c9954d00","revisions":{"2740050a514367271a1d7b4418bb38a28383faf9":{"kind":"REWORK","_number":1,"created":"2024-07-23 11:01:56.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/19/924719/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/19/924719/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/19/924719/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/19/924719/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/19/924719/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/19/924719/1"}}},"commit":{"parents":[{"commit":"d163d9173050a53eb52f3889e8a1cdcb3ce1bfdd","subject":"Combine Ubuntu/Debian vars together","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/d163d9173050a53eb52f3889e8a1cdcb3ce1bfdd"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2024-07-23 10:59:04.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2024-07-23 11:01:39.000000000","tz":60},"subject":"Use platform dependant sshd service name in restart handler","message":"Use platform dependant sshd service name in restart handler\n\nChange-Id: I192b8213f1fd90e28d15691e233f2fd671429a94\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/2740050a514367271a1d7b4418bb38a28383faf9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/2740050a514367271a1d7b4418bb38a28383faf9"}]},"branch":"refs/heads/master"},"d1c1fad704350cd821821ed14f468cf3c9954d00":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-08-05 11:20:35.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/19/924719/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/19/924719/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/19/924719/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/19/924719/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/19/924719/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/19/924719/2"}}},"commit":{"parents":[{"commit":"d163d9173050a53eb52f3889e8a1cdcb3ce1bfdd","subject":"Combine Ubuntu/Debian vars together","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/d163d9173050a53eb52f3889e8a1cdcb3ce1bfdd"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2024-07-23 10:59:04.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2024-08-05 11:20:35.000000000","tz":0},"subject":"Use platform dependant sshd service name in restart handler","message":"Use platform dependant sshd service name in restart handler\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-repo_server/+/925571\nChange-Id: I192b8213f1fd90e28d15691e233f2fd671429a94\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/d1c1fad704350cd821821ed14f468cf3c9954d00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/d1c1fad704350cd821821ed14f468cf3c9954d00"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
