)]}'
{"id":"openstack%2Fopenstack-ansible-os_keystone~927155","triplet_id":"openstack%2Fopenstack-ansible-os_keystone~stable%2F2023.2~Ib10022bc2b29e556d5254827122c138e90381c0b","project":"openstack/openstack-ansible-os_keystone","branch":"stable/2023.2","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2024-09-09 22:47:33.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib10022bc2b29e556d5254827122c138e90381c0b","subject":"Do not install uWSGI as part of Keystone","status":"MERGED","created":"2024-08-26 15:01:31.000000000","updated":"2024-09-09 22:48:24.000000000","submitted":"2024-09-09 22:47:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"927155","cherry_pick_of_change":924928,"cherry_pick_of_patch_set":3,"meta_rev_id":"4a763624897e045068736aa2116510013b1d234b","_number":927155,"virtual_id_number":927155,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-09 22:47:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2024-08-27 07:05:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-08-30 08:45:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-08-30 08:45:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}],"CC":[{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-26 19:06:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-30 08:45:05.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2024-08-30 12:37:02.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"CC"}],"messages":[{"id":"0510ec53c1237313bb4ee2ded2131c0ef881e443","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-08-26 15:01:31.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"6c2f3b5eac2617acdf9afde3af7fb56664249660","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-26 19:06:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a33bce445053400abef6354aecda62be\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/38f2c0ce7f93488c81c02cc710ff879d : SUCCESS in 10m 32s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/c6403b836387416f84874ba9ada364f1 : SUCCESS in 1h 24m 30s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/5421e3a5352348f882c64aa4ba4d4531 : SUCCESS in 1h 23m 06s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/b043ad98a9194a708848c0fe1c175a04 : SUCCESS in 1h 22m 55s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b8c8009e70634745af5b80047e6b9fb8 : SUCCESS in 1h 23m 18s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2270a816947f4cc5bcb6c420e515fe47 : SUCCESS in 1h 27m 40s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/359130772de445d6b94df15bd768447b : SUCCESS in 1h 40m 01s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c74a1d68ed7c472cbc68224cc0f16e85 : SUCCESS in 1h 25m 42s\n- openstack-ansible-deploy-aio_metal_quorum-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/50d4fb0b60b943e4ad9f04492a8158e9 : SUCCESS in 1h 09m 59s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a1f69e0096254465869f0abdcce44722 : SUCCESS in 2h 27m 18s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c2c4cff0347c4668aee7c12d9e4e224e : SUCCESS in 2h 18m 40s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/db9f194c9b5e43f5841b8b72a70dbe0d : SUCCESS in 2h 11m 08s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2c40312e08a04ba98dc8d946bc498b3f : SUCCESS in 1h 56m 44s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/11b3daf655ab4f079d2d6db4e5197e05 : SUCCESS in 1h 12m 58s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/40d4eefb4d7b425985ca371ad78ac707 : SUCCESS in 1h 11m 13s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/b5b5a2b9ecb54f6dbfec3f4987c0fbcd : SUCCESS in 1h 41m 08s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2907e9c15bb74f32a07023a3dd7244f7 : SUCCESS in 1h 22m 56s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9fbaab7ebd6741479e13f7a4f7b95cf9 : SUCCESS in 1h 39m 43s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/babba58d22424069bcf81bd2bd51197e : SUCCESS in 1h 09m 34s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/146d25417f364d77a2707fb24f2d5eb5 : SUCCESS in 1h 10m 31s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5e81cf9ad0684bbea7884cb1b1f82db5 : SUCCESS in 1h 09m 38s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e82516a3ea4146aeb66ed54ac7807a34 : SUCCESS in 1h 08m 07s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7d4fbfaf8122478e83a46cde169b6dd2 : SUCCESS in 1h 10m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d5a11bd5879481ea36a0cc793f82b38 : SUCCESS in 4m 42s","accounts_in_message":[],"_revision_number":1},{"id":"ef4ecb7512b6a3eeccd9bb8fefabdca5ecae58ab","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-08-27 07:05:22.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f9ea351c7a9a57efc78bbee06e803abda2220ffa","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2024-08-30 08:45:05.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"01fdbb425422ba9cc581655bfcd4ae01612caf69","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-30 08:46:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"810b13e568ce884f046d5dd213c971880da7f14d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-30 10:30:12.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4034f9e0b742428c944a6cefb3fd2a47\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7cb18232585b4423903f9f29b0011b0b : SUCCESS in 5m 46s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/22ab9b9036cc42cc9d3d21e9059e4bda : SUCCESS in 1h 08m 17s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/29b23e28c332403d93a2b0ed605ca5f8 : SUCCESS in 1h 07m 12s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7c6ef814eaf7437c985ef91df674eabf : SUCCESS in 1h 11m 16s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0ad791843bfd4253aa74e68f1f179fa6 : SUCCESS in 1h 07m 46s\n- openstack-ansible-deploy-aio_metal_quorum-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a22ed52ee1834f0aa5d1bc0be256c9bb : SUCCESS in 1h 11m 09s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/424a3b88b69c488c9e6a49de9a8105f3 : SUCCESS in 1h 15m 39s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c56327723dae4b44b4e0d0d9d2fa674f : SUCCESS in 1h 10m 21s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f118662dfea14616ac601e026a660ce9 : FAILURE in 1h 35m 59s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f7dc438162754390a8a1fca309ccf900 : SUCCESS in 1h 25m 33s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/1967b9e838964b70aefa47c0ac8480a7 : SUCCESS in 1h 10m 24s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/73539e5a595548278b9e2bf7b0156d93 : SUCCESS in 1h 14m 41s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c80cc4768de2457d91cba2b99df9a0a7 : SUCCESS in 1h 09m 23s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8a5c615b4bac40928f6b5e4afdc0e4ce : SUCCESS in 1h 16m 03s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/77411b0477ac486d90b32fcc7abc87f6 : SUCCESS in 1h 11m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/277d59de08964ac2a8d2864e1d599227 : SUCCESS in 4m 03s","accounts_in_message":[],"_revision_number":1},{"id":"4d3595d5b67eaa810edec3d265a2b8eae26f63e2","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2024-08-30 12:37:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e399cfecbeaca04943c1733d7c174a670dd515c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-30 15:16:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b76586c9ab444e1884beb1f6f92e2ac\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5c9959afb2c84547bc66f064e7f57a8b : SUCCESS in 7m 48s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/9a884e1bb5e84c55a898e4a6fb32cb6f : SUCCESS in 1h 22m 10s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/eeb9d3c14d5249cdaa897641d83fe9bb : SUCCESS in 1h 22m 41s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/2204a971a90c4c5aa894c608985652d9 : SUCCESS in 1h 09m 04s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6095481166ea4b86bc28ac1b83847b20 : SUCCESS in 1h 15m 43s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/88a0e52bd7154e0d8f8d2f841cd533de : SUCCESS in 1h 04m 48s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9f7903a947a54afcae82ee11d275aaf9 : SUCCESS in 1h 16m 38s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4764b0db6f954bf8b532630af47fdbcc : SUCCESS in 1h 09m 48s\n- openstack-ansible-deploy-aio_metal_quorum-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f58b02c108a54367ad2910f39134cdde : SUCCESS in 1h 03m 42s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/69c4884851074c559a9f755880e9397d : SUCCESS in 2h 31m 47s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/57c94b75a0454d1cbcabba6deb0cbc47 : SUCCESS in 2h 03m 22s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7b372d63f1db4d62a4585656b0612a93 : SUCCESS in 2h 10m 05s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/137e73aa43ef4bec8c712f1d80d7f488 : SUCCESS in 2h 05m 40s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/091545734247404a99d44391400792b9 : SUCCESS in 1h 00m 39s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/26b490f68d6540e995b645883b8f5d69 : SUCCESS in 58m 59s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/cb9162de7dee406a87f4d242db8be790 : FAILURE in 55m 30s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7f915f510e9c45ceba7de95bf4cc7474 : SUCCESS in 1h 26m 13s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5652e93b0c0646f3877cf2323d50dbfe : SUCCESS in 1h 25m 30s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/fdf7a9f02ce944638e5e6e9d8a2f502a : SUCCESS in 1h 06m 20s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/a74e4333779c4bc1857722d10204411d : SUCCESS in 1h 08m 23s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/955b018c9e18467b86caea5d3ae35a17 : SUCCESS in 1h 27m 23s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7e1d795a1cdd425bad75b8100633cd01 : SUCCESS in 1h 17m 10s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1efdc617c2f14a0f8f9c000c8e88bc37 : SUCCESS in 1h 20m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8f5d9c6d53e405fac921de973e7ea35 : SUCCESS in 3m 59s","accounts_in_message":[],"_revision_number":1},{"id":"d9e90f37cf6a062409aa631afa2bcb4e51557cd9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-30 15:18:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"718aedae9d1449e25c3ebdf37c88e52bfe5e2a1f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-30 16:57:44.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e1b85f498d78497f95d23b8c1d166e23\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a8136087607c4d4e8e1da62fba293447 : SUCCESS in 6m 58s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/b1471e4fb3f84b638339a80e4918c2ab : SUCCESS in 58m 20s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/95f33479aa954bee8a770f95fcbba9d8 : SUCCESS in 1h 29m 43s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7b1b79bf70164855be9dcac30951a478 : SUCCESS in 1h 29m 04s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6a32363a6ac04e5f8e86630ce0c7efa8 : SUCCESS in 1h 10m 24s\n- openstack-ansible-deploy-aio_metal_quorum-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7ac7b93b46c740a1bf26b5c9a1a843a0 : SUCCESS in 1h 17m 16s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c8b5267357694df88ba9143bc23742a3 : SUCCESS in 1h 13m 15s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f8f7c8f6a6ba49dabc06768a178a16ae : SUCCESS in 1h 05m 49s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/bd9fbf937fe048f791865799239187a3 : SUCCESS in 1h 25m 42s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4e60d7f1b5c847669114d3f334838962 : SUCCESS in 1h 21m 56s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/90494270ee024fcca09b0a25ca0b672c : SUCCESS in 1h 15m 57s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/5855754905094a15aa5356c559a8f655 : SUCCESS in 1h 20m 00s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a45c3416163d41408b05ea7ab1a0b13e : SUCCESS in 1h 12m 59s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3867c3b0c03c44f9b70a4b59b5b9a689 : SUCCESS in 1h 16m 08s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/39f80947a5ab47dd8fbb03981ba84c10 : FAILURE in 1h 13m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1e353f7b60c4bed87c680fd890131bd : SUCCESS in 8m 44s","accounts_in_message":[],"_revision_number":1},{"id":"68d225248d39af08f275cc1068c8766b41feca18","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2024-09-02 07:16:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ed02646562f792e0f782df2923406e0176b537cf","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-09-02 09:13:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dd2a5e350cd80f3017e4159ecd051e30fd1d0a72","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-02 09:47:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9b90804fac24e6e94aa0932f7c582a2\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a16ee496e72142b4a2c4b7852a1f662a : SUCCESS in 6m 30s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/5558f966efd0463480e1dd6da2a746cb : SUCCESS in 1h 19m 41s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/b4ce701a498c4239bba4cfbc011cc5d6 : SUCCESS in 1h 08m 35s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/158e70e97f314b7a8095784fb743c3fc : SUCCESS in 1h 19m 50s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0a9be27d49064f39b5fbe2a8323e28da : SUCCESS in 1h 19m 40s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9586f94c3eb14bfeb67441ecf0bbc01a : SUCCESS in 1h 13m 33s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/76799f758e9e4d42bc4560ca86e7af2f : SUCCESS in 1h 30m 04s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c8bbb571a9d448aaa79c4e1779851a7f : SUCCESS in 1h 21m 04s\n- openstack-ansible-deploy-aio_metal_quorum-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/54553508d66749ac8dec16b77198b203 : SUCCESS in 1h 15m 03s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/8648150b29ad48f087182138fda36793 : SUCCESS in 2h 24m 08s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/89f0961da88645eb9fdf1716c86ae63d : SUCCESS in 2h 01m 10s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/136cd910d6c744d1a15d9a3615c88449 : SUCCESS in 2h 16m 09s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/14dc0ad16eeb4c198ee1233bb9ffbf93 : SUCCESS in 2h 11m 12s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/10233bf2f51a4d309f0fc42969d174e1 : SUCCESS in 1h 01m 42s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f4befca56f4b46cfb3d637bcaac780f3 : SUCCESS in 1h 03m 36s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/491ad570bb474e1eb7a00b67c835e841 : FAILURE in 56m 21s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f5d403f025c442388078fb79d6161c24 : SUCCESS in 1h 18m 22s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/26aab4eb236245c4b17e7ee77111389e : SUCCESS in 1h 31m 28s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/7ca19660e3184efc90b1fc2651a1439c : SUCCESS in 1h 17m 06s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/605992e6132b4c9789a6d1636a256ddc : SUCCESS in 1h 14m 51s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cdbf271697164c74bfb7df6b253e5f63 : SUCCESS in 1h 10m 02s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7be9d9cac0b044088430d16afe802649 : SUCCESS in 1h 24m 04s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/365569804a214655984970f1a2504e99 : SUCCESS in 1h 18m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01fd5459a6114249a18ae4cc33f4e01b : SUCCESS in 3m 58s","accounts_in_message":[],"_revision_number":1},{"id":"2375a71f670d03944e7ea6e090d6eb09070fe150","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-02 09:49:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9067c4f4a98ffdc84d2ac3482eb18caa7a3e72ac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-02 11:23:39.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/aa8f0a4e37a4431d907a1c83b5479f36\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cc7c18ca5dba456c8d1c2a5023cb2a1b : SUCCESS in 6m 14s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/a711764bf6db482a8a87ede21caddffe : SUCCESS in 1h 06m 40s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/ce7eadf2f76442c6906a915bd2004673 : SUCCESS in 1h 13m 49s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2571b919aad446bfa36e27c4d70fafc6 : SUCCESS in 1h 02m 19s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/33ff17e9415a42f29b24a5e115a3d3b7 : SUCCESS in 1h 12m 42s\n- openstack-ansible-deploy-aio_metal_quorum-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a158e7074520454d8f46a9cca091d115 : SUCCESS in 1h 10m 23s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5918128785c1492eb25a448d737081b4 : SUCCESS in 1h 20m 11s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/18958c60704c41e389255f7b264a76f2 : FAILURE in 1h 04m 09s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ea7ae2ed3aa94b06853c8be61e33f6ac : SUCCESS in 1h 25m 04s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a9b73a1ffe68496f8bf0745e441d015c : SUCCESS in 1h 18m 41s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/ff17a0f23c9b4a429806ed8e780745fc : SUCCESS in 1h 09m 32s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/4cfd255893e54c448cae6e38d86c8cbc : SUCCESS in 1h 13m 49s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/901dd439e346423ca5a6efff085c9df7 : SUCCESS in 1h 02m 34s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/71514c5dc02c4f0890a35add7c15600e : SUCCESS in 1h 09m 57s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fdd372e3c4534ec1a17da7c3ef55bda7 : SUCCESS in 1h 06m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de916769374c49789379eb3cbf3ca718 : SUCCESS in 4m 05s","accounts_in_message":[],"_revision_number":1},{"id":"9339b4a806b9e5d3716518fda827a8f4c10514ed","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-09-02 13:21:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"692c0c6a22fcdc8bc7cd046ca7dd36073854fc0d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-02 15:42:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/25869b40cc4345c8a44f23477100609f\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/59b216db40154425a6573fe6dbb90703 : SUCCESS in 7m 28s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/bfb8205499704933b30dc301cd5d27e9 : SUCCESS in 1h 07m 06s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/061cf43769ba47948c0acfcec4beaad9 : SUCCESS in 1h 17m 28s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/c6c06cc070474177b7bbeff52a14a4e5 : FAILURE in 1h 01m 57s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a3771fceda0546b7ac5bd9e6bdfcc2f2 : SUCCESS in 1h 55m 52s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cdf11e5cc1774d88a1bfb1b39acf7424 : SUCCESS in 1h 10m 44s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/aaec2fd6a9fd4d7397771dc9330ed0b8 : SUCCESS in 1h 23m 34s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e5bc4405cadc4f50b9227769fe685eaa : SUCCESS in 1h 18m 02s\n- openstack-ansible-deploy-aio_metal_quorum-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2fdd457db7a3494a9b852bbcd2df9f8f : SUCCESS in 1h 11m 55s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/cd396da1b31e4cdd80fec44c82113a9e : FAILURE in 1h 06m 56s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/40a739cb5b2e40469c07f43b0a6f5228 : SUCCESS in 2h 09m 01s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/448c6ebe40b144ae989905af608f8971 : FAILURE in 1h 07m 11s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/901b188637d84fb7be04974449bd789b : SUCCESS in 2h 05m 27s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0b9adb90288b42ef92aeddacc0f2a9c3 : SUCCESS in 1h 08m 10s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a43bedf142084efbbe47c08a9ca2c7e0 : SUCCESS in 50m 08s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/866a6c9c427e474eaa96179a62f48326 : FAILURE in 59m 53s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/94f35df44aac4cd1b1462d0adea5b4c7 : SUCCESS in 1h 20m 03s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/409385b6c7064f098b1bb8079e727585 : SUCCESS in 1h 27m 56s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/1f6bad70e7cc49f5b3181c1b1e4b11ae : SUCCESS in 1h 09m 50s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/d67e850a30b7446f9c603ea52e34037c : SUCCESS in 1h 17m 34s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fa5f3a0db6a94997a3604983ab1689a5 : SUCCESS in 1h 12m 28s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/151488332a2d4204991acc0941437d6a : SUCCESS in 1h 10m 44s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c72105076ad4443480c6843c02599795 : SUCCESS in 1h 07m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4521f8facc644774b786ccbd2f4ffcb5 : SUCCESS in 4m 20s","accounts_in_message":[],"_revision_number":1},{"id":"10dae7c34f266f99486a2841da39b2c35b2f1ccd","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-09-05 07:49:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8485b683148f76b51f61e100c5975ed53983cc92","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-05 10:03:03.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2982e42a65664797b0ae511fb1504638\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/01c29750f57b40e58f512021b8b15ac9 : SUCCESS in 6m 33s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/32e677d73df747ff8ad7043bca064cfe : FAILURE in 38m 51s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/fcb42089d84145a3b85202f852f50336 : SUCCESS in 1h 17m 01s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/c0e2a8994a8b4772b5174f18f177dee8 : SUCCESS in 1h 19m 24s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b069f396ef2d4bb7b363c6439e1c94f3 : SUCCESS in 1h 16m 41s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/38449f4a5a204da0a79932878729ec5a : SUCCESS in 1h 16m 40s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/210a0f178c2d436ebf315842daf3804b : SUCCESS in 1h 21m 39s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1bdb90628faa4fafa931a23ca3a3b11e : SUCCESS in 1h 10m 50s\n- openstack-ansible-deploy-aio_metal_quorum-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e231385916994452921395e0a77f1bc3 : SUCCESS in 1h 17m 43s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/8b948d3804dd44b1b54a058756b6918e : FAILURE in 6m 28s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5beac099234b4d1ea8e9c5f6bccb2b16 : SUCCESS in 1h 56m 54s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d02f4bffe47446d299f00b6bb6d82a72 : SUCCESS in 2h 09m 55s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c2d22b6679b94e6d899807bc14080456 : SUCCESS in 1h 58m 16s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/91d76d39362145ff96fcc238b4b47f90 : SUCCESS in 1h 14m 38s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/29fbbba46f1c4169b76f78dc633f09cc : SUCCESS in 59m 07s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/70011bfb13904b93b5c976ca7d3aefcb : FAILURE in 1h 00m 25s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/12473350551f4ed4b33e0cd56e6f0f6c : FAILURE in 30m 28s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/cedda0d7c8c5420e9b262ff2b964f747 : SUCCESS in 1h 32m 32s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/3c0a2dd64a1f47a8975909fc334aaec7 : SUCCESS in 1h 15m 06s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/698c396d17ab404a8c49a89d339e56de : SUCCESS in 1h 07m 34s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3e739a6cbbcf468c92a4860f3f110b8a : SUCCESS in 1h 09m 52s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d8c61980a3004f929e90b3a946e30169 : FAILURE in 32m 23s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2c40d5aef3ef4d8280e136b35c707db2 : SUCCESS in 1h 16m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/211ae0f3088540b39daad8571eb69dbb : SUCCESS in 3m 52s","accounts_in_message":[],"_revision_number":1},{"id":"8d11c230044a03de1c73056d2c0ebdeefa7b03df","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2024-09-09 17:12:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"255e5384ce63747892fd57cde75c61c86b444020","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-09 21:09:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9bcb79b550e34a549a4264bf8542af0f\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e0835e33d6e347af81db4c94dfd4a8a5 : SUCCESS in 6m 14s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/9e05912ba745441381fdb583114956d3 : SUCCESS in 1h 02m 56s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/f3dd305359d6453289ea8732a58b3acf : SUCCESS in 1h 09m 20s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/41c5d2173e714f0483fbe140b19ca7d0 : SUCCESS in 1h 16m 01s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0896bff9a28f49ba8dabc7e16bb1be04 : SUCCESS in 55m 02s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d0a38a15b81e48dfb452162e36ef5cc0 : SUCCESS in 1h 12m 09s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6ea46528a62a4b52b45e381146671bac : SUCCESS in 1h 11m 53s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c6ba7a288ac44212a1f97a002bb532a0 : SUCCESS in 1h 27m 12s\n- openstack-ansible-deploy-aio_metal_quorum-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8425b8d37cdf417d87a8c791936301c6 : SUCCESS in 45m 25s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e1aafa784685418e9f17a2d7c8c8d1a7 : SUCCESS in 2h 09m 50s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/07ba34292d0c4c6da266ac00c869e7e3 : SUCCESS in 2h 10m 09s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4e66d4407bc14ae6b1046d400126e0a1 : SUCCESS in 1h 58m 53s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a98386df21084e938b464e43125a74df : SUCCESS in 2h 08m 10s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1236d519f0464d3da03834ec8fa54ef2 : SUCCESS in 49m 55s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8704d12b1a3148f3baa07d6cede03440 : SUCCESS in 1h 03m 16s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/b9c51fb91af34b6c86334c3666e2e5bb : FAILURE in 1h 04m 03s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3448ab887e684118a80a21e31956c022 : SUCCESS in 1h 30m 03s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/65fcd18ac7e44fd6af58eb91f8b2467b : SUCCESS in 1h 04m 10s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/0324c773a8d64d0bb777e422480a24bc : SUCCESS in 1h 14m 21s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/5759100a84d7407fa1a3fc500722dbc2 : SUCCESS in 1h 08m 09s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bd748c59b79542658e24e1c69cc6e34e : SUCCESS in 1h 19m 37s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b12d810a460a406fa5a5c26d6187edb9 : SUCCESS in 1h 09m 39s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/48dc8dcc68144cc38345fffd07a775ba : SUCCESS in 1h 13m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ffb9553ac85f4d619830fdcf72556075 : SUCCESS in 4m 23s","accounts_in_message":[],"_revision_number":1},{"id":"dbd542891f84e46e498ceb5da9cd9f0a39710a32","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-09 21:11:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"50703d98ebbb8e302d1d953dc8e847a644c861c6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-09 22:47:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/839030b4256c45919880357c1b99d2a2\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b0fffea05d48429c98373d95aa332cc5 : SUCCESS in 5m 28s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/19d1e270e69247cd933e9f2737c6f1e1 : SUCCESS in 55m 20s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/15063186b47f4db28e49a38a61687b7b : SUCCESS in 47m 04s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d53c5c7750e942d3b56200bf19e7b5a7 : SUCCESS in 51m 15s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7c7d56f132aa4be9b3c950445a1b14d5 : SUCCESS in 1h 03m 16s\n- openstack-ansible-deploy-aio_metal_quorum-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bd8f9f69a297441199e3424704f615ab : SUCCESS in 1h 12m 55s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2969d6c940d549ee961f1a876e4b3201 : SUCCESS in 1h 10m 09s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ffe800e4dff4437398d3ce79b5176e2e : SUCCESS in 1h 14m 52s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a7330fc91a424a819b7e89493175811e : SUCCESS in 1h 26m 24s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/38aae03822f04b1fa9243a34175b4b04 : SUCCESS in 1h 25m 50s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/ffa72b0ce33a405aa473865f206a077c : SUCCESS in 1h 12m 32s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/b0e40d9bf6c040488254b6ae68cce0a9 : SUCCESS in 1h 23m 30s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/915882c872c24027aa947263a18b9544 : SUCCESS in 1h 27m 49s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fdddcccae0a04cafbace2d47e9040a5b : SUCCESS in 1h 11m 19s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/041f7619f8e5459c97974f171eabb670 : SUCCESS in 1h 14m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1e7a83ebb27485096b7ab454f7da8f4 : SUCCESS in 4m 09s","accounts_in_message":[],"_revision_number":1},{"id":"45ed5a49a79f9b1fe166a04ead4733cadaa33a62","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-09 22:47:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4a763624897e045068736aa2116510013b1d234b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-09 22:48:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a582093797c433494660b51903087cf\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bdd8298ce4d2417bb842ee9f8b670927 : SUCCESS in 34s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b7f0278d61f10d7fd854e6aa6fe08338a082d12b","revisions":{"b7f0278d61f10d7fd854e6aa6fe08338a082d12b":{"kind":"REWORK","_number":1,"created":"2024-08-26 15:01:31.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/55/927155/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/55/927155/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/55/927155/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/55/927155/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/55/927155/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/55/927155/1"}}},"commit":{"parents":[{"commit":"9c45b20fd8f5f042887584d199426d89f68548c3","subject":"Remove X-Forwarded-Proto header in apache","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/9c45b20fd8f5f042887584d199426d89f68548c3"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-07-25 12:27:57.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-08-26 15:01:31.000000000","tz":0},"subject":"Do not install uWSGI as part of Keystone","message":"Do not install uWSGI as part of Keystone\n\nuWSGI isntallation is expected to be handled by our uwsgi role, thus\nit\u0027s installation with keystone is not needed neither for source nor for\ndistro path.\n\nRelated-Bug: #2073873\nDepends-On: https://review.opendev.org/c/openstack/ansible-role-uwsgi/+/924922\nChange-Id: Ib10022bc2b29e556d5254827122c138e90381c0b\n(cherry picked from commit 09c41fab997281e3198b494f749832f0424b7435)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/b7f0278d61f10d7fd854e6aa6fe08338a082d12b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/b7f0278d61f10d7fd854e6aa6fe08338a082d12b"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
