)]}'
{"id":"openstack%2Fopenstack-ansible-os_keystone~927156","triplet_id":"openstack%2Fopenstack-ansible-os_keystone~stable%2F2023.1~Ib10022bc2b29e556d5254827122c138e90381c0b","project":"openstack/openstack-ansible-os_keystone","branch":"stable/2023.1","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2024-08-30 12:14:02.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib10022bc2b29e556d5254827122c138e90381c0b","subject":"Do not install uWSGI as part of Keystone","status":"MERGED","created":"2024-08-26 15:02:10.000000000","updated":"2024-08-30 12:14:56.000000000","submitted":"2024-08-30 12:14:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"927156","cherry_pick_of_change":924928,"cherry_pick_of_patch_set":3,"meta_rev_id":"9926b84f95cb0d828a632d9bc3f51213c30ac2e2","_number":927156,"virtual_id_number":927156,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-30 12:14:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2024-08-30 09:36:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":2,"date":"2024-08-27 07:05:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2024-08-30 09:36:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-26 19:20:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-30 09:36:46.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"c7ec9efb5f4a22c26d5927b7620bea58c77aa41c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-08-26 15:02:10.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"61bfd8911c02199186e9284c4dc0a84d2233e052","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-26 19:20:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/82397fb4cc194175ac184f8535f4d259\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e9b1e7d946964c76af147b6a1d0fe754 : SUCCESS in 10m 15s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/175eb70976584215b8231cef5442468e : SUCCESS in 1h 32m 18s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/de49664d48df47e7934309ec0a82b110 : SUCCESS in 1h 25m 38s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/213b02f6579647f3825483fa243458d5 : SUCCESS in 1h 38m 56s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/da1b79d60cf0480b98329af57245b93b : SUCCESS in 1h 37m 30s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c5861757537f43ca84cfcab54e2b629c : SUCCESS in 1h 24m 59s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6d65423bd33649f5961604aad29c6b19 : SUCCESS in 1h 45m 51s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f877e3cb45224f6fb863d7d1c70efb7f : SUCCESS in 1h 26m 18s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6d34e6434b074e2bb85d69d3e9c0c02a : SUCCESS in 2h 10m 26s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e6ee565e81f7438b894516b612d993e5 : SUCCESS in 2h 05m 10s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3f69f9ece51446729b60b249cb67b06d : SUCCESS in 1h 08m 36s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0f8d67be77dc45ea98cd36ab9d9252ce : SUCCESS in 55m 21s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/d0a47eb9e0e94baa9181ccea1cc88a15 : SUCCESS in 1h 26m 59s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0a2a06c2d93c404f9418027f47ee070a : SUCCESS in 1h 24m 13s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ba5dadcde3c545aca322d72650db5335 : SUCCESS in 1h 30m 00s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/cb1db5283d9248a6833fa50fbfa596a5 : SUCCESS in 1h 17m 41s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dbd56b72d0ba4027924056b2c2d4ae5e : SUCCESS in 1h 21m 56s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/63c6fc76b9e9450881d103561edcdceb : SUCCESS in 1h 14m 22s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ef4964e670854b5a93eab3066dda93f0 : SUCCESS in 1h 28m 27s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/09dc150bcad74091b41e5f4cb0357f17 : SUCCESS in 1h 25m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/024c5e6bc372444089ea21043b565149 : SUCCESS in 4m 24s","accounts_in_message":[],"_revision_number":1},{"id":"5b41277ee17b8bdc8633abcea6a126a2c26b2837","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-08-27 07:05:26.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"dc9cf60cdde854ae0483cf2fe0f81244e65ad1b0","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2024-08-30 09:36:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"85afb06aaa964c3057d4ff89d5d7c23d6a3eb566","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-30 09:38:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"714f406544e48d9a29f42af9b6208dc475e4a3c5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-30 12:14:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9412c033a2e04ba89bbf836ff09a683c\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/be4463e3801f439aab1b075a4e68d2cd : SUCCESS in 6m 32s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/93bc8cb7ca6c4b14870d2fd922c0b780 : SUCCESS in 1h 17m 18s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1c873f67ef9448bea4fe51a63f86b0fc : SUCCESS in 1h 34m 52s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/adb5b2a8025a42498e39daa1f7d6f4ae : SUCCESS in 1h 27m 07s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f0d9b4b1efe54f8d959a66f847320a17 : SUCCESS in 1h 33m 35s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5d1a491da0a34cfc850729c1d030a8bb : SUCCESS in 1h 33m 59s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f71b40259ccb44e1bacda696299937b9 : SUCCESS in 1h 27m 01s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c93f72ed98d44d37af58e75b5f050662 : SUCCESS in 1h 28m 36s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/5eec6e3640e945e2bc6fcbb606a30d91 : SUCCESS in 1h 05m 30s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c29b785454c9469a841c75bef95b3e0e : SUCCESS in 1h 10m 00s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1acac3f2bd9041e2a6c27a3d3b39fe21 : SUCCESS in 1h 18m 18s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/35294695853742308b10be8c616bc8c8 : SUCCESS in 1h 17m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/def75fba45af490c8077a9d0ad9456d3 : SUCCESS in 4m 14s","accounts_in_message":[],"_revision_number":1},{"id":"547b9bfd44b18d6a8b29ba4c9704f67fc2cd60c1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-30 12:14:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"9926b84f95cb0d828a632d9bc3f51213c30ac2e2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-30 12:14:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ed0af6cfff04df6b846906a2defb0c9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52a88f745f4949a2a22e0b19f6725cd0 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2e2096a562dfac60c64d28662c0ffe50202a0f78","revisions":{"2e2096a562dfac60c64d28662c0ffe50202a0f78":{"kind":"REWORK","_number":1,"created":"2024-08-26 15:02:10.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/56/927156/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/56/927156/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/56/927156/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/56/927156/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/56/927156/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/56/927156/1"}}},"commit":{"parents":[{"commit":"a45472ac4fbd8eb4a3230296b820f494650ef722","subject":"Remove X-Forwarded-Proto header in apache","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/a45472ac4fbd8eb4a3230296b820f494650ef722"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-07-25 12:27:57.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-08-26 15:02:10.000000000","tz":0},"subject":"Do not install uWSGI as part of Keystone","message":"Do not install uWSGI as part of Keystone\n\nuWSGI isntallation is expected to be handled by our uwsgi role, thus\nit\u0027s installation with keystone is not needed neither for source nor for\ndistro path.\n\nRelated-Bug: #2073873\nDepends-On: https://review.opendev.org/c/openstack/ansible-role-uwsgi/+/924922\nChange-Id: Ib10022bc2b29e556d5254827122c138e90381c0b\n(cherry picked from commit 09c41fab997281e3198b494f749832f0424b7435)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/2e2096a562dfac60c64d28662c0ffe50202a0f78"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/2e2096a562dfac60c64d28662c0ffe50202a0f78"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
