)]}'
{"id":"openstack%2Fopenstack-ansible-os_keystone~929691","triplet_id":"openstack%2Fopenstack-ansible-os_keystone~master~Ia3e4af7986166f0729840d2a61fb8f52ea053676","project":"openstack/openstack-ansible-os_keystone","branch":"master","topic":"osa/apache_mpm_alignment","attention_set":{},"removed_from_attention_set":{"31542":{"account":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"last_update":"2024-09-27 12:25:33.000000000","reason":"Change was submitted"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2024-09-27 11:03:18.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2024-09-27 12:25:33.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia3e4af7986166f0729840d2a61fb8f52ea053676","subject":"Ensure that selected Apache MPM is enforced","status":"MERGED","created":"2024-09-17 18:07:46.000000000","updated":"2024-09-27 12:26:28.000000000","submitted":"2024-09-27 12:25:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"929691-osa/apache_mpm_alignment","meta_rev_id":"390ad8530db39facd97fad3c9267fed935cc57b5","_number":929691,"virtual_id_number":929691,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-27 12:25:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":2,"date":"2024-09-27 10:14:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":2,"date":"2024-09-27 11:03:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":1,"date":"2024-09-27 11:03:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":1,"date":"2024-09-27 11:03:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-17 21:53:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-23 10:43:19.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2024-09-25 07:13:40.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2024-09-27 10:14:45.000000000","updated_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"reviewer":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"state":"REVIEWER"}],"messages":[{"id":"fa2b0c6db64f19ae421451b3fe7245076be42788","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-09-17 18:07:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"30ca22cf600caa7c1c571bfa01de1c63d7158dc8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-09-17 18:12:29.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"d897c8c19fecc488e2e9589b3992c105138f5ef5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-09-17 18:13:57.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"e1e78066da4a8a1ef63effa3e57a82e50a4d10da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-09-17 18:15:59.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"61a1afd089bbeee0a6721d07091a78bdbb5bf375","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-09-17 18:25:15.000000000","message":"Patch Set 4: Backport-Candidate+1","accounts_in_message":[],"_revision_number":4},{"id":"4a25d4b35ce5339ac86c2fc57504f4c47d0c28b4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-09-17 19:52:28.000000000","message":"Patch Set 5: Published edit on patch set 4.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR **is:MAX** OR is:MIN\")\n","accounts_in_message":[],"_revision_number":5},{"id":"71bf62cff34eae89cdc15a317386793091966dec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-17 21:53:11.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0574afbde6854d5daa756a8655c7c3e3\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e1d2d169fee444c0838a04a8c7ae29af : SUCCESS in 6m 05s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/2f3761ee2c9f46629c793a02b6d1a130 : SUCCESS in 1h 20m 58s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/9b4fac0ed4454ef49dbab742ea1728f0 : FAILURE in 30m 20s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d9b2b317d4bc4d1caa0285100f184d89 : FAILURE in 35m 58s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0db322e3dc2e475d87a0201b3243bf31 : SUCCESS in 48m 10s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8913e7de8af54d4d9a657580d9786dc4 : SUCCESS in 1h 11m 47s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/053a07854e634ec3beaa78a08fc1b845 : FAILURE in 33m 44s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/26c430cd66444417b8f5158149bc3045 : SUCCESS in 1h 18m 48s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/69a0825e93b9465e9f792c93a24415cf : FAILURE in 49m 45s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5a528191f7394a728abe1c4185bf4474 : SUCCESS in 1h 21m 50s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c2e629d7523345c192f8526e690f7761 : SUCCESS in 1h 28m 26s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/689b3120e3ac451795a7dfe3570f5149 : FAILURE in 1h 19m 19s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d20a9f0b0277484c85098053825bef70 : SUCCESS in 1h 06m 36s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/da680f6a628b41e3b59298a9a1cf78f9 : FAILURE in 1h 18m 27s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3cf1c5448a2c4dd48bdd3630815455b4 : SUCCESS in 1h 57m 20s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/97e656db9c8a463f8358bbf57bd897e4 : FAILURE in 27m 24s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5c7a1a016d87410aa58f42d1d4602d4b : SUCCESS in 1h 01m 21s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/69778bb3c6b74f5b979028daa14ebd3a : FAILURE in 54m 36s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e83e2fb46e4048f8b86ad0076b065710 : FAILURE in 49m 24s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ac76e2ea51e54720b0e7d999357b1834 : FAILURE in 1h 11m 50s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/c755ef41c2e34852b1dd6f195d61f0c1 : SUCCESS in 1h 17m 37s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2cb6c2443f084c4ebd500e2f17a0597f : SUCCESS in 1h 06m 24s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e97ea4e35bea4272ba97c2011dde200a : SUCCESS in 1h 09m 25s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bc5c14de656a4aaf8c4c056f95306167 : SUCCESS in 1h 27m 36s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2f9ab84369124df99c6db933cc6d08f7 : SUCCESS in 1h 14m 24s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cbf94d23483e43fea80cb0658630bfcc : SUCCESS in 1h 17m 58s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/95918ed5f23f453d82ab45f87200bfc2 : SUCCESS in 1h 09m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2b3448865544c10a8a52bf273b6a99a : SUCCESS in 3m 31s","accounts_in_message":[],"_revision_number":5},{"id":"d8d772cd1cd9c2818b39348833e1407ecaa3f494","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-09-18 08:26:59.000000000","message":"Patch Set 6: Published edit on patch set 5.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR **is:MAX** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"49b6241350dffba66862468532fd594bf7dcff85","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-18 10:33:17.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c7bbf2ae16e34b65be1ba579ee4e2022\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3953b482d6884bb0bb030044e9d83033 : SUCCESS in 7m 55s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/778b277803a84d449b59902209119755 : SUCCESS in 1h 22m 04s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/1a19333c1308428d9e0bef1fd3d4c656 : FAILURE in 32m 34s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/dbd32c955fa14395ba495efb90f44247 : FAILURE in 36m 20s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/54fbe58822c3459e9039d226f18d2199 : FAILURE in 1h 01m 13s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a19312771bbf4f498e6ca7f3bd4268b2 : SUCCESS in 1h 16m 39s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4c21ba476e914db9a62762798a3762b3 : FAILURE in 35m 38s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/22c97e81ce1e44658dbeaf05af295f6f : SUCCESS in 1h 17m 33s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f355f28fb03345d6bbd8305b4e0131d7 : SUCCESS in 1h 03m 56s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cbc360767ae34fa7bdde14610e356ec1 : SUCCESS in 1h 17m 17s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-noble https://zuul.opendev.org/t/openstack/build/947edea06dfb4609b4d5ae2cae8e62b4 : SUCCESS in 1h 17m 53s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/16cedb94d62e4946ad83fc2c458fa315 : FAILURE in 1h 23m 38s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/47455bc0c873444aa46f4be2bcde2360 : SUCCESS in 1h 42m 41s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6854f0e6f8e249288ed4960bac5781cc : FAILURE in 1h 24m 19s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1fac9db0721f46c39c8f2e75dc6d9fb7 : SUCCESS in 1h 54m 28s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6e51592e98a7470aa7ddf80b150ecebe : FAILURE in 29m 14s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f576ad93ed50491ca56d316b7628ea7a : SUCCESS in 57m 17s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/4ffb55f89f27454a99dbf22e10717920 : FAILURE in 56m 13s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d7ade2fc63da471cb7623f6edbe11089 : FAILURE in 1h 13m 53s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/29ed93e18cb847859c7af09ee3e6bf82 : FAILURE in 1h 15m 05s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/1326e6c142ec441c84bcb72b5c8c8cc1 : SUCCESS in 1h 20m 09s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/82efa1f12e0547d68c7d545e54461fae : SUCCESS in 1h 12m 49s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0e9dbb535a0e4a67b94381cbca8c5599 : SUCCESS in 1h 03m 36s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fd1c15927b4d497f9358c35fcaeb7125 : SUCCESS in 48m 56s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/eb392e6f3f534590bbe3fa4dc8eba39c : SUCCESS in 1h 20m 30s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1e8e407872bc4b3c99d7f04fdc204d14 : SUCCESS in 1h 18m 58s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ab083400d6794dd7ba9a82d96673b955 : SUCCESS in 1h 20m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4bc61cfe2ae4cfb8880b690ef131404 : SUCCESS in 5m 54s","accounts_in_message":[],"_revision_number":6},{"id":"be9a3515a84fdadef9d4f987942415e432895e40","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-09-18 13:29:08.000000000","message":"Patch Set 7: Published edit on patch set 6.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR **is:MAX** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"feb3c5998c9f337ec7509b0b1772719dcf4585e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-18 15:42:15.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec422469619645ad97608057019a6d78\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ae58086bd88646b5a7a8b45ca72e7167 : SUCCESS in 8m 38s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/063b3edab6584e34a0ff799ad53c9fd9 : SUCCESS in 1h 27m 04s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/ffbb2c0b0cc0449f8c632278474803c0 : SUCCESS in 1h 22m 42s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3afedb6b3cf64972b53167492a525e44 : SUCCESS in 1h 28m 13s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6fb4606ffea34941b3f964ffc0151a81 : SUCCESS in 1h 19m 32s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/af9bbe57cb93469884f0d8012ecb7686 : SUCCESS in 1h 29m 06s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ae9eaaa4505b469684e1e16f697e1228 : SUCCESS in 1h 27m 20s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e6ae6bfef83e4a2f8d80e114f2ec3535 : SUCCESS in 1h 14m 17s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c043af30d1f443e7ba265bc64334d776 : SUCCESS in 1h 19m 32s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b668d46650b3445aad8fefb27f60edf0 : SUCCESS in 1h 24m 36s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-noble https://zuul.opendev.org/t/openstack/build/74b31f709d5f4cefbcae86d3441e2185 : SUCCESS in 1h 18m 25s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/96f962ade4e84ac3a80e5220a2883820 : SUCCESS in 1h 12m 58s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4ca2a4210b2441ca97c719e48cfba571 : SUCCESS in 1h 45m 03s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6cd2d0f062e6478681fc3c0155bf688c : SUCCESS in 1h 55m 24s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8a15f84e1c0746529c0fd445d5ebeec1 : SUCCESS in 2h 05m 31s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/21e83ab910054de4a7402f54a2a148ef : FAILURE in 56m 43s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4eb7eb64a0b04be6b4eb1eaf861932d0 : SUCCESS in 1h 01m 11s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/0a750d8fc0684e2887d720f354fb2cb1 : FAILURE in 51m 01s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/22d59aa26af341949fe1efa1887da96e : SUCCESS in 1h 21m 20s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/103f03db11b94a8eb1d55bc39b93b08b : SUCCESS in 1h 15m 09s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/e863759eedfa457ab93d37f68ac65dd1 : SUCCESS in 1h 15m 47s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/26679d38a82045178c5e139f3091209a : SUCCESS in 1h 15m 48s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/53b6c07fa44a4e828c2f692e8dc853ae : SUCCESS in 1h 17m 21s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ac9a2a6eb08149c88e0aed6d0577914e : SUCCESS in 1h 18m 57s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/960fc307638843249a1d1eaacbced381 : SUCCESS in 1h 21m 41s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/abc8b452face4d8a9d2b0a4cfab2a04b : SUCCESS in 1h 21m 50s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b241cf0e0e184bef945175a403a2f0a4 : SUCCESS in 1h 23m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1284d13e82c44c74acf1503bd7714f33 : SUCCESS in 4m 09s","accounts_in_message":[],"_revision_number":7},{"id":"c525fb9012c93e64aad33575afee10e5dbcef9c5","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2024-09-23 10:43:19.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"b612f24641d2b355ac741aea1a2bad66aa399684","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2024-09-25 07:13:40.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"716177e45b256ae34cd3869c6dd36942a6447a83","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-25 07:15:41.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"cb86653e16d50e37127260ae062c9adec5aaf709","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-25 09:14:42.000000000","message":"Patch Set 7: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cc500bef163e43cbab51558757dbcb75\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/93d4929abea3422d992257d271392b65 : SUCCESS in 6m 01s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/be444a43db51449ba5b5a020294bb397 : SUCCESS in 1h 13m 04s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/07fc93d80d904ddcb18db3765eb86f45 : SUCCESS in 1h 21m 49s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7b9deae5fedc451589191a5ecfda115c : SUCCESS in 1h 08m 58s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/550db37435e241b684e66a13468ce47e : SUCCESS in 1h 21m 49s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b823042bde5f4d57b13d3444c2585d6f : SUCCESS in 1h 15m 33s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2e9d1fab8bc740b780d9492064833962 : FAILURE in 24m 01s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/871b72db20394b31b59dd4b4f9e718cb : SUCCESS in 1h 15m 45s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/193cf0f8447c4e9e98ba652e7409a6ff : SUCCESS in 1h 09m 48s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d30ac19222c44432b9e042ec6f7ae7d4 : SUCCESS in 1h 16m 48s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/07c875a84d444c359b91d0830fea300f : SUCCESS in 1h 11m 52s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/13dcac50a49f4d59b7e3bfd2375f5acf : SUCCESS in 1h 13m 24s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/ba2002bc7db64739972023824a4085f1 : SUCCESS in 1h 07m 21s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/91f10d155ada4278a8ba87e9d1bd6ba4 : SUCCESS in 1h 04m 46s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6c28f581616045d5841d71e9f2bb4f10 : SUCCESS in 55m 07s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5ff3a87a56b045dcb88ed29bfb74ad31 : SUCCESS in 1h 09m 04s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/51e2736a92c143c69bc7d6ad73bce0d3 : SUCCESS in 1h 06m 39s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3a080df605eb42fe999867724a0169a3 : SUCCESS in 1h 10m 19s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ad34139e95274d1a8dfce0ce6ce0659b : SUCCESS in 1h 11m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74b7671e80824cd7b0a856942f5358a8 : SUCCESS in 4m 23s","accounts_in_message":[],"_revision_number":7},{"id":"28deceaa6f283f9dd9769b172448ab32e4da551b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2024-09-25 15:48:55.000000000","message":"Patch Set 8: Commit message was updated.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"**changekind:NO_CODE_CHANGE** OR changekind:TRIVIAL_REBASE OR **is:MAX** OR is:MIN\")\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-2\n* Workflow+1\n","accounts_in_message":[],"_revision_number":8},{"id":"aa8d0d71d7fd9b3a3f8be70893a5653076e30060","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-25 18:02:41.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/baf311637ff741ba807984e2f0435f59\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/675427cf50ee468286a150cb6d44a59c : SUCCESS in 7m 59s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/88a4f7e510e54b45ba9189c986d33269 : SUCCESS in 1h 05m 21s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/224225e6668c4c0e85d9e7ba659344e2 : SUCCESS in 1h 19m 36s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/70061bc2da994ee59a2fee8d910afe3a : SUCCESS in 1h 14m 02s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/556b5a68372348fba30715345a2baa31 : SUCCESS in 1h 14m 15s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1157ae915170414c9c740e7192188b4a : SUCCESS in 1h 27m 53s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a5580e341d674071ba36619d1e6d18b5 : SUCCESS in 1h 15m 14s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/242d77414dcb4e978d7c6a42468ef9a2 : SUCCESS in 1h 11m 27s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f604206e1ece4e41847717f0f3e2489b : SUCCESS in 1h 08m 37s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/352fcfef1f5441928e9e0d43708f995e : SUCCESS in 1h 19m 48s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bc3a62dd35654e4d8eff8eea8ed5e7b5 : SUCCESS in 1h 12m 46s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/57d1150b3f0f4fdfa45cc88079ec3b40 : SUCCESS in 1h 57m 01s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9d7a26fd95514f8287bb8649e96a3f4b : SUCCESS in 1h 41m 03s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/496107c4c4e04175b59fb3a8f748a170 : SUCCESS in 1h 16m 54s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d1d0926f556748bdaa4d05bc2c93a125 : SUCCESS in 1h 44m 16s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/12cade0d088f4416890633f27d890ff7 : SUCCESS in 1h 02m 51s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/408f7d7aa0c24c3cac6c5b7059a1edb1 : SUCCESS in 52m 37s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/281af38f35944c66bfc9e9b7c236319e : SUCCESS in 1h 23m 19s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f4c1908d616a46d198b98dbbcdedac78 : SUCCESS in 1h 21m 17s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/559feb569b3f46c199cad8fe5353914d : SUCCESS in 1h 23m 48s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/6c18e30f8e324880a4203c889f066aad : SUCCESS in 1h 08m 56s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/82418b591ba745b08c00f1a9acc721e4 : SUCCESS in 1h 10m 54s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/88d56c931ed247feb4294cf861ff62d5 : SUCCESS in 1h 04m 05s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9c9743bda873495f91f6390c00c2bdc2 : SUCCESS in 1h 14m 55s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d10c7c9378184e5eb2d1d7282f5dbed9 : SUCCESS in 1h 16m 23s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/889695f796a9438db2b6d5f7eab8006e : SUCCESS in 51m 51s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/69c1f150fe7c4944895112901c2527a9 : SUCCESS in 1h 13m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d70aa5aa6a814647bcd22b0c698ebcef : SUCCESS in 5m 42s","accounts_in_message":[],"_revision_number":8},{"id":"a95cdb360a23465ed3af7c7b4e3a4d6aa67d49f4","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2024-09-27 10:14:45.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"7abab9702b3594ed8f26ccf1211c97cbe44d960e","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-09-27 11:03:18.000000000","message":"Patch Set 8: Backport-Candidate+1 Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"150896d06f952acb18063692affbaead503e9ba4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-27 11:05:20.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"380396a8236a86156a8034d98a1f0782d4256ddb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-27 12:25:33.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fde86c57a9ac4d45a689d724347ccadd\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/09a6cf740bce4186b0ab8e53d21db866 : SUCCESS in 6m 40s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/59b28cb58a2840d7a013ca5165bdbaba : SUCCESS in 1h 10m 00s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/feabb190c7fd4d1496b7599974d758b7 : SUCCESS in 1h 14m 52s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e54931d49d5b4be0814b06c342c80a31 : SUCCESS in 1h 11m 10s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/02a87416aa6342f9a392bfde8942daf3 : SUCCESS in 1h 10m 56s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/95695e813aab4304a6b9651e2875897c : SUCCESS in 1h 10m 08s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0cd2b0d69d8a4c4f85be560a870ce119 : SUCCESS in 1h 15m 49s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9537463e6cdb44a18c636b2fb7524485 : SUCCESS in 1h 17m 37s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8451699c924c42b4b714728ace4b1095 : SUCCESS in 1h 08m 35s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a5add92606134460b5097bb1b206949d : SUCCESS in 1h 16m 39s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/34860228eb6941938ac548df02fbebca : SUCCESS in 52m 46s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/470b9d22db604f42a2c7bdbd5b5a846a : SUCCESS in 1h 19m 32s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/c3fe6902da1f416298c00ae1285cf685 : SUCCESS in 49m 04s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7fcd5df46df14d7fbabdc2f4442cbe47 : SUCCESS in 1h 09m 36s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ce4d5bed3f394870878d014c5938bab4 : SUCCESS in 1h 09m 26s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e0495de6cbbe4e48813c2c4319bb0b98 : SUCCESS in 48m 14s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c8ecf7a6ae4643e09977519a55ccaa1f : SUCCESS in 1h 14m 48s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/37117254a2294238833cffd00ab2ed00 : SUCCESS in 1h 15m 11s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dee84831894345b4bcfbe633b4c007c8 : SUCCESS in 1h 13m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82c2ea62ca2b4b0eb4374feac186ccf6 : SUCCESS in 3m 22s","accounts_in_message":[],"_revision_number":8},{"id":"f6090bde19a0253f96908d508dc85d613538a837","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-27 12:25:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8},{"id":"390ad8530db39facd97fad3c9267fed935cc57b5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-27 12:26:28.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/07f26f47746f4461b2df4481e9cedb70\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1169124cd904c579ff4bcef30568d7e : SUCCESS in 35s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"870451e554d480ee8974827ada13b0b1042a812d","revisions":{"e6a0b3563e2e0012b094a6cb30dc57d31027f5cd":{"kind":"REWORK","_number":1,"created":"2024-09-17 18:07:46.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/91/929691/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/91/929691/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/91/929691/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/91/929691/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/91/929691/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/91/929691/1"}}},"commit":{"parents":[{"commit":"95641cbd26b35ed59e0614bac4cfdf1e6b13a07a","subject":"Remove excessive bindings for uWSGI","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/95641cbd26b35ed59e0614bac4cfdf1e6b13a07a"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-09-17 18:07:35.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-09-17 18:07:35.000000000","tz":120},"subject":"Ensure that selected Apache MPM is enforced","message":"Ensure that selected Apache MPM is enforced\n\nAt the moment services might have different MPM selected while all\noperating the same Apache setup, ie on metal setup.\nThis results in failures to set selected MPMs, so eventually second run\nof roles after initial deployment will end up in failure (ie upgrade).\n\nThis patch ensures that all except selected MPMs are disabled and do\nrole get\u0027s the desired state of deployment.\n\nWe also need to align selected MPM across all roles to avoid\nfuture conflicts.\n\nChange-Id: Ia3e4af7986166f0729840d2a61fb8f52ea053676\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/e6a0b3563e2e0012b094a6cb30dc57d31027f5cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/e6a0b3563e2e0012b094a6cb30dc57d31027f5cd"}]},"branch":"refs/heads/master"},"98d4ec860eb8e3e83f27a2cb32b58fc931656f4b":{"kind":"REWORK","_number":2,"created":"2024-09-17 18:12:29.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/91/929691/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/91/929691/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/91/929691/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/91/929691/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/91/929691/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/91/929691/2"}}},"commit":{"parents":[{"commit":"95641cbd26b35ed59e0614bac4cfdf1e6b13a07a","subject":"Remove excessive bindings for uWSGI","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/95641cbd26b35ed59e0614bac4cfdf1e6b13a07a"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-09-17 18:07:35.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-09-17 18:12:29.000000000","tz":0},"subject":"Ensure that selected Apache MPM is enforced","message":"Ensure that selected Apache MPM is enforced\n\nAt the moment services might have different MPM selected while all\noperating the same Apache setup, ie on metal setup.\nThis results in failures to set selected MPMs, so eventually second run\nof roles after initial deployment will end up in failure (ie upgrade).\n\nThis patch ensures that all except selected MPMs are disabled and do\nrole get\u0027s the desired state of deployment.\n\nWe also need to align selected MPM across all roles to avoid\nfuture conflicts.\n\nChange-Id: Ia3e4af7986166f0729840d2a61fb8f52ea053676\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/98d4ec860eb8e3e83f27a2cb32b58fc931656f4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/98d4ec860eb8e3e83f27a2cb32b58fc931656f4b"}]},"branch":"refs/heads/master"},"3eacc502b1d2bb7557fa0f7dfcdcca781ffd7345":{"kind":"REWORK","_number":3,"created":"2024-09-17 18:13:57.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/91/929691/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/91/929691/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/91/929691/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/91/929691/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/91/929691/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/91/929691/3"}}},"commit":{"parents":[{"commit":"95641cbd26b35ed59e0614bac4cfdf1e6b13a07a","subject":"Remove excessive bindings for uWSGI","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/95641cbd26b35ed59e0614bac4cfdf1e6b13a07a"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-09-17 18:07:35.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-09-17 18:13:57.000000000","tz":0},"subject":"Ensure that selected Apache MPM is enforced","message":"Ensure that selected Apache MPM is enforced\n\nAt the moment services might have different MPM selected while all\noperating the same Apache setup, ie on metal setup.\nThis results in failures to set selected MPMs, so eventually second run\nof roles after initial deployment will end up in failure (ie upgrade).\n\nThis patch ensures that all except selected MPMs are disabled and do\nrole get\u0027s the desired state of deployment.\n\nWe also need to align selected MPM across all roles to avoid\nfuture conflicts.\n\nChange-Id: Ia3e4af7986166f0729840d2a61fb8f52ea053676\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/3eacc502b1d2bb7557fa0f7dfcdcca781ffd7345"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/3eacc502b1d2bb7557fa0f7dfcdcca781ffd7345"}]},"branch":"refs/heads/master"},"49bacf11ffad50c427ff919c5dedf53072eb8f54":{"kind":"REWORK","_number":4,"created":"2024-09-17 18:15:59.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/91/929691/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/91/929691/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/91/929691/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/91/929691/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/91/929691/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/91/929691/4"}}},"commit":{"parents":[{"commit":"95641cbd26b35ed59e0614bac4cfdf1e6b13a07a","subject":"Remove excessive bindings for uWSGI","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/95641cbd26b35ed59e0614bac4cfdf1e6b13a07a"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-09-17 18:07:35.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-09-17 18:15:58.000000000","tz":0},"subject":"Ensure that selected Apache MPM is enforced","message":"Ensure that selected Apache MPM is enforced\n\nAt the moment services might have different MPM selected while all\noperating the same Apache setup, ie on metal setup.\nThis results in failures to set selected MPMs, so eventually second run\nof roles after initial deployment will end up in failure (ie upgrade).\n\nThis patch ensures that all except selected MPMs are disabled and do\nrole get\u0027s the desired state of deployment.\n\nWe also need to align selected MPM across all roles to avoid\nfuture conflicts.\n\nChange-Id: Ia3e4af7986166f0729840d2a61fb8f52ea053676\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/49bacf11ffad50c427ff919c5dedf53072eb8f54"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/49bacf11ffad50c427ff919c5dedf53072eb8f54"}]},"branch":"refs/heads/master"},"989f38143563659a4ec59fe15af4c40548efc480":{"kind":"REWORK","_number":5,"created":"2024-09-17 19:52:28.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/91/929691/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/91/929691/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/91/929691/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/91/929691/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/91/929691/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/91/929691/5"}}},"commit":{"parents":[{"commit":"95641cbd26b35ed59e0614bac4cfdf1e6b13a07a","subject":"Remove excessive bindings for uWSGI","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/95641cbd26b35ed59e0614bac4cfdf1e6b13a07a"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-09-17 18:07:35.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-09-17 19:52:28.000000000","tz":0},"subject":"Ensure that selected Apache MPM is enforced","message":"Ensure that selected Apache MPM is enforced\n\nAt the moment services might have different MPM selected while all\noperating the same Apache setup, ie on metal setup.\nThis results in failures to set selected MPMs, so eventually second run\nof roles after initial deployment will end up in failure (ie upgrade).\n\nThis patch ensures that all except selected MPMs are disabled and do\nrole get\u0027s the desired state of deployment.\n\nWe also need to align selected MPM across all roles to avoid\nfuture conflicts.\n\nChange-Id: Ia3e4af7986166f0729840d2a61fb8f52ea053676\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/989f38143563659a4ec59fe15af4c40548efc480"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/989f38143563659a4ec59fe15af4c40548efc480"}]},"branch":"refs/heads/master"},"cd28bb85d89e358310d39c8401ddf01e8e2608b0":{"kind":"REWORK","_number":6,"created":"2024-09-18 08:26:59.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/91/929691/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/91/929691/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/91/929691/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/91/929691/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/91/929691/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/91/929691/6"}}},"commit":{"parents":[{"commit":"95641cbd26b35ed59e0614bac4cfdf1e6b13a07a","subject":"Remove excessive bindings for uWSGI","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/95641cbd26b35ed59e0614bac4cfdf1e6b13a07a"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-09-17 18:07:35.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-09-18 08:26:59.000000000","tz":0},"subject":"Ensure that selected Apache MPM is enforced","message":"Ensure that selected Apache MPM is enforced\n\nAt the moment services might have different MPM selected while all\noperating the same Apache setup, ie on metal setup.\nThis results in failures to set selected MPMs, so eventually second run\nof roles after initial deployment will end up in failure (ie upgrade).\n\nThis patch ensures that all except selected MPMs are disabled and do\nrole get\u0027s the desired state of deployment.\n\nWe also need to align selected MPM across all roles to avoid\nfuture conflicts.\n\nChange-Id: Ia3e4af7986166f0729840d2a61fb8f52ea053676\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/cd28bb85d89e358310d39c8401ddf01e8e2608b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/cd28bb85d89e358310d39c8401ddf01e8e2608b0"}]},"branch":"refs/heads/master"},"9a462efa97e9b1e8b9ae208db311e04f42980e4b":{"kind":"REWORK","_number":7,"created":"2024-09-18 13:29:08.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/91/929691/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/91/929691/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/91/929691/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/91/929691/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/91/929691/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/91/929691/7"}}},"commit":{"parents":[{"commit":"95641cbd26b35ed59e0614bac4cfdf1e6b13a07a","subject":"Remove excessive bindings for uWSGI","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/95641cbd26b35ed59e0614bac4cfdf1e6b13a07a"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-09-17 18:07:35.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-09-18 13:29:08.000000000","tz":0},"subject":"Ensure that selected Apache MPM is enforced","message":"Ensure that selected Apache MPM is enforced\n\nAt the moment services might have different MPM selected while all\noperating the same Apache setup, ie on metal setup.\nThis results in failures to set selected MPMs, so eventually second run\nof roles after initial deployment will end up in failure (ie upgrade).\n\nThis patch ensures that all except selected MPMs are disabled and do\nrole get\u0027s the desired state of deployment.\n\nWe also need to align selected MPM across all roles to avoid\nfuture conflicts.\n\nChange-Id: Ia3e4af7986166f0729840d2a61fb8f52ea053676\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/9a462efa97e9b1e8b9ae208db311e04f42980e4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/9a462efa97e9b1e8b9ae208db311e04f42980e4b"}]},"branch":"refs/heads/master"},"870451e554d480ee8974827ada13b0b1042a812d":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2024-09-25 15:48:55.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/91/929691/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/91/929691/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/91/929691/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/91/929691/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/91/929691/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/91/929691/8"}}},"commit":{"parents":[{"commit":"95641cbd26b35ed59e0614bac4cfdf1e6b13a07a","subject":"Remove excessive bindings for uWSGI","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/95641cbd26b35ed59e0614bac4cfdf1e6b13a07a"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-09-17 18:07:35.000000000","tz":120},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2024-09-25 15:48:55.000000000","tz":0},"subject":"Ensure that selected Apache MPM is enforced","message":"Ensure that selected Apache MPM is enforced\n\nAt the moment services might have different MPM selected while all\noperating the same Apache setup, ie on metal setup.\nThis results in failures to set selected MPMs, so eventually second run\nof roles after initial deployment will end up in failure (ie upgrade).\n\nThis patch ensures that all except selected MPMs are disabled and do\nrole get\u0027s the desired state of deployment.\n\nWe also need to align selected MPM across all roles to avoid\nfuture conflicts.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-rabbitmq_server/+/930446\nChange-Id: Ia3e4af7986166f0729840d2a61fb8f52ea053676\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/870451e554d480ee8974827ada13b0b1042a812d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/870451e554d480ee8974827ada13b0b1042a812d"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
