)]}'
{"id":"openstack%2Fopenstack-ansible-os_keystone~931192","triplet_id":"openstack%2Fopenstack-ansible-os_keystone~stable%2F2023.1~I5d9505ed4b385673cc719e9821ee2e1c2ba5c754","project":"openstack/openstack-ansible-os_keystone","branch":"stable/2023.1","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2024-11-26 16:04:07.000000000","reason":"Change was submitted"},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2024-11-26 14:09:43.000000000","reason":"\u003cGERRIT_ACCOUNT_25023\u003e replied on the change","reason_account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}}},"hashtags":[],"change_id":"I5d9505ed4b385673cc719e9821ee2e1c2ba5c754","subject":"Change example to contain domain name instead of UUID","status":"MERGED","created":"2024-10-02 15:13:14.000000000","updated":"2024-11-26 16:05:14.000000000","submitted":"2024-11-26 16:04:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"931192","cherry_pick_of_change":919563,"cherry_pick_of_patch_set":4,"meta_rev_id":"2ab4a50712cca7f9711ffbdef6ab5036cfb798c1","_number":931192,"virtual_id_number":931192,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-26 16:04:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2024-11-26 11:18:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-11-26 14:09:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-11-26 14:09:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-02 17:24:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-20 08:34:58.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"ffa41ce3d5f3c32dc71322028da0795147f68b30","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-10-02 15:13:14.000000000","message":"Patch Set 1: Cherry Picked from branch master.\n\nThe following files contain Git conflicts:\n* defaults/main.yml","accounts_in_message":[],"_revision_number":1},{"id":"7c84d3a6758494cddb4904be798aec2d5bf0e82f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-10-02 15:18:21.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"589fcc0da80323ef11ca232a2b9ca3cc67d0bcea","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-10-02 15:18:33.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"917a9ad4ab4114071e6ec6b2515003d4e612f8d8","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-10-02 15:18:40.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":3},{"id":"f21b1d486105b719ca0aebbdd1de6b8851a36dea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-02 17:24:17.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5b7bce48ce0c43a5aa56aa0ecd76e678\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d1cb9ed59803488cb93d3c72f1b392e7 : SUCCESS in 5m 09s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/85187eb57bc3470ba00d941a65cf0cc5 : SUCCESS in 1h 28m 15s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/c4de01ce82694c1590d53b9eca865ffe : SUCCESS in 1h 26m 42s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/548fc77400694da99c6085be57f36cd4 : SUCCESS in 55m 51s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/67bc82021df14c4db08cfe4bc2232a0e : SUCCESS in 49m 52s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1cc58cebabb44706b1e7957ba221b229 : SUCCESS in 51m 37s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/538dead9c07848a9b6d8cfb5f96bd000 : SUCCESS in 57m 07s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/de31ab89b77f45788ecfe57fad8d9138 : SUCCESS in 53m 10s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/fcd27fe9deae435f9b637487356c66f2 : FAILURE in 13m 18s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f9024e92bc41479f97a0bb6d602d8441 : FAILURE in 20m 49s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/74ff2a40bc6d46fba65360a41a69383e : SUCCESS in 1h 03m 42s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4fcc456e57314f88841c65fe7c9e70bb : SUCCESS in 1h 01m 51s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/856821a28d794b39a8c7dd48e3cdd2df : SUCCESS in 1h 24m 58s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d59c1cfcdc21444180c5bafbff6ef7ea : SUCCESS in 1h 33m 47s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/518289b8cf2249c49c0b8767f665d32b : SUCCESS in 1h 33m 47s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/236445bbbc1d47668232e87344315ad7 : SUCCESS in 1h 11m 42s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8e24d1df3a7f4f46b757baca42b70242 : SUCCESS in 1h 15m 14s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ef9109b0a983475b8cd2a641c40def6c : SUCCESS in 1h 12m 32s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/151c5bec17e54c80ada7d47f1dfd466b : SUCCESS in 1h 24m 35s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e1398c92a6b4473eaa1069ac1027150b : SUCCESS in 1h 24m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1622bd57feb44542ac1eddf8636913f7 : SUCCESS in 3m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9b1fab3900c449b7b644ed6c60254930 : SUCCESS in 2m 13s","accounts_in_message":[],"_revision_number":3},{"id":"86da647dd277f04349f42034bd25595335bd0504","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-10-03 09:41:14.000000000","message":"Patch Set 4: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"c4382957609e8c2a8f3e9cd68bc1db5ac7b47564","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 11:34:25.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/34047a0fab9c4c3aa76da2967a070446\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bf048998dde24de5ad56c0e17cda4eb5 : SUCCESS in 6m 31s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/369332f8656241ef91733b07f4316e41 : SUCCESS in 1h 19m 16s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/dacfc2a94f624c3496ef596fa6df5632 : SUCCESS in 53m 37s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/299707450ea54a2093b271cf5387e61f : SUCCESS in 1h 22m 48s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/74cb2c8150004e98b97e35f45d5c77b8 : SUCCESS in 1h 45m 41s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9af846c662b14469a9598c8119215b76 : SUCCESS in 1h 28m 40s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/bc0b271c34e449ba8e8fd1b2da826790 : SUCCESS in 1h 41m 05s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b32cc7af9cdf4558b6ac62f1bbc7dbc7 : SUCCESS in 1h 25m 48s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a45e1c3f8704481bb4e8db5f2a3fb2c2 : FAILURE in 13m 05s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bde3d1a94a5649c9837d1ecd94112371 : FAILURE in 19m 54s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5e5aad87405d414c8dbaf6d930c1d51b : SUCCESS in 1h 04m 42s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0b00839f2a3e44aebdba672acada3faf : SUCCESS in 1h 02m 07s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/49984732583a4ceba4843f60aa5a86cb : SUCCESS in 1h 21m 18s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f7edf535429b42e4be935beedbb734a3 : SUCCESS in 1h 27m 13s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/eebcb68525c9489bbec5243a2cd8ce9f : SUCCESS in 58m 07s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/3bdace60cd0a4bd0bf59cc89daa9ce79 : SUCCESS in 1h 34m 52s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/29e4ea8dfcff4269980d5a651924e93a : SUCCESS in 1h 18m 18s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/aae98cb710e14d8ea375b5028b9964bf : SUCCESS in 1h 39m 05s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/10b7e15faccb4ccc9e760018a1494fba : SUCCESS in 1h 09m 31s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dc2bc106079545eead68ab84d3cfa482 : SUCCESS in 1h 37m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2dd9ca4af1d1451998acc83378fe7f96 : SUCCESS in 4m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ad72198c4a0a497d8db0a4f3ee4b2acd : SUCCESS in 2m 31s","accounts_in_message":[],"_revision_number":4},{"id":"cc36671701174dc3a3dab0d706bbea5dec86fdae","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-10-03 15:19:19.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e8e478c0367cacdc3f005aa3e5edeb153d55afde","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 18:34:19.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b64af812686f4d73a567b5b2735e0f92\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6717468b1e494b74a94ca180fe1af112 : SUCCESS in 6m 12s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/5fcce84706804d55a20294d66bdd6362 : SUCCESS in 1h 26m 12s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/dbb738fdb2cc445d8434d1e9e658ec5b : FAILURE in 1h 29m 48s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/484ed40fd19f4694ab24b7c9840ed53a : SUCCESS in 1h 38m 12s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cec18466cbf94addac5c7991494382d0 : SUCCESS in 1h 59m 34s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8d0db5b91b3a44c2bc56b4a4fcd7833e : SUCCESS in 1h 23m 38s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a0e0927f9cb545728454ab84efc98f9b : SUCCESS in 1h 33m 02s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e2d261ab8b13435596003c14cdbc4dee : SUCCESS in 1h 23m 55s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/46ad7203c7034420bba46c12d7d53c32 : FAILURE in 18m 05s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2c8fc605462748f09eca2d4c382dffb5 : FAILURE in 19m 44s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3fc20ba5cbe54d6387fc33b84b6260f8 : SUCCESS in 52m 55s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9f87bcdf6f45465c81ae5375cc7ecd07 : SUCCESS in 1h 05m 03s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/0fa570f337bf47c0852096d9777369a5 : SUCCESS in 1h 26m 14s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c6f003ef252e4bd6843ef337eb67e75d : SUCCESS in 1h 27m 08s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b7b76cb6f3d7470cb6e3e5cdd8043fc4 : SUCCESS in 1h 30m 21s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/11e72e20f155443fa8151bfee6efa964 : SUCCESS in 1h 40m 26s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c8769882c9524c83b3acc4a927fe7af7 : SUCCESS in 1h 02m 19s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5b0926632bb340b39d92d948cc73faef : SUCCESS in 1h 29m 16s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/17c2e544edee44d6a64b5a42eb164076 : TIMED_OUT in 3h 03m 24s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6753f014a0614f84bf9a89201e34d611 : TIMED_OUT in 3h 03m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34bf16eff391421593a98fc63b576373 : SUCCESS in 4m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/06c1ae09c7064056874177724f675ad4 : SUCCESS in 3m 15s","accounts_in_message":[],"_revision_number":4},{"id":"2a62f431ea5da58d2b6934c174f590d9879145ca","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-10-04 07:48:29.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ca9ee23615139da2ac558b8d752da35f389d63d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-04 09:38:29.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d058ca3535a44793adc7a71236e9a9c7\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3e178250f3254fbfbe14d1033946eb0a : SUCCESS in 6m 42s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/cce3eab4e3094de99e4736c442294469 : SUCCESS in 1h 24m 41s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/4f06bb58c0e74cce89cd3cb45e4ba39c : SUCCESS in 1h 19m 12s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/463f6ffa8e2941af96419992b87e5ce3 : SUCCESS in 57m 15s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3b88b97b15394664969936f1e8675333 : SUCCESS in 1h 39m 12s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/47eb580939d8475187c3806e86eb3dc5 : SUCCESS in 1h 27m 40s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b25e092a3de5404ea2f0c6a9efdefd83 : SUCCESS in 1h 23m 12s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d181207c19c44f00a367d1da26c6e37d : SUCCESS in 51m 02s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3c8ff35fffd74052be0b0c1befd8829c : FAILURE in 19m 13s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/be5905ee4cea4e2aa1d629404c19abcc : FAILURE in 20m 46s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3feaa28878774c9599a51a0a5788e7d7 : SUCCESS in 59m 31s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7bf2ee032efb4ee294cc1be49ed6840b : SUCCESS in 1h 03m 19s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/503b0ad2822044819175da14b35dac22 : FAILURE in 1h 21m 51s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a91df14df12c4de79a20574cd163fef2 : SUCCESS in 1h 29m 48s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/11c878015e71451fa3f116366e3f7d13 : SUCCESS in 1h 32m 35s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/4a3c0e20a7e1452ca4eeb9c01b978838 : SUCCESS in 1h 10m 18s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9ca93c5fa1ea4a6e800e7d2e6ec918f2 : SUCCESS in 1h 13m 52s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5114421c4b1041b096f5bcb2d20ce203 : SUCCESS in 1h 33m 00s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8d6584354cbf4d80aa11f68c7a555b0c : SUCCESS in 1h 14m 51s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d9d5a831c9f3446282fad6ebeab2b254 : SUCCESS in 1h 38m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d0997d7470444d9bff13dc37486b28c : SUCCESS in 3m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2aefacbfdcdc4ff9b13104992df5238f : SUCCESS in 3m 35s","accounts_in_message":[],"_revision_number":4},{"id":"f565e7fbae45995c65099914024e6fdc1eeabf71","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-10-08 07:31:38.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"72f6b99ebeb85271c900612593b233efebd2f9e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-08 09:26:12.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e5890082d60a4e60bff7558e41836b8c\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2e0c9954eb694cd2a64e1f932f7ade70 : SUCCESS in 5m 50s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/167e07820eb34fb799ac818dfd8d07d2 : SUCCESS in 1h 26m 52s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/4e7ed0f91a974232aff42cd41f7cbe02 : SUCCESS in 1h 33m 47s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7a09e901993a4bf09bcbb59d05ecadc4 : SUCCESS in 1h 30m 15s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/49c0d0cc26394b70b9b5cda5f79c8ac4 : SUCCESS in 1h 33m 49s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7d3efc985eb64c2f8bd41da3233a3971 : SUCCESS in 1h 27m 57s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1b7ac49303f04f99ad8f8113106887ac : SUCCESS in 1h 34m 12s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/088a3fda90974d34ac3a00b17ba31036 : SUCCESS in 1h 52m 31s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1a25d1bc40ca44d0b4b5e9bb81d9c314 : FAILURE in 19m 08s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/88671eda6ae54c3e92f02063da550a4e : FAILURE in 15m 10s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e02efd63ed404e4f81533b14580813fe : SUCCESS in 54m 19s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/632a66afd3f941b19f78ddd82e1eee7d : SUCCESS in 1h 07m 04s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/8a3ccc164f3c4ef9a7f1a62c681f9113 : SUCCESS in 1h 18m 11s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/8bd6877893cf4ed49d11a318d09d71f5 : SUCCESS in 56m 46s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/20c4c9028ae145fe937852ef8a1c9db0 : SUCCESS in 1h 27m 34s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/99e359183a2f445ba11de2068e8560f2 : SUCCESS in 1h 10m 23s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bb645a47af614537a491c916534f3dde : SUCCESS in 1h 40m 55s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e77a27d86bd94cdf8f9ab020b303494f : SUCCESS in 1h 16m 27s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/345468a77d6241faba8d62fe5ad51900 : SUCCESS in 1h 36m 22s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a8ee38f9d08645849dbe5d6606aea15e : SUCCESS in 1h 10m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/365466dacd5d4b3e86c935444aee9553 : SUCCESS in 3m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d9a7da39554646d29f9c7bbb9d3b0f8f : SUCCESS in 2m 14s","accounts_in_message":[],"_revision_number":4},{"id":"a414962160e7c8748d7db3c28956573209cde6a5","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-11-19 12:22:24.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"340fb19dd217953c427505592e6f7b7778163a94","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-11-19 12:22:51.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"11094a6922356038f83d8cfdd9cefc11f0a45e3a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 14:19:03.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6ad5b3ba50f84a058af47dc60020cb18\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b5e82e40e1424085989f25c488f455ef : SUCCESS in 6m 19s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/de2d3e11cb77440383385bacc1c2912c : SUCCESS in 1h 14m 44s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/9e94a3ca102546dc9f45f83bcd53945f : SUCCESS in 1h 39m 37s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1c1bc1728eec4843abc367fea4fe3be2 : SUCCESS in 1h 24m 03s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/453f089131fe4f79b48edc16dcd39ae1 : SUCCESS in 1h 42m 10s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d34a13dd942f4d1ea56b813abf8d8093 : SUCCESS in 1h 27m 47s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7d632b04a3ae45aa9bcca0a81ec7bf04 : SUCCESS in 1h 49m 02s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0b6d47a8c90c4aa68571ba24ae2f30fd : SUCCESS in 1h 39m 09s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/309f9347c94a499dbd0b25c461a02888 : SUCCESS in 57m 54s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dc00297a93f74800a736fff4e9f2cd35 : SUCCESS in 1h 01m 54s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/cf30aaa8746a4f43918ce654dba233cb : SUCCESS in 1h 06m 10s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c608970de9744dd5b7f98f8dabf6848c : SUCCESS in 1h 27m 13s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/435c727bab0a4521b5abb33209f2b40d : SUCCESS in 1h 27m 21s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/87a4c1eed87943268690a4715ce08de2 : SUCCESS in 1h 00m 13s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5f3711117b4f44d8b509f33fc29f7f73 : SUCCESS in 1h 13m 49s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9b870a7716424f139b2cbad9c7953aa9 : SUCCESS in 56m 44s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/600ae6c11f204c2a817a94e1b2dc0d89 : SUCCESS in 1h 14m 37s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1589a915342144fab559c62205206a09 : SUCCESS in 1h 12m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/475d39f81fb54af89182653bdb22ac99 : SUCCESS in 4m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8c8fcae6ac2e4bfe925e08238ca09e6e : SUCCESS in 2m 49s","accounts_in_message":[],"_revision_number":4},{"id":"14ddffd19591483399f521f6fd899395f2033874","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2024-11-20 08:34:58.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"4dae04266b534b90bf29f4a2c54a2ac5fe9f7a21","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 08:37:04.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"459b70a8efc904a621ac9f83db6de04630f0f1dc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 11:01:43.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/78ce51e20e5d45758121cc60ca4698a8\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e0343d51d4564b77913edd6bc80722eb : SUCCESS in 6m 26s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/eb137ca247324e109a041f77cd49a781 : SUCCESS in 1h 17m 47s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1b467b7895d542218accf5e0c1cf6d5e : SUCCESS in 1h 25m 16s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fc201db717d943ada8ea537ed384087d : SUCCESS in 1h 08m 26s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/af7ad261ea72418fa3a384160a496356 : SUCCESS in 1h 19m 03s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c927b5bb363c42118943726c939722d2 : SUCCESS in 1h 25m 48s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/fbba44e6bcd847d8b63df2d79a74b87d : FAILURE in 57m 19s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1224d0c5eebd4204ab98b605dfce1d79 : FAILURE in 53m 38s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/3b0e521a119b4c32b73d22cb59aef9df : SUCCESS in 1h 06m 36s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cf321b15152f4306b48e435bf892b790 : SUCCESS in 1h 19m 52s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3979a667d9f2408dab01204405d1502e : SUCCESS in 1h 23m 11s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8a708e20bbba484c9dcd10288c3f44c6 : SUCCESS in 1h 09m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e37753c9dd44c43a5894b17b6c86cf2 : SUCCESS in 4m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e2c1078e9dd64ed8a9ff609da0d4c3a1 : SUCCESS in 3m 35s","accounts_in_message":[],"_revision_number":4},{"id":"9daf576c9c7b36b5d5fd8ff15002b063e787ae32","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2024-11-20 15:49:44.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"6b00d0356a3d6f95cedc51da05b89cd78a71e993","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 17:30:51.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f836f0ecb4a54bc991656a480f13cd5d\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ea7057b03bc144ae81c54b90b01f8719 : SUCCESS in 6m 07s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/f810823a2e4447dfb2dd158458ad05a7 : SUCCESS in 1h 20m 17s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/389da1b4d285451297ef587c517acd6a : SUCCESS in 50m 28s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9dc41c5fe442420bbf6698547db3cf33 : SUCCESS in 1h 24m 10s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0ee50f7b233c4a66b9b4b2a42bedb636 : SUCCESS in 1h 29m 39s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d851845ac1e041da8b9b6320b56913e2 : SUCCESS in 1h 29m 24s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/542c3a97af6f47a18514d5e93830ee46 : SUCCESS in 1h 33m 55s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5f5df14cbb1840209a3b81e2b83662d7 : SUCCESS in 1h 32m 21s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ec70492edb454b00bc94a7be07125763 : SUCCESS in 55m 46s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3f9d00ac76974392a038d4ba2cdbce53 : SUCCESS in 46m 37s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/f6796519b16c47c796995103d7c02382 : SUCCESS in 1h 24m 59s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3a60c300456c441d9bfe43dbf0b05e33 : FAILURE in 1h 05m 51s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/8076b8f9615846da8a48ba943fa42855 : FAILURE in 54m 38s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/3d7e0895104c43dba39621b215bf0412 : SUCCESS in 1h 06m 29s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5a4a6b78fac4481aaf7ca799c9cb6e08 : SUCCESS in 1h 04m 18s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0a61953b16b248498798e654f96d777e : SUCCESS in 1h 07m 52s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ff981097e8a946c98297a7fc69376a2f : SUCCESS in 1h 05m 34s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/681e4c020a18496a8ee52330a9535f87 : SUCCESS in 1h 22m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b76b0298ec34efd8851fc32473bbeca : SUCCESS in 4m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8daa197971d041cf842df013e1754215 : SUCCESS in 2m 40s","accounts_in_message":[],"_revision_number":4},{"id":"feddb71d20b59fbaff638fe746a3f0e9c9189284","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-11-20 17:31:56.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"f2e9aeff74c2fb70d2da9768c2786789b1b8ca5b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 19:31:56.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6f9ce6f23ae04fe88a263f670063d0bf\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ccb689799f5a4d0b9a3b5940d4eccfb3 : SUCCESS in 5m 51s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/f4dd1afec38d4699bbd7043921f4188a : SUCCESS in 1h 22m 43s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/9c9b80ba43c34fb189e03bc788c57f33 : SUCCESS in 1h 42m 52s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d69b6036c7c14168bea54453adf71ed5 : SUCCESS in 1h 34m 38s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/11f643b8a8a546548305192c03668a1a : SUCCESS in 1h 39m 33s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f6a3d9901a2e4c32b93ddb8face69a89 : SUCCESS in 47m 25s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d4d78e50a17745c4934ed22b5bfd259b : SUCCESS in 1h 46m 25s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/168d5e48eb0c484584184f91281f2b7a : SUCCESS in 1h 33m 43s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5a115a8ccd89423da13ff854fca25676 : SUCCESS in 36m 36s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fd84030c2d8d47f0a602f712f39dffe3 : SUCCESS in 59m 10s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/6afb67b83e3e4cfda3e749b19be92a08 : SUCCESS in 1h 25m 29s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f67a215f46464f2a830cf5785dd22026 : FAILURE in 38m 48s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/26c73dcc2bb64e169703d0fff1381c90 : FAILURE in 1h 03m 05s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/6bad362845464c7da448bc6cc9ad3996 : SUCCESS in 1h 12m 54s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/17841148727a4741b575f60cdafa427a : SUCCESS in 39m 56s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/80e4ef0d41a04930afa5dd073df51eb1 : SUCCESS in 1h 11m 24s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6069c320b7f74978849db15e1fd9a296 : SUCCESS in 1h 20m 18s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eb726a0d1f82497ab5c236ea1c206c5a : SUCCESS in 45m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ef6431c586c42f6a85d35f166be98fa : SUCCESS in 3m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/96c33506c8b14cd1b8d19ae85e30d32d : SUCCESS in 2m 45s","accounts_in_message":[],"_revision_number":4},{"id":"e9302aff85c7eba5e0236f22aed92e1268c951b9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-11-26 11:18:25.000000000","message":"Patch Set 5: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":5},{"id":"21450cdff6d260d2e4ef3a4994e51a8084182a4b","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-11-26 11:18:30.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"b37d8a9fa0f8399e035063edbab8bb1041951950","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-26 13:04:51.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0680f5ee93744da6945e740ec00bf92b\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/83e469bb8ff74849b7562c7754e0d93d : SUCCESS in 7m 38s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/9e6e86c1bc084482b2a67f4848608087 : SUCCESS in 1h 27m 43s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/04b89b39dccd4627bc84bc8200ec2d1f : SUCCESS in 1h 26m 05s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/49137b39932b4b06a564cfe4ea8a35e0 : SUCCESS in 1h 30m 14s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cbfbfa2f5eda428aa0dd644781cee0f6 : SUCCESS in 1h 35m 11s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8cd4b23fa46144b59bd6c47cedf53827 : SUCCESS in 1h 21m 52s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2d4668f26f124be89498f8ef39674587 : SUCCESS in 1h 40m 34s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5b47760167d54880a303aca3ac97b86e : SUCCESS in 1h 27m 43s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ed859656b77045a08f90a4f63326d231 : SUCCESS in 1h 10m 09s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b614844827a94c3a99986cd7c2c7ea15 : SUCCESS in 59m 26s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/4a4ab2e9008048ad98f04fb8acec865a : SUCCESS in 1h 22m 49s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f158b6c14353446f8cf08dcb0167efaa : SUCCESS in 1h 34m 06s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/32593bdae24646dcb29d10f6d7a3d7a5 : SUCCESS in 1h 12m 39s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/3e85af2078e94350a81f73b513880685 : SUCCESS in 1h 16m 16s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0988d106ee4c4027b21f40c7de6c779e : SUCCESS in 1h 08m 51s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ee34c3f23b2a4405916b3d548b56133c : SUCCESS in 1h 19m 37s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7944881de910427189d1bcce22efb2b9 : SUCCESS in 1h 07m 53s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bdd75ed09e544bd9af184d6644cc4d34 : SUCCESS in 1h 10m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2cb2d774def040bfbe881379c6fe3b39 : SUCCESS in 4m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/966d95d56ea24ead838599b7c1160fe0 : SUCCESS in 2m 46s","accounts_in_message":[],"_revision_number":5},{"id":"523d389e2615bcdb456d37ec66cadcb485e8bbf7","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2024-11-26 14:09:43.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"184a5e53167ad73b50b7ff37e6d0508ed236c2b8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-26 14:11:56.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"66bdae3c58cea9a5e6da654529e08bd24d1a34b8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-26 16:04:07.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b7bad40e509492b9a5e3587a646d124\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7fb5cdcd06f04c1ab4fa4a99b38ec456 : SUCCESS in 6m 38s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/b927cd99570142618801adfe077ec9a4 : SUCCESS in 1h 15m 49s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/efcae36784cb4130b00048412201bc4e : SUCCESS in 1h 28m 34s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/60de69a411d24ac2950d2896d159dd03 : SUCCESS in 1h 16m 37s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c7de11cd58a44c77bf711cf48247bc46 : SUCCESS in 1h 46m 23s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/45001e88e7e34583b57287b8c7f9a37f : SUCCESS in 1h 38m 05s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2be2c25576ff43f19629d454f387b4e2 : SUCCESS in 1h 21m 21s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/fbf0ceaa17df4e69ba5d1f08bd1c2629 : SUCCESS in 1h 27m 09s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/e49f0ff72d8549b5a9148c2f9fc5626c : SUCCESS in 1h 07m 46s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e50d2377868d43ddb79580fff3a5dc13 : SUCCESS in 46m 00s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ac40506b4f754f82932cb16ba8ceff03 : SUCCESS in 1h 17m 23s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/04dc92de254d4d36b1225f21089b1c11 : SUCCESS in 1h 12m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f544e4692f7e43e6b138183af1ddffa0 : SUCCESS in 4m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/72b5595dac0a47faa19e60a6ca0ee4bb : SUCCESS in 2m 09s","accounts_in_message":[],"_revision_number":5},{"id":"54d7b76f2bccbf4a8e44d6572d76d4f110acc49d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-26 16:04:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"2ab4a50712cca7f9711ffbdef6ab5036cfb798c1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-26 16:05:14.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/054e7dfe773e40e0b73f8e45c2929d49\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68951950dc614f5db1bb3c6cc70bb644 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"32a5098302a26b735a178f8207126b6a9fa4871e","revisions":{"ecf1d91d485a352208748cb7b1800f7201dd16cc":{"kind":"REWORK","_number":1,"created":"2024-10-02 15:13:14.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/92/931192/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/92/931192/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/92/931192/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/92/931192/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/92/931192/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/92/931192/1"}}},"commit":{"parents":[{"commit":"2e2096a562dfac60c64d28662c0ffe50202a0f78","subject":"Do not install uWSGI as part of Keystone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/2e2096a562dfac60c64d28662c0ffe50202a0f78"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-05-14 11:41:30.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2024-10-02 15:13:14.000000000","tz":0},"subject":"Change example to contain domain name instead of UUID","message":"Change example to contain domain name instead of UUID\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-openstack_hosts/+/930272\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-rabbitmq_server/+/930446\nRelated-Bug: #2065680\nCloses-Bug: #2064718\nChange-Id: I5d9505ed4b385673cc719e9821ee2e1c2ba5c754\n(cherry picked from commit c3aa3c36233b79b3391e95a4ce100bde3e667391)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/ecf1d91d485a352208748cb7b1800f7201dd16cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/ecf1d91d485a352208748cb7b1800f7201dd16cc"}]},"branch":"refs/heads/stable/2023.1"},"af8064d4412fe6c621483fe38a23e7ed05902c9d":{"kind":"REWORK","_number":2,"created":"2024-10-02 15:18:21.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/92/931192/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/92/931192/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/92/931192/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/92/931192/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/92/931192/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/92/931192/2"}}},"commit":{"parents":[{"commit":"2e2096a562dfac60c64d28662c0ffe50202a0f78","subject":"Do not install uWSGI as part of Keystone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/2e2096a562dfac60c64d28662c0ffe50202a0f78"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-05-14 11:41:30.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2024-10-02 15:18:21.000000000","tz":0},"subject":"Change example to contain domain name instead of UUID","message":"Change example to contain domain name instead of UUID\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-openstack_hosts/+/930272\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-rabbitmq_server/+/930446\nRelated-Bug: #2065680\nCloses-Bug: #2064718\nChange-Id: I5d9505ed4b385673cc719e9821ee2e1c2ba5c754\n(cherry picked from commit c3aa3c36233b79b3391e95a4ce100bde3e667391)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/af8064d4412fe6c621483fe38a23e7ed05902c9d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/af8064d4412fe6c621483fe38a23e7ed05902c9d"}]},"branch":"refs/heads/stable/2023.1"},"1edcfc14a498cd3cdcb84ae147713587fd8b0f6a":{"kind":"REWORK","_number":3,"created":"2024-10-02 15:18:33.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/92/931192/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/92/931192/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/92/931192/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/92/931192/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/92/931192/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/92/931192/3"}}},"commit":{"parents":[{"commit":"2e2096a562dfac60c64d28662c0ffe50202a0f78","subject":"Do not install uWSGI as part of Keystone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/2e2096a562dfac60c64d28662c0ffe50202a0f78"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-05-14 11:41:30.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2024-10-02 15:18:33.000000000","tz":0},"subject":"Change example to contain domain name instead of UUID","message":"Change example to contain domain name instead of UUID\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-openstack_hosts/+/930272\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-rabbitmq_server/+/930446\nRelated-Bug: #2065680\nCloses-Bug: #2064718\nChange-Id: I5d9505ed4b385673cc719e9821ee2e1c2ba5c754\n(cherry picked from commit c3aa3c36233b79b3391e95a4ce100bde3e667391)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/1edcfc14a498cd3cdcb84ae147713587fd8b0f6a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/1edcfc14a498cd3cdcb84ae147713587fd8b0f6a"}]},"branch":"refs/heads/stable/2023.1"},"b8d3ab18d6e325acc5e7d5312edb00e62dfd40bc":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2024-10-03 09:41:14.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/92/931192/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/92/931192/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/92/931192/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/92/931192/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/92/931192/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/92/931192/4"}}},"commit":{"parents":[{"commit":"2e2096a562dfac60c64d28662c0ffe50202a0f78","subject":"Do not install uWSGI as part of Keystone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/2e2096a562dfac60c64d28662c0ffe50202a0f78"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-05-14 11:41:30.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2024-10-03 09:41:14.000000000","tz":0},"subject":"Change example to contain domain name instead of UUID","message":"Change example to contain domain name instead of UUID\n\nRelated-Bug: #2065680\nCloses-Bug: #2064718\nChange-Id: I5d9505ed4b385673cc719e9821ee2e1c2ba5c754\n(cherry picked from commit c3aa3c36233b79b3391e95a4ce100bde3e667391)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/b8d3ab18d6e325acc5e7d5312edb00e62dfd40bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/b8d3ab18d6e325acc5e7d5312edb00e62dfd40bc"}]},"branch":"refs/heads/stable/2023.1","description":"Edit commit message"},"32a5098302a26b735a178f8207126b6a9fa4871e":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2024-11-26 11:18:25.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/92/931192/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/92/931192/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/92/931192/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/92/931192/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/92/931192/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/92/931192/5"}}},"commit":{"parents":[{"commit":"2e2096a562dfac60c64d28662c0ffe50202a0f78","subject":"Do not install uWSGI as part of Keystone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/2e2096a562dfac60c64d28662c0ffe50202a0f78"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-05-14 11:41:30.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2024-11-26 11:18:25.000000000","tz":0},"subject":"Change example to contain domain name instead of UUID","message":"Change example to contain domain name instead of UUID\n\nRelated-Bug: #2065680\nCloses-Bug: #2064718\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-openstack_hosts/+/936232\nChange-Id: I5d9505ed4b385673cc719e9821ee2e1c2ba5c754\n(cherry picked from commit c3aa3c36233b79b3391e95a4ce100bde3e667391)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/32a5098302a26b735a178f8207126b6a9fa4871e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/32a5098302a26b735a178f8207126b6a9fa4871e"}]},"branch":"refs/heads/stable/2023.1","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
