)]}'
{"id":"openstack%2Fopenstack-ansible-os_keystone~933572","triplet_id":"openstack%2Fopenstack-ansible-os_keystone~master~Icfcc1ccf7970853f480a922c40a8e0f5b7bbe4f2","project":"openstack/openstack-ansible-os_keystone","branch":"master","attention_set":{},"removed_from_attention_set":{"31542":{"account":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"last_update":"2024-11-08 23:48:20.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Icfcc1ccf7970853f480a922c40a8e0f5b7bbe4f2","subject":"federation: ensure cloud credentials are found on utility host","status":"MERGED","created":"2024-10-28 14:14:56.000000000","updated":"2024-11-08 23:49:20.000000000","submitted":"2024-11-08 23:48:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"933572","meta_rev_id":"6f34d1a648c142a4a03182272a505652b214e1c3","_number":933572,"virtual_id_number":933572,"owner":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-08 23:48:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2024-10-28 14:39:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-11-08 22:10:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-11-08 22:10:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2024-10-28 14:39:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-28 14:39:37.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2024-10-28 16:27:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-08 22:10:55.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"9db6cd59cb25c91b3f00f7fed9395b1f6f600437","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2024-10-28 14:14:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"87fc95ab6e860bb754f2470847dfdd9f5942ea7e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2024-10-28 14:15:47.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"7f2920e49d02aea0d0a6b5796e804c1abcce7b44","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-10-28 14:39:37.000000000","message":"Patch Set 2: Backport-Candidate+1 Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"3129b7bfb53fa373fcfed473fb26dc399952884a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-28 16:27:28.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/77cfb61720424443a187525a21bef610\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/413cdaf4ba32463b86dbfca735ee6866 : SUCCESS in 6m 23s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/89e4c2a10bd948bca1cd5c9ea6f95155 : SUCCESS in 1h 12m 15s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/3ef2e395627d4bb59565f9302195285a : SUCCESS in 1h 23m 23s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2f9471e7b9e0459c946383ba30f9f2b3 : SUCCESS in 1h 17m 26s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c2d89d4a7dc34da1a4825f9bcb03ae4b : SUCCESS in 1h 04m 20s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0ebfb3847a6d427a802b0a1045c35959 : SUCCESS in 51m 51s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3f430dae88444e7e9ba940ccc7243159 : SUCCESS in 1h 25m 58s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a2c6932047c84cf09f839a89f097cdc5 : SUCCESS in 1h 20m 22s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/15a11ac207dc4815b64a80329b796a29 : SUCCESS in 1h 16m 07s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2a3a1a710bbe4ea4a3fbc560e1f71c4a : SUCCESS in 1h 31m 43s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4214d7323a0d46b890c46b6211360e0f : FAILURE in 1h 41m 46s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/285af6e138ee4f5cb2d09c43a0168a71 : SUCCESS in 2h 01m 15s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f5456340d96f450d8ea5f35ce337c3f9 : SUCCESS in 1h 56m 17s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/12530c1d910649d1be44c6d5bc6b9985 : SUCCESS in 2h 03m 53s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/83cfe3bf762045a9918153bdd0d9f701 : SUCCESS in 1h 47m 59s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/14c70b4370ab4307aa2baef97bfcf766 : SUCCESS in 1h 05m 23s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/940894b913364b3b9d58bfdd4d6e1d97 : SUCCESS in 56m 46s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/c1d3bc0b91144b61b5d5b4fc129dbf65 : SUCCESS in 1h 28m 30s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5a42c82afc2c40dda4d3c5ac748776f0 : SUCCESS in 1h 18m 43s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/831f1c4236cb4ca3b9a1497da0a6fce8 : SUCCESS in 1h 19m 50s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/bfd5b1cc76a94b2fb045a36f2d5c6a51 : SUCCESS in 1h 07m 49s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c9ef91d0370544d7a62ecf676abc8136 : SUCCESS in 1h 02m 29s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7a57c57ad9e6480083b6b3cd44144c3a : SUCCESS in 1h 12m 05s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6f7e741fc7c843a1ba109517ea67f898 : SUCCESS in 42m 46s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c5c1c22856984cac8437b8bb22e0d42d : SUCCESS in 1h 10m 22s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d2d40be8a6184d8fb3d8179c5c09e200 : SUCCESS in 1h 07m 47s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/26d818e3feab4b33887e64d5ffb0a45e : SUCCESS in 45m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49005221ac104d4ea1c7c4dc51fd4656 : SUCCESS in 4m 03s","accounts_in_message":[],"_revision_number":2},{"id":"e47549c527e415e5bec5d347a703f619b308997d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-11-04 10:44:22.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1c646f3d802d9e59c4c08ea19958f09c184efb07","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-04 13:20:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/30d1c721d5ea46889199305d48a3f8ac\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/155b16251f104987907da566b7c621d1 : SUCCESS in 6m 06s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/5a1bf1fa6129489c9dd654373e195be8 : SUCCESS in 1h 14m 42s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/4c46424a998f482aa4fe07f4e3d76042 : SUCCESS in 1h 18m 02s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/cefc589b9a3d489ab9c94c20928d3b7d : SUCCESS in 1h 22m 30s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fb77f298e408421fb904cd347c36d64e : SUCCESS in 1h 18m 07s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6b25d64d75464a72a04ef44c26d4435b : SUCCESS in 1h 25m 12s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6173a6dfc768488dae12295a24c32073 : SUCCESS in 1h 20m 14s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dcf64b639434482bafcc3c03f5ddf1a0 : SUCCESS in 1h 13m 35s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0ac4da418ad74295b2fd63283a07066f : SUCCESS in 1h 09m 20s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/26eeef6805a34a9183f7c8ff2b4a2877 : SUCCESS in 1h 18m 06s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c0d94246524940d890266a9f2835dd0a : SUCCESS in 1h 16m 26s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/199582ef319f4963aa3152b4b90d748c : SUCCESS in 2h 09m 25s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3425b1b36d064ea9ad17b35a9d7ab933 : SUCCESS in 1h 48m 00s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6b86b0d392624e468be5f563b26bebc3 : SUCCESS in 1h 52m 51s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/556bda928395443a93033004158fcc53 : SUCCESS in 1h 56m 44s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d67f25fed53d4915b1d6ac97bf751ddf : SUCCESS in 1h 01m 09s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/998b1244a32c4581aa59f9e4052eee73 : SUCCESS in 55m 02s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/b0d4615c09984325bc18706a442b41c3 : SUCCESS in 1h 23m 18s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e5d60c37a8c44c88b6d6d35a0b0843a9 : SUCCESS in 1h 16m 12s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ec06a4532a044ab29548233963e7f578 : SUCCESS in 1h 27m 12s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/fbc478cf87d94f02a1c883d737ae0efd : SUCCESS in 1h 12m 42s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ab63b642ee4a4c00be734e28279bb413 : SUCCESS in 1h 12m 44s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d02643f70e6f404f8fd27534c36d122f : SUCCESS in 1h 12m 29s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c6d52d313ac842a882797b04793d1a2e : SUCCESS in 1h 12m 12s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b677729ee8ff4f8a9dda95cb003afeb5 : SUCCESS in 1h 16m 45s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f827c314c886431885320fe2e6ef830e : SUCCESS in 1h 18m 55s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4a53997cc259416fadd6a124dce6af28 : SUCCESS in 1h 16m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/abcf216562414379919457789c9d4d15 : SUCCESS in 3m 54s","accounts_in_message":[],"_revision_number":2},{"id":"cace1ddc4a0e42bf83e3f34a7909f21346b660a3","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2024-11-08 22:10:55.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e9f4825d2099c53c41129935f53bc97ba96ea26a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-08 22:13:06.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a90c91cfc4ea891b59ec5ca673cfaa89f27ae2db","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-08 23:48:20.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b85ba2bfc4a048beb0a2213b5908ca24\n\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/d3f15e3d498b458fa13265dc5a118256 : SUCCESS in 1h 12m 32s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e89260f36ea948a48898ccf30eb5c99f : SUCCESS in 1h 16m 27s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/72cfb8cfc45f439d982f4223f7df447f : SUCCESS in 1h 14m 09s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/606187d7fcfe41e09bd2aa9d1ae61896 : SUCCESS in 1h 12m 24s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4404395601c14abf9c0b392fbd03cf88 : SUCCESS in 1h 16m 42s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-noble https://zuul.opendev.org/t/openstack/build/745ba1f8d5e6400ab19c9ffe0bb7cca5 : SUCCESS in 50m 28s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c706bfe1fb9d40348dfb24651aba865d : SUCCESS in 1h 22m 21s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7bf8c9688af64b6f9ff088306e961e34 : SUCCESS in 51m 28s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/67607552e68743c5bb1b89c9012c6052 : SUCCESS in 1h 17m 24s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/32d477f2eb9b4ba190a1c7929dbcf66a : SUCCESS in 1h 22m 04s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6238578cbe444c6793788f2b7b7d5dee : SUCCESS in 1h 28m 34s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/6fe2aab0a4144606ae974ddbc1e7a2e0 : SUCCESS in 51m 06s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e9241f4e461d4fbc9384603da2bbb051 : SUCCESS in 1h 10m 01s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/379ec83e1a7443cd9e25c763d82547c0 : SUCCESS in 52m 04s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9bddee40d52a49df85ad39dd23928b6b : SUCCESS in 1h 15m 32s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/701bb64864834996aa856ff0945fa3ef : SUCCESS in 1h 25m 18s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/58600c139a514c4e8838b2cf02cf95ed : SUCCESS in 49m 23s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dcb2150a77f044d7891a5f19f7046037 : SUCCESS in 53m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ebccb58200024e4281a94bfef1812aab : SUCCESS in 4m 00s","accounts_in_message":[],"_revision_number":2},{"id":"99b5c0af8a718f5b166ce69b174e7bbd1d8f81f1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-08 23:48:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"6f34d1a648c142a4a03182272a505652b214e1c3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-08 23:49:20.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6fd0f5745736428087201098aaabad1c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1fbfa0b8517147c0bd50bc1e8b2285a2 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"953b1ac80ff0d613d7c6e3cb6d832af69ef79612","revisions":{"807e1eb80acdece35d6beb2588eaf2f29d9de22e":{"kind":"REWORK","_number":1,"created":"2024-10-28 14:14:56.000000000","uploader":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"ref":"refs/changes/72/933572/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/72/933572/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/72/933572/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/72/933572/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/72/933572/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/72/933572/1"}}},"commit":{"parents":[{"commit":"199ed67d8131f2f7c889005b035607e75bd2c2e0","subject":"Merge \"Remove db_sync --migrate command\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/199ed67d8131f2f7c889005b035607e75bd2c2e0"}]}],"author":{"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","date":"2024-10-28 14:12:21.000000000","tz":0},"committer":{"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","date":"2024-10-28 14:12:21.000000000","tz":0},"subject":"federation: ensure cloud credentials are found on utility host","message":"federation: ensure cloud credentials are found on utility host\n\nChange-Id: Icfcc1ccf7970853f480a922c40a8e0f5b7bbe4f2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/807e1eb80acdece35d6beb2588eaf2f29d9de22e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/807e1eb80acdece35d6beb2588eaf2f29d9de22e"}]},"branch":"refs/heads/master"},"953b1ac80ff0d613d7c6e3cb6d832af69ef79612":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-10-28 14:15:47.000000000","uploader":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"ref":"refs/changes/72/933572/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/72/933572/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/72/933572/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/72/933572/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/72/933572/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/72/933572/2"}}},"commit":{"parents":[{"commit":"199ed67d8131f2f7c889005b035607e75bd2c2e0","subject":"Merge \"Remove db_sync --migrate command\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/199ed67d8131f2f7c889005b035607e75bd2c2e0"}]}],"author":{"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","date":"2024-10-28 14:12:21.000000000","tz":0},"committer":{"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","date":"2024-10-28 14:15:02.000000000","tz":0},"subject":"federation: ensure cloud credentials are found on utility host","message":"federation: ensure cloud credentials are found on utility host\n\nDuring upgrade, we found that these tasks were asking for an\nauth_url, indicating the credentials could not be found.\n\nFixing these tasks to match the ones above resolved this issue.\n\nChange-Id: Icfcc1ccf7970853f480a922c40a8e0f5b7bbe4f2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/953b1ac80ff0d613d7c6e3cb6d832af69ef79612"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/953b1ac80ff0d613d7c6e3cb6d832af69ef79612"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
