)]}'
{"id":"openstack%2Fopenstack-ansible-os_keystone~934533","triplet_id":"openstack%2Fopenstack-ansible-os_keystone~stable%2F2024.1~Icfcc1ccf7970853f480a922c40a8e0f5b7bbe4f2","project":"openstack/openstack-ansible-os_keystone","branch":"stable/2024.1","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2024-11-18 10:01:04.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Icfcc1ccf7970853f480a922c40a8e0f5b7bbe4f2","subject":"federation: ensure cloud credentials are found on utility host","status":"MERGED","created":"2024-11-09 14:01:09.000000000","updated":"2024-11-18 10:02:12.000000000","submitted":"2024-11-18 10:01:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"934533","cherry_pick_of_change":933572,"cherry_pick_of_patch_set":2,"meta_rev_id":"c555b42be026bf42a1779b5e985edd61322e6db1","_number":934533,"virtual_id_number":934533,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-18 10:01:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2024-11-16 20:48:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-11-18 08:22:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-11-18 08:22:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}],"CC":[{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-09 14:01:09.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"CC"},{"updated":"2024-11-09 15:56:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-12 16:23:41.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"CC"},{"updated":"2024-11-18 08:22:26.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"2de617d25bf85a5cb919771e652042a9f1bef063","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-11-09 14:01:09.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"805c7526f41fbee0a9f44db5558c47b823402a4a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-09 15:56:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0f51a7c5d95b46d6a6d5a80f77000577\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1b377e1b232c46c7a750a6be99a64532 : SUCCESS in 6m 02s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/f9938f5fd3c046dcb920718010bcffcd : SUCCESS in 1h 17m 17s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/ea2f66e331f74d76b3c9264e06c4a32a : SUCCESS in 42m 55s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/bcb61aeccba14b3a805f9e75fd249936 : SUCCESS in 1h 06m 07s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/aecad045893340bbb873b1e2005e53a4 : SUCCESS in 1h 28m 39s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c30d41d09c2b43e08d5a4d732d83297d : SUCCESS in 1h 02m 27s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/fe45d502bef64c5f80d29c076bc15201 : SUCCESS in 46m 49s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7b2dfe72681d4b30b82f0ca3553e622e : SUCCESS in 1h 25m 43s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7e7084aa2c8847c6acfe2bdf5ccdc05c : SUCCESS in 1h 24m 27s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3d35259272ec4137895e5e8fcd5e2c76 : SUCCESS in 1h 35m 23s\n- openstack-ansible-upgrade_2023.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/122503116ec842c5a3fd28c45d9d2684 : SUCCESS in 1h 52m 41s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d0a13f6d869d4d16ab4aae5171d663df : SUCCESS in 1h 26m 47s\n- openstack-ansible-upgrade_2023.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a59fc773009e497e89b722f64a3c88f9 : FAILURE in 1h 42m 04s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/cff3c5f01e2b45a4b845d11cfe678770 : SUCCESS in 1h 35m 41s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/055950f86e3a427b961df8a04093cadd : FAILURE in 11m 21s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0c3c319f9cbb4acbb11adaa19da707cf : SUCCESS in 39m 09s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d73702eb74f94ed591e0566aebe245ea : SUCCESS in 48m 16s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/4e91f8c49d4d412e82594df035f81f3f : SUCCESS in 1h 19m 48s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5f5fbaf5c8b741c5bc20e1ce3b352d5e : SUCCESS in 1h 28m 48s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/16d8a52c98f4439d8f38750cd750d444 : SUCCESS in 1h 05m 47s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/f9745c39ab1646b1a4c777bc19fc4144 : SUCCESS in 52m 28s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/8f13749a8a944d56ba57e870b5d10a65 : SUCCESS in 45m 50s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/67a4607683274758ac3956cf5c3d75e4 : SUCCESS in 1h 08m 48s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c57d60a27ca148c082ffc7996fed50b8 : SUCCESS in 57m 09s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cec5a99b338d4eee8888dda7ba2270a4 : SUCCESS in 1h 01m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5edeb70787f4461283cec85078d3f43d : SUCCESS in 2m 53s","accounts_in_message":[],"_revision_number":1},{"id":"e132cfba83f13d17ad6009a2485b61b69de4f2e5","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2024-11-12 16:23:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"057174a767e6c129d71f7f152c2e38854bb5f6e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-12 19:01:18.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/97f994c29b31450da8068191923403f9\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c133779a2e994d3c85e616f99e4e35fc : SUCCESS in 6m 25s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/66b2d5bc965d47dcae5b1b73b695fe57 : POST_FAILURE in 1h 13m 09s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/f1aab125ac8248839198892a1b0630eb : SUCCESS in 1h 23m 19s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/f53b1df9794144ada5330d85c926cd7b : SUCCESS in 1h 26m 32s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/cc79bf58c0b9431caae6b79e0c0d5efb : SUCCESS in 1h 22m 21s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/970d50c3b9c1405ea3b3957feb004c45 : SUCCESS in 1h 13m 49s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7a492c00538b4694acc90459d72ed06d : SUCCESS in 1h 31m 30s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/82b4ed0fbd554988b0cfeeb66cbaa409 : SUCCESS in 1h 13m 01s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1ae8d47072e247d191725b6a12d68337 : POST_FAILURE in 1h 14m 22s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/11dc6f8bbac24fcfaa3dacfe2912f19a : SUCCESS in 2h 13m 57s\n- openstack-ansible-upgrade_2023.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ec8fe2aaed3f4f53bd31527d2744851f : SUCCESS in 2h 21m 48s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1da6bb5d5ade4f438fee8de48d3a995c : POST_FAILURE in 1h 11m 59s\n- openstack-ansible-upgrade_2023.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eb17d4504eec4d56866acff166d065cd : SUCCESS in 2h 23m 32s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9a747918e93b4378b7ac8783eea83af9 : SUCCESS in 2h 18m 38s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e21ba88be5524fe595ed773f1056255a : FAILURE in 44m 29s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ab25119f1f5b41de8087012690b69846 : SUCCESS in 1h 04m 24s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8816827af9fa44a4a230cde9fb0df3e8 : SUCCESS in 53m 20s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/f4369b3f4f4d4f69b7f50b94f1bc4b10 : SUCCESS in 58m 04s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/51bebb8a58e54e518b5173fb5016f02d : POST_FAILURE in 1h 23m 21s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6f4229295a9547d9a4199ea01c009653 : SUCCESS in 1h 31m 16s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/1ce32a9b14ac471596d0cc0445c5743b : POST_FAILURE in 1h 02m 14s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/9d8985f50d6e4c22bf59e0eb053e44af : SUCCESS in 1h 22m 44s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/44d470fda2a546d09c2ba6bc76486587 : POST_FAILURE in 1h 06m 35s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/70ffdb90a9cc47ed928f28864b99049d : SUCCESS in 52m 33s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a67115c7f83242f99f53eb52b0d7e07c : SUCCESS in 1h 21m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e40e780235024878b0e3b08ca0c219c3 : SUCCESS in 3m 35s","accounts_in_message":[],"_revision_number":1},{"id":"27f283c51d620375ea0351c4c4518632556a5121","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2024-11-16 10:39:59.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"913ef1168dcb09e00c415d8e12e9e1228df53d66","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-16 13:24:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ba9ba74ce6464c43b768c111ddbba939\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e28f38d85c11409cb9aee7c8d2d567c7 : SUCCESS in 5m 46s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/1ce6bdc393e54791bb549289288c0efd : SUCCESS in 1h 01m 05s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/03b3ff2d9d034ea49e1f11b4338a5838 : SUCCESS in 1h 03m 12s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/a393efada93349b8b4a2d064d9f2678f : SUCCESS in 1h 06m 58s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c5553e213a19496d88797c3257505040 : SUCCESS in 1h 10m 54s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3ec0397665784931a0a81b1e70fbb8ad : SUCCESS in 1h 17m 57s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/03c3213227024b5187a729ea5a1d3dce : SUCCESS in 53m 12s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9c77438888504c319d71ad4d216fa4c3 : SUCCESS in 1h 04m 47s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/604a20e42755489b8287a2684860e646 : SUCCESS in 1h 02m 31s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d52e6c85faf74129aa23cd2ca5153374 : SUCCESS in 1h 37m 04s\n- openstack-ansible-upgrade_2023.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5cc083b24b75470dadc5628688739e62 : SUCCESS in 2h 37m 16s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/310d287833fc4355973e5e19ef2647bc : SUCCESS in 1h 04m 54s\n- openstack-ansible-upgrade_2023.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9d3ad292aa3044cbbcc996b94600d22c : SUCCESS in 2h 23m 17s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/351119416e8e457e88afe3dd3faaa018 : SUCCESS in 1h 39m 16s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/409d825eccc24b1184ff1d08c5cbb50a : SUCCESS in 1h 48m 39s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/be7759f9bdc94e58a79663fc8bced0e3 : SUCCESS in 39m 34s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6a949542e3d0422bbfbfe11ad0ef3750 : SUCCESS in 51m 30s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/ecc55f2715ce48daacaff08e383a7042 : SUCCESS in 1h 08m 32s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ef66f178b4364d82ad0e85fbbe37b286 : SUCCESS in 1h 22m 14s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3a090582a0054206a9b69bd55eaa70cd : SUCCESS in 1h 26m 00s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/787000717d6441a2a5170c81f68276ea : SUCCESS in 47m 13s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/8f564dae63b24fe6b4437360dd98a2cc : SUCCESS in 1h 13m 09s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3557b6257f0d4e229075ad8932f54d8a : SUCCESS in 1h 03m 22s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7b2fc108b1a547ff91f7ecff437e9004 : SUCCESS in 1h 11m 24s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/96876f53b39f4bd386577dffde3193e0 : SUCCESS in 49m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa1d552a94d54360a5c35cb4d1a0182c : SUCCESS in 4m 28s","accounts_in_message":[],"_revision_number":1},{"id":"1a29626eca47994ae5c551893e541b0540fed630","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-11-16 20:48:27.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"81484c6979e9cd22dcd8f5a84d279a30a5ba1a2a","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2024-11-18 08:22:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6417d8afd070ad00359f1df5d46afae59200268c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-18 08:24:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"36216fa54da036e5ec55a63325a15d209adf3bea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-18 10:01:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d412bc47c2f4fd5932477298e26eeb2\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c0a7206ca6d0470d9627f1e59ee4368b : SUCCESS in 6m 25s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/25627c1f20904a66b96a732486d0bf27 : SUCCESS in 1h 16m 38s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/0f5184215e4248aebf71f15514aaf07a : SUCCESS in 1h 15m 02s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f720ce51cc0045bcbba79eaaea222869 : SUCCESS in 1h 22m 03s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/97fb877e1de14fa798127a20aebec9d7 : SUCCESS in 1h 25m 45s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ffe14ec5fbde4112b5c1dd30d0b94f30 : SUCCESS in 1h 19m 44s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/72d4c9b11fa34cebbea031746c673a93 : SUCCESS in 1h 26m 22s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a81434bf9591401ebd5df15ba098e2fa : SUCCESS in 1h 15m 08s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a200479949a84b18b38245978a2ce92e : SUCCESS in 1h 19m 04s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1b1269d7f38d4909bd52d5f732bd40c2 : SUCCESS in 1h 26m 10s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/23de4d34747b43bdaf48bc3410c95f6d : SUCCESS in 46m 00s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/2ccae2f641484691a186166e6dfb2819 : SUCCESS in 1h 20m 03s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3868b6b23fd44f9f89af71c5a550b180 : SUCCESS in 1h 17m 22s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f4febcb0c3ff430da45271a03342aa30 : SUCCESS in 47m 18s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7149c68498fe45408760a3092f39f5ab : SUCCESS in 1h 12m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4fb6aca586e4ee99689f4c4277ebf3a : SUCCESS in 3m 41s","accounts_in_message":[],"_revision_number":1},{"id":"4182b79dac05fd809fdc3adf8f44cb500318c4dc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-18 10:01:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c555b42be026bf42a1779b5e985edd61322e6db1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-18 10:02:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/523510649e0b432d93e0a0f07f49c188\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b793d74a7764a4c8682593dd9286e22 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"54591057cf40c363826e612d8748ee3e85d2c2cc","revisions":{"54591057cf40c363826e612d8748ee3e85d2c2cc":{"kind":"REWORK","_number":1,"created":"2024-11-09 14:01:09.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/33/934533/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/33/934533/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/33/934533/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/33/934533/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/33/934533/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/33/934533/1"}}},"commit":{"parents":[{"commit":"81ac8f5ee2089cb28dd76d1f4cb264b4e2121b49","subject":"Change example to contain domain name instead of UUID","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/81ac8f5ee2089cb28dd76d1f4cb264b4e2121b49"}]}],"author":{"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","date":"2024-10-28 14:12:21.000000000","tz":0},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2024-11-09 14:01:09.000000000","tz":0},"subject":"federation: ensure cloud credentials are found on utility host","message":"federation: ensure cloud credentials are found on utility host\n\nDuring upgrade, we found that these tasks were asking for an\nauth_url, indicating the credentials could not be found.\n\nFixing these tasks to match the ones above resolved this issue.\n\nChange-Id: Icfcc1ccf7970853f480a922c40a8e0f5b7bbe4f2\n(cherry picked from commit 953b1ac80ff0d613d7c6e3cb6d832af69ef79612)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/54591057cf40c363826e612d8748ee3e85d2c2cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/54591057cf40c363826e612d8748ee3e85d2c2cc"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
