)]}'
{"id":"openstack%2Fopenstack-ansible-os_keystone~941364","triplet_id":"openstack%2Fopenstack-ansible-os_keystone~master~I433d6fe347e21098f563881f3c1fe494231a0b62","project":"openstack/openstack-ansible-os_keystone","branch":"master","topic":"osa/fqcn","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2025-02-28 08:36:28.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}},"hashtags":[],"change_id":"I433d6fe347e21098f563881f3c1fe494231a0b62","subject":"Auto-fix usage of modules via FQCN","status":"MERGED","created":"2025-02-12 11:53:37.000000000","updated":"2025-02-28 12:28:29.000000000","submitted":"2025-02-28 12:27:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"941364-osa/fqcn","meta_rev_id":"2d64edf9fc304099027ce0ec69ee2b70f76e2e63","_number":941364,"virtual_id_number":941364,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-28 12:27:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"all":[{"value":2,"date":"2025-02-25 12:38:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-02-27 08:31:37.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":2,"date":"2025-02-27 09:16:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":1,"date":"2025-02-27 09:16:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-12 14:11:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-25 12:38:06.000000000","updated_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"reviewer":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"state":"REVIEWER"},{"updated":"2025-02-27 08:31:37.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2025-02-27 09:16:51.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"fdc0f349b67d1cb71f79909ce97b0f1e9a6bab12","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-02-12 11:53:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"12a5f903e9801a088da9df2f594807afa050de32","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-12 14:11:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3f28dc9534d446b49a354fe42a29cbeb\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0c808d28b3f346f8afe671c195ce7bf2 : FAILURE in 6m 46s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/ade3a749c0594280a1511db95f5087ec : SUCCESS in 1h 14m 56s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/7548bb81299e42bebeda6cedf3119d65 : SUCCESS in 1h 15m 21s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/eecd283b86554167a71893f6df2904f7 : SUCCESS in 45m 30s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7d01e15844534217a3edb289d5fd61f0 : SUCCESS in 1h 11m 00s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/84efd8d4a2a3495c9fda9f26995fbb4e : SUCCESS in 1h 24m 36s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9a2dc2b8e7614f0baa4c7dcdd2712c5b : SUCCESS in 1h 22m 07s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/08ff9bd076bb4f6fafcf2b0b54f2fd18 : SUCCESS in 1h 16m 34s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7fb0ecfd121d42db84d43ad2ad0360ca : SUCCESS in 1h 17m 03s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b841c29cbe7349af9e36037d19c508e7 : SUCCESS in 1h 38m 26s\n- openstack-ansible-upgrade_2024.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/661438a5127740d5b244a3f1b5f281f5 : SUCCESS in 2h 06m 39s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8c17e07a4bfe4d7c9fc0c0b96d6ea3c1 : SUCCESS in 1h 50m 17s\n- openstack-ansible-upgrade_2024.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/264ebd0afc974a8287e10ec4b2aebc99 : SUCCESS in 1h 41m 33s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1d0c99335b41423f855386a02a465463 : SUCCESS in 1h 48m 22s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9c9db0a1f5e94aea8c0c8a31dc843643 : SUCCESS in 1h 46m 58s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4b992307d759401b998dc94ca45659c9 : SUCCESS in 1h 05m 30s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4c8ad60e6d49403c81112b79966219cf : SUCCESS in 58m 35s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/d775ecdbf65348628774dfc640ff395b : SUCCESS in 1h 26m 33s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1af019d73dd64282b541e1d28f288ddb : SUCCESS in 1h 10m 13s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b03183525486433e84c98e89ffc74c6d : SUCCESS in 1h 19m 06s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/b82bda8e70c7441b920f35d00b0f4921 : SUCCESS in 1h 11m 18s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/56e62e6a166a40a18ceba3f1f8f7ffac : SUCCESS in 1h 18m 50s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f302f8adf4f341d0af4bb0d1993a872f : SUCCESS in 1h 03m 40s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b3f28b37b8f4449f8a162fde527dbe80 : SUCCESS in 1h 11m 29s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/220742eb9e6f4b00ad1ec0f058993c57 : SUCCESS in 1h 11m 54s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5d70b5453c6d4973a13d92e2e406f2c1 : SUCCESS in 1h 23m 06s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/93449eb8ac6543e4a001e933f9844311 : SUCCESS in 1h 16m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1a4456e41f74c71b7784aa4a33d5e55 : SUCCESS in 4m 15s","accounts_in_message":[],"_revision_number":1},{"id":"027a900fa3963ab0e684e0ae64f3bd8e0e5e05b3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-02-12 15:44:03.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"6f8acff797c4da4757eadddf043c8f52738a5a97","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-12 18:02:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b59436466f094e218a91f279402f074d\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/278ab76f647a47fca65d7cc2f12d90c1 : SUCCESS in 7m 17s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/02a06474f4894c0d973381fb20ff5b6e : SUCCESS in 1h 12m 10s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/1fe1d1dc174a4a9a9e0eba73bf1a20d5 : SUCCESS in 1h 21m 46s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f0893dc72c8748328a561d83839e8309 : SUCCESS in 1h 20m 47s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8d3ca68bd8674797a305d3cd326d8a07 : SUCCESS in 1h 26m 56s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/85b66278991442a3a0aa5783b1dd4af9 : SUCCESS in 1h 20m 57s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1981138c64c54815bed76b468df8f8b4 : SUCCESS in 1h 23m 14s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3ba8a5fa6225446db9a1ecb59f5c0880 : SUCCESS in 1h 20m 04s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cc2b8906db8e453aa4be5e7be2f7610f : SUCCESS in 1h 22m 16s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/23757272bcd54165981f9006bb5d90da : SUCCESS in 2h 10m 08s\n- openstack-ansible-upgrade_2024.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/99a58ea56a414804975d3756901a88b5 : SUCCESS in 1h 53m 29s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2aed9a4ff9934d24a7f7cf73027f1c39 : SUCCESS in 1h 38m 33s\n- openstack-ansible-upgrade_2024.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fb139b4e63c247d4bd88f586174d5525 : SUCCESS in 1h 43m 53s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/695ac974cae241c480c5970ae4c04187 : SUCCESS in 1h 56m 35s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/046bab929df8406db6170cc74dc2437c : SUCCESS in 1h 49m 08s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b9653f9115284eed8b7205591d52d318 : SUCCESS in 58m 48s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/69a7d1e8318d423e87e1378e3ddb60b5 : SUCCESS in 1h 01m 06s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/25b3c405b7f843b294153a9cdea2acdb : SUCCESS in 1h 15m 25s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f0a90fe59f4f40cab2f8dedea2ec144c : SUCCESS in 1h 19m 27s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7983bf564221457ba050932e2b1d5817 : SUCCESS in 1h 08m 44s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/66ef0d6c22de42a497cdb6811574aa82 : SUCCESS in 1h 17m 03s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/49de673a20ef4147bc18ad37f9f0259c : SUCCESS in 1h 22m 08s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0feb77e6ea3a4027b33bf466bdefcac4 : SUCCESS in 1h 04m 50s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a1277e30974a4627a556ef08542f867a : SUCCESS in 1h 09m 46s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fbc74d6996d441d8875fffeb40c1081e : SUCCESS in 1h 10m 02s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5c8a6f2c978241fc8f577971a66406b1 : SUCCESS in 1h 18m 03s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5d43a5db461640429f0458ed63c58890 : SUCCESS in 1h 16m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91f1c4677f4c4a2bb1c16aac6ff78ae1 : SUCCESS in 3m 59s","accounts_in_message":[],"_revision_number":2},{"id":"32785978bd51ffaa96e001f971825cdcfc52f030","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2025-02-25 12:38:06.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"0141689bf31a643f6371eb9ecf7dff890ebe5a5a","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-02-27 08:31:37.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"51584abda9e8e3a655c2645b8d6c370e5bf42288","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2025-02-27 09:16:51.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"fc4e8a6fc31d3dc6996395c62d94e7d60abd22b8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-27 09:19:05.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9f11d7d653ebaf85bfd25a3017f8b272e53aad84","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-27 11:51:11.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/90b313340df14ff0b05093fe2656e7c6\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d2247b9145fb4ae4b2e0371f5244f19a : SUCCESS in 9m 51s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/b1684e019d544afe82e59be4a6e3ed82 : SUCCESS in 2h 00m 50s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b96102fa71ad4193b05a325af7f6d31b : FAILURE in 1h 19m 16s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1d2e8017a93c4e2588219eb80af8a7c8 : SUCCESS in 1h 15m 22s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/75d7ffccb5094c25b7ca76d7faa9507d : SUCCESS in 2h 01m 52s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3ff5bb99aa4b4e0690b262117e688221 : SUCCESS in 1h 17m 24s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c42ed50060ae476583159cd55a69e72d : SUCCESS in 1h 54m 32s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1baa401aabab4c66a5a190593d6fe172 : SUCCESS in 1h 19m 57s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d0868ce0f6144d419f4b37ce651b3d06 : SUCCESS in 1h 57m 11s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5de4611b8bf648c485ca9303f6e691eb : SUCCESS in 1h 16m 32s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/f0086715ebfb4b70a9a44a986075e932 : SUCCESS in 1h 43m 01s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8bd3a6a68696493d9abda05dce31ed35 : SUCCESS in 1h 05m 36s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6a5a51fbe3564c34a615aa881fd06eb1 : SUCCESS in 2h 15m 53s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f71835e304aa42499960861a7c281057 : SUCCESS in 1h 13m 58s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/467534006e994d58aee1825f1052510c : SUCCESS in 2h 26m 04s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/761c79e6bef04808ab1ea33730b09a3e : SUCCESS in 1h 21m 11s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d56281e4443a49739869eb785a056daa : SUCCESS in 1h 23m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3a2097991ac4dcdbb2be457950c1e3f : SUCCESS in 5m 08s","accounts_in_message":[],"_revision_number":2},{"id":"77d9eac473e94baa6a096f508e9f97823237d730","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-02-27 11:54:16.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2530ab52d8cf04579d5cb6bd35b6c9c665916a8d","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-02-27 17:33:59.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c53c34c764318673e83cfa83db7b5ecb007f8c72","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-28 00:51:53.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ad42b5c2303e4b52b6c179d30a498d26\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d6e7e8e0c1bd4de3b00a19e2ed19b420 : SUCCESS in 14m 34s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/03a8e00062914e88a5b55434209be8c2 : SUCCESS in 2h 00m 12s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/f3a401bf79384b7aa52e9f85ee1b0f0f : SUCCESS in 1h 20m 14s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c916bd67ebd7492bb1d425886b43145d : SUCCESS in 1h 12m 02s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/993b68e2c29e4b9098352e86f2351d79 : SUCCESS in 1h 28m 10s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f9b54111d4f44f0283fb39f3b5a1f57f : SUCCESS in 1h 21m 19s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/71412dcf9f824393bcbbd1d0ce755cd4 : SUCCESS in 1h 26m 59s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1a40d97c7ed44f7084ec96f3211eafff : SUCCESS in 1h 13m 02s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/74cd2aee114343ada655ac964502cb46 : SUCCESS in 1h 53m 40s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/76347e665e72480a9a252401ed794beb : SUCCESS in 2h 30m 23s\n- openstack-ansible-upgrade_2024.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/edb713f646e14a50a402f59b85a499c9 : FAILURE in 48m 20s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/14359e1bb9074327bb1a2b63820f8cfe : SUCCESS in 1h 50m 14s\n- openstack-ansible-upgrade_2024.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9fc0e9db3408466a89f11d7b118eee32 : SUCCESS in 2h 03m 44s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0fa97a348d4a4aa4ab0d991370c051ed : SUCCESS in 1h 58m 27s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5282b7fa31004728ab8cced4c8aa118d : SUCCESS in 1h 58m 43s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/dddfc54964074c14a83deba22a5ee537 : SUCCESS in 1h 01m 34s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/316314f5310c4800a0da980137b31bc0 : SUCCESS in 1h 04m 54s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/09ca11bc3ab046f389b9f8939f3c2546 : SUCCESS in 1h 25m 16s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/75bfd47201c64f92932446086da8ecba : SUCCESS in 1h 39m 02s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/41aaf824bb5f47a98efa597e754310bc : SUCCESS in 1h 33m 09s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/5c7df5bb19124d2e9230fbf27c1c7917 : SUCCESS in 1h 11m 58s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1877c1eaa06b4cbdbe2ac2486ea13817 : SUCCESS in 1h 11m 38s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/072143c3dcc54d8f860485914e8bbc46 : SUCCESS in 1h 11m 27s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/95e5b6c26ad4430eb2314523d31f54fc : SUCCESS in 2h 14m 26s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ae3831e90e8a4b7fa3653bf96d913905 : SUCCESS in 2h 06m 49s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bc96c17bf87c4371bc5829e7252339ae : SUCCESS in 2h 08m 30s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f0b422c8c6e74104ae81f276d6a90041 : SUCCESS in 1h 10m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a18f6b560ea4b76b079977cd834027b : SUCCESS in 27m 47s","accounts_in_message":[],"_revision_number":2},{"id":"d2e53bedf43f0a79130cc39419d9e2e1eaea7b3d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-02-28 08:36:28.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7f5efbf9e6fe2302cdb6bcf560759aa134dc5a6c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-28 10:48:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec7de9065cd041d7bc6d5968caf694b5\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7892b84377894033a2f8fc7a3c65bdbd : SUCCESS in 6m 18s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/388a6b0f933845b0832f70749a83f9b8 : SUCCESS in 1h 16m 48s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/9f0fe33ca2f941c69a4a10784cc2cc11 : SUCCESS in 1h 17m 32s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ecfb75257d834653982f63557791634d : SUCCESS in 46m 31s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/03f57577f9124ee08dd498fb46fa9d76 : SUCCESS in 1h 28m 21s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c34cd982e96640c285a3e15f65b733a8 : SUCCESS in 52m 08s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b7e580523c874f2fbbf94f0bdeca4bd7 : SUCCESS in 1h 32m 36s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e1c749bee9204b2187af57a1020d18c4 : SUCCESS in 1h 11m 34s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/01ba9819d715479ebde9a7f9bab583f2 : SUCCESS in 47m 02s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/228f713bdefd43bab47c01f2c012d576 : SUCCESS in 2h 04m 53s\n- openstack-ansible-upgrade_2024.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2a790e2755fe4585a546b4f6ed8d6f9b : SUCCESS in 2h 09m 20s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/81653c976de84d779e580335cec6ca16 : SUCCESS in 1h 59m 56s\n- openstack-ansible-upgrade_2024.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a3f7f3ed638e40a58d3b868ff160d148 : SUCCESS in 1h 51m 36s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f777719b9a754dfc817ba45836cbf26b : SUCCESS in 1h 03m 47s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2e01193f9f1048968903192c926343c4 : SUCCESS in 1h 45m 06s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1c0b9b975a05450b9fe89b26136fcb41 : SUCCESS in 59m 59s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4062ab55f20a44f094bb84bcf7aa7216 : SUCCESS in 1h 00m 47s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/3dd44133cf76457aab7b718ddc7206a7 : SUCCESS in 1h 16m 40s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/beb52000d00c4607ac0d7d497755d427 : SUCCESS in 1h 31m 49s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/8f870364418d400b9cd95220b475f4b1 : SUCCESS in 1h 21m 12s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/1667592238da4043a825240265ce2056 : SUCCESS in 1h 13m 30s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9f2be4b544b04d9495f2dfd98352b1f8 : SUCCESS in 1h 43m 39s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4627b56140034a74b373ac7b81f6a517 : SUCCESS in 50m 44s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dc18ebe093e2411cb7e67a05b979c2a4 : SUCCESS in 1h 13m 46s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8dcec4de43294c60868f3cd30ce82072 : SUCCESS in 1h 09m 52s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/469ac7ebc52b44f19f0c13efd1b6f81f : SUCCESS in 1h 14m 17s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/57f31c558e6248e8aa53ccb3787833ff : SUCCESS in 1h 18m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/879be35f82234e05a3be0f13a21bf047 : SUCCESS in 4m 00s","accounts_in_message":[],"_revision_number":2},{"id":"f3ea4a23d2f88be76879c3191bb78725d29534b4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-28 10:50:27.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9fe9e0c4faede0aa19fc94c2b4b6a24aadef2831","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-28 12:27:14.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2374ea031fd74188bcabc41b6cf6bf73\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3c61c9133a964e2b97a3373705d27127 : SUCCESS in 8m 06s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/6e6d3cebd5d147be9e528d800a6e4b18 : SUCCESS in 1h 23m 25s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/549a8d29165e4a258b345a90066b6c3b : SUCCESS in 43m 01s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7b414aad380943c5bbb39a3916626a27 : SUCCESS in 1h 19m 30s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0aa73780874147cc805b6144d228366e : SUCCESS in 45m 35s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f0e38a0f6f754276a96a059da7979b28 : SUCCESS in 1h 28m 11s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c68c665237de4167b5af4ccf7d2148c5 : SUCCESS in 1h 25m 04s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f4c1eeb276c0405a88c19431d66ce241 : SUCCESS in 45m 25s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0adeb6b83daf480abf6e50764cc31081 : SUCCESS in 1h 29m 14s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/471674850854402cb63cc9e1067774b5 : SUCCESS in 1h 20m 10s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/463d7f90ec5c460c999b41eb2729373e : SUCCESS in 1h 14m 47s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b4039df27f1840be9bd3a9f9e4f4add3 : SUCCESS in 1h 05m 10s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/87e233dcf3774ea285edbe6bd44d3681 : SUCCESS in 1h 09m 20s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0a51cf81d7f64f549c31d3a947853261 : SUCCESS in 1h 19m 26s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6e12eab98e5945308917f1fd610e909d : SUCCESS in 1h 11m 31s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8b974a03363e461e923e5c9fd0343d53 : SUCCESS in 1h 10m 47s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/30d95b7245e84c57bbdae50a36dac866 : SUCCESS in 1h 19m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ad75cbd653946dabf5897b3d6e41562 : SUCCESS in 3m 42s","accounts_in_message":[],"_revision_number":2},{"id":"b34dd684b4eca3151e4d51ffeefcdf410e9b1ac3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-28 12:27:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"2d64edf9fc304099027ce0ec69ee2b70f76e2e63","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-28 12:28:29.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2a13b59c0260411db25965734d26e648\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f687825042b417a99c8878951c78b32 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a6b4f710103cbd4f40b53ac2ae016d16aa743f96","revisions":{"6af69d29764dc6707fc022de7e9286a7f4fb7895":{"kind":"REWORK","_number":1,"created":"2025-02-12 11:53:37.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/64/941364/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/64/941364/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/64/941364/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/64/941364/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/64/941364/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/64/941364/1"}}},"commit":{"parents":[{"commit":"93c3d2e4328166e3e98e64e60f5d0ce39b898d37","subject":"Auto-fix yaml rules","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/93c3d2e4328166e3e98e64e60f5d0ce39b898d37"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-02-12 11:53:23.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-02-12 11:53:23.000000000","tz":60},"subject":"Auto-fix usage of modules via FQCN","message":"Auto-fix usage of modules via FQCN\n\nSince ansible-core 2.10 it is recommended to use modules via FQCN\nIn order to align with recommendation, we perform migration\nby applying suggestions made by `ansible-lint --fix\u003dfqcn`\n\nChange-Id: I433d6fe347e21098f563881f3c1fe494231a0b62\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/6af69d29764dc6707fc022de7e9286a7f4fb7895"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/6af69d29764dc6707fc022de7e9286a7f4fb7895"}]},"branch":"refs/heads/master"},"a6b4f710103cbd4f40b53ac2ae016d16aa743f96":{"kind":"REWORK","_number":2,"created":"2025-02-12 15:44:03.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/64/941364/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/64/941364/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/64/941364/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/64/941364/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/64/941364/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/64/941364/2"}}},"commit":{"parents":[{"commit":"93c3d2e4328166e3e98e64e60f5d0ce39b898d37","subject":"Auto-fix yaml rules","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/93c3d2e4328166e3e98e64e60f5d0ce39b898d37"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-02-12 11:53:23.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-02-12 15:44:03.000000000","tz":0},"subject":"Auto-fix usage of modules via FQCN","message":"Auto-fix usage of modules via FQCN\n\nSince ansible-core 2.10 it is recommended to use modules via FQCN\nIn order to align with recommendation, we perform migration\nby applying suggestions made by `ansible-lint --fix\u003dfqcn`\n\nChange-Id: I433d6fe347e21098f563881f3c1fe494231a0b62\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/a6b4f710103cbd4f40b53ac2ae016d16aa743f96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/a6b4f710103cbd4f40b53ac2ae016d16aa743f96"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
