)]}'
{"id":"openstack%2Fopenstack-ansible-os_keystone~961088","triplet_id":"openstack%2Fopenstack-ansible-os_keystone~stable%2F2024.1~I541435d8741fe568c3a69e064b1cc73ffc217bad","project":"openstack/openstack-ansible-os_keystone","branch":"stable/2024.1","attention_set":{},"removed_from_attention_set":{"32666":{"account":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"last_update":"2025-10-06 13:30:34.000000000","reason":"\u003cGERRIT_ACCOUNT_32666\u003e replied on the change","reason_account":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2025-09-16 07:56:29.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2025-10-27 15:49:06.000000000","reason":"Change was submitted"},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2025-10-27 15:49:06.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I541435d8741fe568c3a69e064b1cc73ffc217bad","subject":"Fix package name for mod_auth_openidc","status":"MERGED","created":"2025-09-14 21:40:19.000000000","updated":"2025-10-27 15:49:51.000000000","submitted":"2025-10-27 15:49:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"961088","cherry_pick_of_change":961084,"cherry_pick_of_patch_set":2,"meta_rev_id":"b40bc5a6570d3104735b78ceb0e2bb6417c616bb","_number":961088,"virtual_id_number":961088,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-27 15:49:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"all":[{"value":2,"date":"2025-10-06 13:30:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":2,"date":"2025-09-26 10:54:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"all":[{"value":1,"date":"2025-10-06 13:30:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"date":"2025-10-27 11:24:20.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}],"CC":[{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-14 21:40:19.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"state":"CC"},{"updated":"2025-09-14 21:40:46.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2025-09-15 00:29:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-29 13:21:18.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"state":"REVIEWER"},{"updated":"2025-09-29 13:21:18.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"81d098c633718eda045a112dc4421179a217cffa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-09-14 21:40:19.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"ed37ab21b9895c2be485cbfbcc04f5b22ed46bdf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-15 00:29:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/762720ade49c4060b0caea8382415c4a\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/229254eed1c64fd581bb8426df48aae8 : SUCCESS in 5m 59s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/c27b7da921ef4e2c9d6a67bf4d5ae814 : SUCCESS in 1h 14m 35s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/7df4aab923824b7b86d4be136b9658a1 : SUCCESS in 1h 20m 15s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/6728bd82453a454d922db85029c21b8e : SUCCESS in 1h 00m 41s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6f49c34bb81b49daa4026b48c7b98aad : SUCCESS in 1h 26m 01s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3d91059bf43b4a23bd15e716481cdbb2 : SUCCESS in 1h 18m 17s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ffe4e331054d47029e339bfb57cd02b4 : SUCCESS in 1h 11m 46s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/386fa650a7674ef98d7478fbde5aab17 : SUCCESS in 1h 22m 45s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3c8c4db0dc8d4bbfa7745697a1543e5a : SUCCESS in 1h 15m 44s\n- openstack-ansible-upgrade_2023.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a971af69d4ed454a832e19a7865253bb : SUCCESS in 2h 32m 33s\n- openstack-ansible-upgrade_2023.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c72fb4c237e0484e9e7a84fed8adebdb : SUCCESS in 1h 53m 52s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/df32d258fd6448508c09140295fa5c81 : SUCCESS in 1h 05m 28s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b7d3321747714f84b8ede86358550a01 : SUCCESS in 1h 09m 15s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/62c53dc0a34f48d29c60123108d12bb4 : SUCCESS in 1h 23m 33s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7fefe3cd3c4a4812ad15d1c1ac8f69c4 : SUCCESS in 1h 06m 55s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/86b8359a774a4b21842e89f283d22e02 : SUCCESS in 1h 25m 33s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/9bf26a692bde438caf03ddcbd7cf9fd1 : SUCCESS in 1h 06m 07s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/f97dd42c0e564019a2cd5b136b2b7750 : SUCCESS in 58m 44s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9bb2f49bbe534a7e92c8f1ef3bfb1b8d : SUCCESS in 1h 13m 23s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bd25032e03d74fe6b5c450ce4473ba40 : SUCCESS in 1h 18m 08s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/32d8d2124a254140aadd74e84eaf5be9 : SUCCESS in 1h 18m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d25b3582ed7541afa9dc8d14d8425de0 : SUCCESS in 4m 02s","accounts_in_message":[],"_revision_number":1},{"id":"3c14a094c9302c9f096707be23b4cd0060a0d551","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-09-16 07:56:29.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"a79c582b6c5b9c168d4ef633d392ac4ddc3a2feb","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-09-26 10:54:38.000000000","message":"Patch Set 1: -Workflow Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7353bcbf7e3fb49da1ec49010d2bcd7301b37ba7","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2025-10-06 13:30:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"cb5cd36c6aaedc407024d7c4257b9b9f3b4ab5ad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-06 13:32:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c02c0ddc7f8fdecf3772fefe7b96dd0fe0c7847d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-06 15:12:18.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/45dac13f219d45ae8e591997ec277688\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/11d888fb3fe7439599368bef041614b6 : SUCCESS in 6m 08s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/787ac44aee0648239c2568f0640ca2f3 : RETRY_LIMIT in 2m 41s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/92b9294e0f224b9a9bc5758fbb595915 : SUCCESS in 1h 19m 11s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5024d874d4304164846d5c4fee869814 : SUCCESS in 1h 28m 11s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d606dc325e3c438eabafefff4800dae6 : SUCCESS in 1h 11m 51s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/674b9367b7ff4dff8eea03c6c169433a : SUCCESS in 59m 10s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ebb83f6cccc94bcaa86697c7452d1c58 : SUCCESS in 1h 33m 59s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cbc74dadff384193b6750092691c35f7 : SUCCESS in 1h 19m 19s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5c6896de53894a7f83bdfea44661be0c : FAILURE in 19m 12s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c3a128df87f44db495126cf6dd22b5c6 : SUCCESS in 47m 28s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/57ec9e2084ed456f8939fa5645874276 : RETRY_LIMIT in 1m 53s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/62bbfa0f14f54d64bb081e0970b45f32 : SUCCESS in 1h 12m 10s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dfaad28150e64a9ba9ae6741ed6e2f06 : SUCCESS in 50m 51s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f980271fdf1b4b11b8c4fb71c95191cc : SUCCESS in 1h 12m 57s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/508d5a83bc15477faa1ecce8a448baee : SUCCESS in 1h 20m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bcf24e972834d768311165f2cb71d31 : SUCCESS in 5m 40s","accounts_in_message":[],"_revision_number":1},{"id":"41bf6f752ae475bd80e0975cfdce0a362dd4d92d","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2025-10-06 23:13:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e1b18ca49b66bf44c036706b896e119c00e00561","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-07 00:51:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5f97f602f38d4a8982f13ef46934512d\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a628021000f34d1a84c638311a54b3ae : SUCCESS in 4m 37s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/838562a430564c0a807bfe2e5d0773d4 : RETRY_LIMIT in 3m 52s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/7f11f0f9baf94482b78f5c1675427004 : SUCCESS in 1h 13m 14s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/7eb468f39fd2404c875ff4ea4986883b : SUCCESS in 1h 20m 55s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/91314766ea9b4e78b787ecb58a476c80 : SUCCESS in 1h 21m 00s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6a35a9339df34cb9a06425f4945c3e23 : SUCCESS in 1h 20m 21s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/00b52dbd71a94492bbd39cd82a1bc4f6 : SUCCESS in 1h 18m 44s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/44dcdfb6f4bf47169006c9d751c61859 : SUCCESS in 39m 36s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5a420894019f4fd78f4833c87e374ab6 : SUCCESS in 1h 21m 05s\n- openstack-ansible-upgrade_2023.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3046edb0834a4628ba5838a6af020308 : SUCCESS in 1h 11m 30s\n- openstack-ansible-upgrade_2023.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1be6fe0ac0dc4268a60e32746bc3e7d6 : FAILURE in 1h 36m 08s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/01b640399bbc40a5a6dd89869ff412d9 : SUCCESS in 1h 05m 58s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7698f4638a4349e48482e0078692dec8 : SUCCESS in 57m 27s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/6a94317de3b34125a5408013d3d44d83 : SUCCESS in 1h 05m 09s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/51b3a37fdb3d4c798bb6766418d53bd5 : FAILURE in 8m 06s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/48ca8d3467364645b3992c7cea46681b : SUCCESS in 1h 17m 09s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/d95550898c324218be6b3b58dfed923c : RETRY_LIMIT in 1m 42s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/852d52e1844c491abd00b850a2670078 : SUCCESS in 1h 17m 52s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f91faf9ba84a43ce8a7fd6bea529abcd : SUCCESS in 1h 09m 06s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8733f7e80dc04097a0de7513ae5243eb : SUCCESS in 1h 07m 49s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/009f75a96a7240c8b1189adb19b0a360 : SUCCESS in 1h 20m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84721938c03c4da6b23cdcd9869c8ea9 : SUCCESS in 4m 01s","accounts_in_message":[],"_revision_number":1},{"id":"62b54e3e9d401f7c1bcaae5b81aa50e976a54a83","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-10-09 18:17:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6e4c553b8fe3ebe0b1527e4245a04be142f9a40b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-09 20:33:02.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2e7bf9f40f614ce3ad6e5603f36cc1eb\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/18816aa4deb240b19b7a8e9cf66bea86 : SUCCESS in 6m 27s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/217706793a0648ca8db2c1c0a2b9132e : RETRY_LIMIT in 2m 28s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/65d9300d88244c3db05133e64eaa9cb2 : SUCCESS in 1h 28m 45s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/62dc393a656e49898c713561b550d725 : SUCCESS in 42m 45s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3d52219f74f2445f9526e9b7487898b5 : SUCCESS in 1h 27m 54s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f3e14d2e92534d34954bd407d6c0bbb6 : SUCCESS in 1h 23m 27s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/555991329d9b40d3babd5fef9e177ff3 : SUCCESS in 1h 29m 54s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bd23ae066fb842f0b3a4aaa37be8e451 : SUCCESS in 1h 11m 55s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f3ed33ac2baf4426a65b827e19e0a78c : SUCCESS in 41m 06s\n- openstack-ansible-upgrade_2023.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9e9d2448071e4758ab3a8284ce4d293e : FAILURE in 58m 06s\n- openstack-ansible-upgrade_2023.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/679cbebd347a41edbd3497d44666d431 : SUCCESS in 2h 07m 51s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9532298ba93c4632b44a7a25f3db44b7 : SUCCESS in 1h 06m 51s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/87f95a4cce1d48fa93ac8a97608c8b1e : SUCCESS in 56m 29s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/4d4a22f64eab4402b073a0da5f3dde9e : SUCCESS in 1h 20m 38s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e3a37fbeb3644b1794e4d22890da79e0 : FAILURE in 30m 24s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/fcece4262d4a48a7b0a9806e756f888e : SUCCESS in 1h 33m 39s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/5c7845526daf4d4db7a0084e0d226750 : RETRY_LIMIT in 1m 47s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/ce031398aba5444cbf8067e3bc98846b : SUCCESS in 1h 18m 41s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e0e251894f7c4a2386056ac1d41bca21 : SUCCESS in 1h 14m 17s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a51b0baddc3647699b0ac824e477a8a6 : SUCCESS in 48m 04s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/35a1240430f64e0a8fc82d5ec6bec681 : SUCCESS in 1h 07m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a47f8397f2c49baab5e0649a3d3f4e1 : SUCCESS in 3m 31s","accounts_in_message":[],"_revision_number":1},{"id":"199d0361ce3faf4ec48e31c0d3d89b8322a0f811","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-10-27 11:24:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"004db0b5b71298581eff566733b9093d961b538d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 14:09:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d410599065c846448b7b8da36d0152d1\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ab4b11ee723d42dbafadc1735fe31c1f : SUCCESS in 5m 54s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/2f7e30b16daf425597760fe69f388cd3 : SUCCESS in 1h 13m 51s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/3d4be9e76ea540de96ffc7cb7e9337f2 : SUCCESS in 1h 22m 25s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/6eae64293a074b80848aa2478f702145 : SUCCESS in 39m 30s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f372a0cff6274683bc7cb4dd55561490 : SUCCESS in 1h 12m 00s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ec55baeb7a3e4ac28863b3f1bad83038 : SUCCESS in 1h 16m 13s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/25008c31dc40465b90805e92b7f1ac49 : SUCCESS in 1h 23m 00s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c3b9fad5ad994bfeadf79096e0dde229 : SUCCESS in 1h 10m 55s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/516b67fe9b9f427589b967fd06c31da8 : SUCCESS in 1h 16m 19s\n- openstack-ansible-upgrade_2023.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a0ece1d1c83946a895035975e2170fb4 : SUCCESS in 1h 08m 27s (non-voting)\n- openstack-ansible-upgrade_2023.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ee20a02fdf9a40b5bf0c02cd295940ec : SUCCESS in 2h 17m 49s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/67db2428bb90446391734483f73f4135 : SUCCESS in 1h 05m 52s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/efc5e7ad793744bf9fa231a846365a1b : SUCCESS in 1h 04m 56s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/0ed2c7d465ae4c5eb613d1d3d3e7c202 : SUCCESS in 1h 07m 27s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5969ade5655c4863b08f127d1961633e : SUCCESS in 1h 26m 58s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/8529974533d3461f9688b40934a378f3 : SUCCESS in 1h 29m 55s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/040b5af22ed243cba9698a6c81719b44 : SUCCESS in 1h 11m 25s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/de6f1c4674dc4d91b84fef5e3dfa5fab : SUCCESS in 1h 22m 33s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0c407f52be624beab7a4ac456197dc00 : SUCCESS in 1h 13m 10s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6b7cda24d70740e995ae6c01b08d573b : SUCCESS in 1h 15m 14s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a296315b1b4e46f887e758405ca8557d : SUCCESS in 1h 14m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ff03c01f891422a877722042ae21b3f : SUCCESS in 2m 58s","accounts_in_message":[],"_revision_number":1},{"id":"102e7486d310519f680e9240d1fbb1aa81845829","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 14:11:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"65e9ec4bef6493c3bd90decd3c29970400f14598","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 15:49:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3406efc9b25c4634a482892ca02569df\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/21cf2f26b4e543e4abce619d90203bf3 : SUCCESS in 8m 13s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/ba204a906b8945b88f3b637bef82ea30 : SUCCESS in 1h 14m 24s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/1cfa9f87f4d143cc8457a6bbd1edc410 : SUCCESS in 1h 18m 22s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/82155e02206e47c0b1aef68017e880f7 : SUCCESS in 42m 57s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9e3a10a7b04943f6a62ee1c543faa8c4 : SUCCESS in 1h 17m 26s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/65d22ef094a54fbd810a07d907ee76ab : SUCCESS in 1h 16m 13s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/307ea820ba9344f09198172a16508e63 : SUCCESS in 42m 19s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f6500fc410d94da49fa8790bbb1d2bce : SUCCESS in 1h 24m 56s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/21670ef34224404c883cbbfab6a48bdb : SUCCESS in 1h 07m 40s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/35e268eda7f7492a8b84e159325e9092 : SUCCESS in 1h 27m 06s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/b0cf3920fb9847fdbe8b7aa2b6bb1747 : SUCCESS in 1h 07m 00s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/a26b6c51489842ffb0abbd8ec95beb3c : SUCCESS in 1h 08m 13s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a5f84d6135424eddb91687f6efff8dcd : SUCCESS in 1h 15m 34s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ad53bcf705bd44f598a059cda1d606b1 : SUCCESS in 40m 49s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ca19b3a0715b48fdafe6b625ba3dcd44 : SUCCESS in 1h 22m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1796cd97e99c41aeb68921158406df0f : SUCCESS in 2m 51s","accounts_in_message":[],"_revision_number":1},{"id":"702377642649a706fde8364d316a668a4abe4225","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 15:49:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b40bc5a6570d3104735b78ceb0e2bb6417c616bb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 15:49:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/129e84ee070a4eef8da96f6ab20bb004\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02221fa6e0f64717a1823dfd6c7d62d8 : SUCCESS in 34s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1a7a1afe793aa54f849164526d242589d4558c1e","revisions":{"1a7a1afe793aa54f849164526d242589d4558c1e":{"kind":"REWORK","_number":1,"created":"2025-09-14 21:40:19.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/88/961088/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/88/961088/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/88/961088/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/88/961088/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/88/961088/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/88/961088/1"}}},"commit":{"parents":[{"commit":"54591057cf40c363826e612d8748ee3e85d2c2cc","subject":"federation: ensure cloud credentials are found on utility host","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/54591057cf40c363826e612d8748ee3e85d2c2cc"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2025-09-14 19:43:15.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-09-14 21:40:19.000000000","tz":0},"subject":"Fix package name for mod_auth_openidc","message":"Fix package name for mod_auth_openidc\n\nUpdate package name from mod-auth-openidc to mod_auth_openidc for RHEL-based systems to ensure proper installation.\n\nChange-Id: I541435d8741fe568c3a69e064b1cc73ffc217bad\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/1a7a1afe793aa54f849164526d242589d4558c1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/1a7a1afe793aa54f849164526d242589d4558c1e"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
