)]}'
{"id":"openstack%2Fopenstack-ansible-os_keystone~962361","triplet_id":"openstack%2Fopenstack-ansible-os_keystone~stable%2F2025.1~I0e4585e85426e57d3a2d6fc86ac2d6df00c09c79","project":"openstack/openstack-ansible-os_keystone","branch":"stable/2025.1","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2025-09-28 15:15:15.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0e4585e85426e57d3a2d6fc86ac2d6df00c09c79","subject":"Fix for fails to reload httpd due to unsupported OIDCXForwardedHeaders","status":"MERGED","created":"2025-09-26 14:23:51.000000000","updated":"2025-09-29 15:06:48.000000000","submitted":"2025-09-28 15:15:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"962361","cherry_pick_of_change":961089,"cherry_pick_of_patch_set":13,"meta_rev_id":"27bb2f6e554942e2aa3be1b0f3efd30e200aea09","_number":962361,"virtual_id_number":962361,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-28 15:15:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2025-09-26 14:24:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-09-29 15:06:48.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":2},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":2,"date":"2025-09-28 13:44:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":1,"date":"2025-09-28 13:44:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-26 14:23:51.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"},{"updated":"2025-09-26 16:31:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-28 13:44:18.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2025-09-29 15:06:48.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"}],"messages":[{"id":"12fffcbc7ee93636154ab871103ef284ed8fd535","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-09-26 14:23:51.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"26c460025cf7faa39abc3daac6f03f97a781c465","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-09-26 14:24:13.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e20dc133f3beb0e841845776190bf4fcfc73385f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-26 16:31:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5db730a9cd2048dbbec10096b600faa8\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bf585c5b6eec49838f0f1d13978996ee : SUCCESS in 7m 15s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/feb27977b9b74b9d8439a44622a91275 : SUCCESS in 1h 14m 55s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/19abd5f05d7a49eaa1d2da08e51501f0 : SUCCESS in 1h 20m 18s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/3f246c0db65a458e9ca3b2ba8ce69d5c : SUCCESS in 1h 17m 58s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/034ecbc800d14bc0bcf302aaaf811c0c : SUCCESS in 1h 08m 08s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4fd24bff21ff488d997f7eb8c0e7f815 : SUCCESS in 1h 02m 47s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0eae452cfd4642199af2239995bd65b0 : SUCCESS in 1h 27m 57s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5f9b10c6b882494b990244630ceacc9b : SUCCESS in 1h 26m 58s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cf7cc825c20a4e979b6178d757b2b63d : SUCCESS in 38m 35s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1264d8f444944efabc689b746bdff73d : SUCCESS in 39m 19s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/42a7352bc8ba4ed38c3a4d02cad788ad : SUCCESS in 1h 59m 30s\n- openstack-ansible-upgrade_2024.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6c2aa0f8ceb54623b076a06d8542158e : SUCCESS in 1h 42m 35s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7c72ce1263ae4b659e81cb1d0c221743 : SUCCESS in 2h 01m 32s\n- openstack-ansible-upgrade_2024.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1330b30b44b543dca40b34d86ec6a7ab : SUCCESS in 2h 00m 11s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0566aaac0e224640af4968956fda1a3b : SUCCESS in 1h 56m 24s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f720aa602c354b3f8372ff00d952ea88 : SUCCESS in 1h 52m 56s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9fdbb21fcc634e7b8e64eed9a982c1b2 : SUCCESS in 1h 06m 01s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fa67da2f422e4104937f3c012a9523a7 : SUCCESS in 1h 09m 38s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/f600dcf072424e9bb76aad47b42a93f3 : SUCCESS in 1h 07m 34s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3e4e68b4287a40909c7c8273b21b512c : SUCCESS in 1h 18m 21s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/aa7d3953421a489597efb7159300dbe1 : SUCCESS in 42m 31s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/8057ca43b09348139bc7046cc7e08931 : SUCCESS in 40m 25s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/e61f4fd9147040d5812e45a84f6d134f : SUCCESS in 39m 29s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/00685032e74642a69c2887aa6f4cd459 : SUCCESS in 1h 10m 17s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b0153d72b8c8442c9df6eb7f28be01a0 : SUCCESS in 1h 10m 28s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/116a0fcd3182476bb4493adf86bffcce : SUCCESS in 1h 14m 08s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/963e3888e29b463fa72867f7f62fd373 : SUCCESS in 1h 18m 27s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7cdf39e499b54e32809bc9b4f5f2c784 : SUCCESS in 1h 16m 53s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f3e5b6a289cb47b498f1cb1de9de4db2 : SUCCESS in 1h 20m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69dcd6309feb4d6c85fa7d70bd08c046 : SUCCESS in 6m 10s","accounts_in_message":[],"_revision_number":1},{"id":"1e8599fb9a0ba76cb008c87e043b860d3c537780","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2025-09-28 13:44:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3793a45c46c3eee79ab633408e5ceea4cbf56a70","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-28 13:46:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b35f75ebd9531f7c5b0249f9c2945760192f65d8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-28 15:15:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/23e3db52eb9c48d9aadc7e4b1029458e\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b17de509d7414ff280cec2b30693b243 : SUCCESS in 7m 05s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/3e8a6b52dd3245b38aa7f6507600d224 : SUCCESS in 38m 04s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/faeeac2f9ac2472cbb14a515e834fbbe : SUCCESS in 1h 15m 31s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1afb749f041747a9bdf383fa77f46f81 : SUCCESS in 1h 08m 51s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d0263e9d536b4ae9bac86cc0cfb43f44 : SUCCESS in 1h 11m 54s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7d7aef4e952245d2b79dda6cef64cc6b : SUCCESS in 1h 22m 07s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5eb68c186f8b4235ab4684b84638eda4 : SUCCESS in 1h 24m 21s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7e657656e27b4b8280a20dd96fa762dc : SUCCESS in 1h 15m 14s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d06385e1ac44424ebbe37aa01d66e105 : SUCCESS in 1h 13m 22s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7e971e454c464de18ba31446151996b6 : SUCCESS in 1h 15m 18s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d79d4865edce418b8389ca8e11de39dc : SUCCESS in 1h 18m 28s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/dabdcb92d92749af9249db26a6dea50f : SUCCESS in 59m 51s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/a5a208c8e92249d7a85a5222ad8b282d : SUCCESS in 1h 12m 11s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/52d8bdbbfe6a42c1a4634b8b3b251cf7 : SUCCESS in 1h 12m 10s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4056a1f1f19249dab1f4c0b9cf7e4cd4 : SUCCESS in 1h 10m 21s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bb9b4f881b984654b5f1b0b8cdbb9772 : SUCCESS in 1h 18m 40s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/56f0a45686454224a0e46db25b9c6749 : SUCCESS in 56m 29s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1f34d91fe11e4a489d314c868a3bbf08 : SUCCESS in 1h 14m 29s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/61ffcb29524448139e08c728db31c4b7 : SUCCESS in 1h 19m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f728a805f864377b9f7601f13c836ac : SUCCESS in 6m 29s","accounts_in_message":[],"_revision_number":1},{"id":"d43972603cbb8cd8af5416d92b2a5a6c8db1fe09","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-28 15:15:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"2620c7f077b18e9e4cbf3f74173b64438bc9a7e1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-28 15:16:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/41ce5be6c4524d1288818cbc885a21b2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ee0abf90f0d4c8e9b781ca1f29eac28 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1},{"id":"27bb2f6e554942e2aa3be1b0f3efd30e200aea09","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-09-29 15:06:48.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9d7938a3e14a5b4d2682715bef2b7455d95c1ea5","revisions":{"9d7938a3e14a5b4d2682715bef2b7455d95c1ea5":{"kind":"REWORK","_number":1,"created":"2025-09-26 14:23:51.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/61/962361/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/61/962361/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/61/962361/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/61/962361/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/61/962361/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/61/962361/1"}}},"commit":{"parents":[{"commit":"39f54edcbcee776a38dcd192064bf0df0fc6cc4c","subject":"Update TOX_CONSTRAINTS_FILE for stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/39f54edcbcee776a38dcd192064bf0df0fc6cc4c"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-09-14 22:04:03.000000000","tz":0},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2025-09-26 14:23:51.000000000","tz":0},"subject":"Fix for fails to reload httpd due to unsupported OIDCXForwardedHeaders","message":"Fix for fails to reload httpd due to unsupported OIDCXForwardedHeaders\n\nIn Rocky Linux 9, maximum available version package is\nmod_auth_openidc-2.4.10-1.el9_6.2.x86_64\n\nIn Rocky Linux 10 is mod_auth_openidc-2.4.15-4.el10_0.1.x86_64\n\nIt is necessary to check the value of:\n_keystone_sp_apache_mod_auth_openidc_gte_2_4_11\n\naccording to the version of operating system.\n\nCloses-Bug: #2123161\nChange-Id: I0e4585e85426e57d3a2d6fc86ac2d6df00c09c79\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n(cherry picked from commit 9852f644f79e7bd44d2de51485f4b7cccc448a37)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/9d7938a3e14a5b4d2682715bef2b7455d95c1ea5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/9d7938a3e14a5b4d2682715bef2b7455d95c1ea5"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
