)]}'
{"id":"openstack%2Fopenstack-ansible-os_keystone~965200","triplet_id":"openstack%2Fopenstack-ansible-os_keystone~master~I57f9423e69e7353f89b5571e7eb540a31c9bce6e","project":"openstack/openstack-ansible-os_keystone","branch":"master","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2025-11-05 13:17:12.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}},"hashtags":[],"change_id":"I57f9423e69e7353f89b5571e7eb540a31c9bce6e","subject":"Fix URI for Shibboleth on EL","status":"MERGED","created":"2025-10-29 09:14:27.000000000","updated":"2025-11-05 17:30:47.000000000","submitted":"2025-11-05 17:29:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"965200","meta_rev_id":"01dac3604a96311fbeb50b44b29e356c36304663","_number":965200,"virtual_id_number":965200,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-05 17:29:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"recommended":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"all":[{"value":1,"date":"2025-10-30 10:53:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":2,"date":"2025-10-31 10:38:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-11-03 16:04:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-11-03 16:04:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":1,"date":"2025-11-05 08:54:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-29 11:42:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-30 10:53:55.000000000","updated_by":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"reviewer":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"state":"REVIEWER"},{"updated":"2025-10-31 10:38:14.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2025-11-03 16:04:14.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"ccd8b13a8366d3cfb7aa73a9f15167a93ff9f9c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-10-29 09:14:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cd050a04e6d2fe03b62b333144b039eccb65b464","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-29 11:42:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/76b35937d50b498e99928229e6626ee1\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b4fcc15098ba42468d1c7f134e2eb33a : SUCCESS in 5m 04s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/792fa9486f51470fa830bfee612181b8 : SUCCESS in 1h 26m 05s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/e7135b3b5de6402d95fdb0c0f8a59f10 : SUCCESS in 45m 32s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/4b3902c93ba84f7a8f7f2aae0128de4e : SUCCESS in 1h 18m 40s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/5cf6edef5f9a4097bbfc6ff7d3c5b395 : SUCCESS in 45m 36s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/575203d5cd51453f931f85b572fccb0c : SUCCESS in 39m 28s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0d4932f3cfff4879ba9d09b23706f08f : SUCCESS in 1h 20m 11s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/33aa03f4443b4ff0ae7b678375861cbf : SUCCESS in 58m 25s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bb31341d3360400b81b23f5be2e0a642 : SUCCESS in 2h 21m 27s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/a35fba0bbd084a058fa0caa831c80aa7 : FAILURE in 6m 06s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c37365515605469890d69b77e0702304 : FAILURE in 47m 51s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/f01778044bab483b8b6708344c04a10c : SUCCESS in 1h 09m 50s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/872aa94d11a94dc88b201db71a2e1199 : SUCCESS in 1h 12m 08s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/5690a2dadfd842ff9fbcf6de2c088334 : SUCCESS in 52m 33s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/81f64cbfc719437ca4a82e5ec9782400 : SUCCESS in 1h 09m 01s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6443091c523f43e596fa92784ee64686 : SUCCESS in 1h 16m 19s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/22a3674b660b42e6bfc7905b39511b83 : SUCCESS in 1h 22m 28s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/417ca5d6fe37478ebf310424807ec00f : SUCCESS in 1h 10m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/771fd69a0445494a9e84168950ee7bdf : SUCCESS in 4m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4fe094937e88442cb3f2109040f773f2 : SUCCESS in 3m 29s","accounts_in_message":[],"_revision_number":1},{"id":"66800bc524a54860e95a3adcd75137da8d238b96","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2025-10-30 10:53:55.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"8d2702a7030aaa5de2b3da0b02713d0a95ae97e4","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2025-10-31 10:38:14.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3c948a37fad33639d6c6e735ae4a5817eea4f250","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2025-11-03 16:04:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4e959bbd3654fdf64461b4d4464ee837971804e0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-03 16:06:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"49ed30dfc139a48a38fcd27a89806900f0c882ea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-03 17:47:49.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a8bd616cb8f54867b1fd56ce62858fe3\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bd018f1475494857843d552b6b931072 : SUCCESS in 7m 02s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/755e6935b83c42ba815e6648de4e711c : SUCCESS in 47m 21s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/8ba20f9fbe9d4b4e97adfaa5e23dcb5e : SUCCESS in 1h 25m 06s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/8da64785e02e48e5bd4526f1509b1712 : SUCCESS in 1h 19m 04s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/edcaa3c6bd6742eeba359e2a2e8c55e6 : SUCCESS in 1h 12m 30s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9d6e7faf769d4a2c8f9ba671029d67fb : SUCCESS in 1h 27m 51s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fd625cfdd8c34bc4be56ca17e5659e3a : SUCCESS in 1h 30m 52s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/a7d5fcb57f5d4cfab3c2ac318525c5d7 : FAILURE in 15m 01s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/9eef9263f6794cbba305f43f0e4b47e0 : SUCCESS in 1h 18m 14s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/204488036a844cd19c70a3aeb95730de : SUCCESS in 1h 22m 20s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/4bed693e3c8248b99180b8c596cd6d6a : SUCCESS in 1h 24m 32s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b09932907d5d4a069ac6e3b255050e0b : SUCCESS in 1h 12m 40s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/27648d4f70c441c995ba945f738a795d : SUCCESS in 1h 15m 27s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9619ec9556b0444d88ee87c97fb09a9a : SUCCESS in 1h 22m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4801a50013194387909b0d08ecbd254e : SUCCESS in 6m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5918d377540d45949f86fb08a0a801c7 : SUCCESS in 2m 50s","accounts_in_message":[],"_revision_number":1},{"id":"7280f6ba720ec020aca46ef68d23eb03cbc894d5","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-11-05 08:54:40.000000000","message":"Patch Set 1: Backport-Candidate+1","accounts_in_message":[],"_revision_number":1},{"id":"d7a722ce9d637ce47be273ac529c84dd2fff3988","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-11-05 08:55:12.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a7dd19702decf46efe65d093136cb9a5633fee57","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-05 11:18:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c6ae94ff4587492f827ee75e565c9b2b\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1afc9a741d094f5a9c0e26e234efc87f : SUCCESS in 6m 42s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/c67b540c84e049109a86e99e7c1978f1 : SUCCESS in 1h 04m 56s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/c9aee9ca5f16407d99514c9e97e1b3c1 : SUCCESS in 1h 22m 03s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/364ee0954b7445019a3494711e6c9759 : SUCCESS in 1h 17m 31s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/2943b60d6ae74219930c2b1f1f6bc30c : SUCCESS in 1h 09m 35s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/4ddffe249af549fa8a9da839758ed89a : SUCCESS in 1h 09m 13s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2543e83e9baf44619f00cba29a699a04 : SUCCESS in 1h 30m 37s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2a5b3bfaddd1459cb27239e2186ae015 : SUCCESS in 1h 25m 04s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b0b2a5c67e124eed989f9db90f444b2c : SUCCESS in 2h 03m 36s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/259e4e67593e46708d3d1bf853259014 : SUCCESS in 2h 13m 05s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/9dfe4a0b96e14846b810e84b88933e2d : FAILURE in 8m 16s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/77752d139f674bbdb9ef55e8f4451fff : FAILURE in 31m 52s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/34bfd5fe504a4b57b258eb7b1881c852 : SUCCESS in 48m 33s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/d83b060c91034d65a943cde73255d85f : SUCCESS in 1h 15m 30s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/11e761cbd6554f72a6867017e9e574da : SUCCESS in 1h 15m 50s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/451d4628ddb4473082ad3da3dc9b8f9c : SUCCESS in 1h 25m 09s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/6b56aff1a624440faa0ca66532e7b35b : SUCCESS in 1h 27m 14s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/48cdc23502e14b57a43b3cf4dffb07ca : SUCCESS in 1h 17m 31s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7f858e225127444e996f3fdf5d74f8b2 : SUCCESS in 1h 16m 44s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/41cd24c2d14a43ac8167a004f3994208 : SUCCESS in 1h 14m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3943d5a6f49e4b618b7eaa6e535d3c88 : SUCCESS in 7m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/49c1fffc5a0346f3b0cb8b2e569c8a69 : SUCCESS in 3m 00s","accounts_in_message":[],"_revision_number":1},{"id":"06194c24b7ae007412303e8bf473471edeb76281","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-05 11:20:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c2876673789c19e28711aa243af81d9fdb4f83ae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-05 12:56:31.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/51c0ae1143704d208e56c13057d69627\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/596b46c50765454c9265cde0dab66516 : SUCCESS in 6m 16s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/96c622fa3fba4f51955059b9ecc06a58 : SUCCESS in 1h 06m 15s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/71645fc77bd5436d908a6e603b347d33 : SUCCESS in 1h 21m 56s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/a6133c362de9409394c9fb47faeeed87 : SUCCESS in 1h 25m 04s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/2e8e485a8fb0474cbfd40171b6847692 : FAILURE in 1h 07m 16s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/dd9707c8fc0746a0a9ae7b183d449900 : SUCCESS in 1h 09m 19s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/69ed13c3636b4a3ba6d848b14a6e0bcb : SUCCESS in 1h 25m 34s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9c5005d1db8748848a4437a0b27a8326 : SUCCESS in 1h 27m 06s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/a04f7cafdb9e44018856aaabb3b14a3a : SUCCESS in 1h 19m 42s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/7bf79e8f8325425e873d2b7eeb326754 : SUCCESS in 1h 14m 19s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/a7410c5965744fef88828b9c7c50a390 : SUCCESS in 1h 22m 00s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/c1a64e57f5d74c4983f8c0fea8f368b9 : SUCCESS in 1h 20m 11s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/e14fcfe82ac9410fa858d971e3824473 : SUCCESS in 1h 06m 25s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a54c59aaac9a448aaa2c43914ac0c729 : SUCCESS in 1h 13m 41s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/44ee2c3bab3d4ee481a6fbb3a78168b5 : SUCCESS in 1h 07m 02s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c9afda4e89d9467aa840e61ac6251292 : SUCCESS in 1h 16m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b05c602eef1f43cc92f49959e7dc0833 : SUCCESS in 6m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e74dd99c09f34e419ff4bbc9601a7297 : SUCCESS in 2m 35s","accounts_in_message":[],"_revision_number":1},{"id":"e8ae7548b14a50fd668875f39cb86c674704c05f","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-11-05 13:17:12.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"79fc9e5938357e4680591483021c81dde0767a0a","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-11-05 13:36:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d75e3badd6caabf866c31eb44921b114f46851ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-05 15:27:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e2bb8678da70424b88f9d8ede3c7ec14\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/34890f24b567455583f040506a9cc2bc : SUCCESS in 6m 22s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/086d752ab9334668aaafa51da6b898dc : SUCCESS in 1h 07m 14s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/1badd12d365f42849376dd04fe38a3f9 : SUCCESS in 1h 17m 15s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/d40d16861edc470080ed446865d050f4 : SUCCESS in 1h 29m 52s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/361b682380744aaaaea8e5aea23ed6e1 : SUCCESS in 1h 31m 09s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/7f8f43e7d12347a9bb0f8874fbfa289f : SUCCESS in 1h 15m 05s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/67c64a1cbd6942178420bca6bdaac32e : SUCCESS in 1h 26m 04s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/78870c4ee2ef4d7394762bcc29096062 : SUCCESS in 1h 28m 13s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7be201c7bd7d45ed93662dd0463044c0 : SUCCESS in 1h 59m 59s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c5b0b32c51434251aade598660a04594 : SUCCESS in 2h 04m 19s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/dded2e4257f446928f40e849970076a9 : FAILURE in 8m 47s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4899d1a29c0b43e0ae859e2c9599b781 : FAILURE in 55m 10s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/bcdab5403eea4b54b4d557ca2cb37cb7 : SUCCESS in 49m 09s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/3dae94df53024b0a9b25e114abbe0193 : SUCCESS in 1h 05m 09s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/0235d8e103e34ae6928cdd02644e186b : SUCCESS in 56m 53s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/6ef7b9ffec4e4190bb8320a5953d0c72 : SUCCESS in 1h 22m 27s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/2fa646b8ca3f462f8fc1281fe81c36c5 : SUCCESS in 1h 19m 10s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/09aaa7e156dc462fb8b1260c50274064 : SUCCESS in 47m 56s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9b15985e158a49a4969fcacddf5cf173 : SUCCESS in 1h 20m 44s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f91f70cfe19241c2a21f58fc09ff5b49 : SUCCESS in 1h 05m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92f110d39c5d4650b404aa08a0570776 : SUCCESS in 6m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1999f9edadf7477288c8588c5dcab7b4 : SUCCESS in 3m 05s","accounts_in_message":[],"_revision_number":1},{"id":"f6939d89425c6f07d2cea2138a8a3ab53ce28452","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-05 15:29:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c7706ee2c91280e7f57538f73a41274e58edf0f9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-05 17:29:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/49455a0c75394c398e47ab459ff0dfdc\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/329a0c27f58b47cab57276e2151131b0 : SUCCESS in 6m 27s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/0f7830097daf4cb284c91527aa3f22ac : SUCCESS in 1h 27m 52s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/22860bf53e4247deb65260640684f145 : SUCCESS in 1h 15m 59s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/de1967501a60402994fd2ad21efdf4d8 : SUCCESS in 1h 11m 38s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/f7da58372931448dabe62026befa35fa : SUCCESS in 1h 13m 32s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/b9b68e9d4329439080cda18960b71cc9 : SUCCESS in 1h 12m 57s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6692e83e76f24d94bce89612573e7402 : SUCCESS in 1h 26m 18s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/81e36f7f2ef74be08a5811c8dffcfb74 : SUCCESS in 1h 47m 47s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/d3f453bb221e4824826546b16202eea7 : SUCCESS in 52m 23s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/b12954184a574b7babae5257c7a589e9 : SUCCESS in 1h 13m 19s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/12545e615e6d42a49843b8e076964ad4 : SUCCESS in 1h 12m 40s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/adc14c02f5974455b0ff233addc39da9 : SUCCESS in 55m 09s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/83331062ccef409f9516dbb50c17711e : SUCCESS in 1h 26m 46s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d9756cc1281c41bd817ff2a9f15c502c : SUCCESS in 1h 14m 07s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c2588b1c459a42cb8b2c7399ed5b4ba6 : SUCCESS in 1h 14m 12s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/01fbf4fe143d42f4a4833836b887e8d9 : SUCCESS in 1h 17m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1070ef2b563e4f60a6a5c28d514b5993 : SUCCESS in 6m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ece7185dfc4942bfac3b1b52afe124ca : SUCCESS in 2m 33s","accounts_in_message":[],"_revision_number":1},{"id":"f0dc49c9c0dee8cd42fecf6bbdd56d0157286f6e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-05 17:29:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"01dac3604a96311fbeb50b44b29e356c36304663","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-05 17:30:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4660c321ee37402692ea8aff88640a61\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc2c3779e13b4f6ea801fd7e32c6593a : SUCCESS in 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"334882bb46a3e73ccbef7bcf15afd54e04ad5035","revisions":{"334882bb46a3e73ccbef7bcf15afd54e04ad5035":{"kind":"REWORK","_number":1,"created":"2025-10-29 09:14:27.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/00/965200/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/00/965200/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/00/965200/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/00/965200/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/00/965200/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/00/965200/1"}}},"commit":{"parents":[{"commit":"860875cbdc0a5656b4d55cc38fa046bf43ab2828","subject":"Merge \"[doc] Update correct link for Keystone mapping combinations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/860875cbdc0a5656b4d55cc38fa046bf43ab2828"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-10-29 09:14:20.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-10-29 09:14:20.000000000","tz":60},"subject":"Fix URI for Shibboleth on EL","message":"Fix URI for Shibboleth on EL\n\nWe use Rocky repos, as modern Shibboleth is available\nfor Rocky 8/9/10 only. It should be compatible with other\nEL-based distros as well though.\n\nThis is alternative patch to [1]\n\n[1] https://review.opendev.org/c/openstack/openstack-ansible-os_keystone/+/963474\n\nChange-Id: I57f9423e69e7353f89b5571e7eb540a31c9bce6e\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/334882bb46a3e73ccbef7bcf15afd54e04ad5035"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/334882bb46a3e73ccbef7bcf15afd54e04ad5035"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
