)]}'
{"id":"openstack%2Fopenstack-ansible-os_keystone~966206","triplet_id":"openstack%2Fopenstack-ansible-os_keystone~stable%2F2025.1~I57f9423e69e7353f89b5571e7eb540a31c9bce6e","project":"openstack/openstack-ansible-os_keystone","branch":"stable/2025.1","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2025-11-07 12:47:40.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I57f9423e69e7353f89b5571e7eb540a31c9bce6e","subject":"Fix URI for Shibboleth on EL","status":"MERGED","created":"2025-11-05 17:30:22.000000000","updated":"2025-11-07 12:48:30.000000000","submitted":"2025-11-07 12:47:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"966206","cherry_pick_of_change":965200,"cherry_pick_of_patch_set":1,"meta_rev_id":"dc705bd7a5687dd8cf83fdc6f7e4d0eee1238041","_number":966206,"virtual_id_number":966206,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-07 12:47:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2025-11-07 10:55:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":2,"date":"2025-11-06 18:42:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2025-11-07 10:55:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-05 19:53:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-07 10:55:48.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"8574277a207f8cd0f7c8248b5ea5dd953747d0c8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-11-05 17:30:22.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"26acab3a9f09677238486774654d40fff2d2b94e","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-11-05 17:30:42.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3e37565035e24a17c7ed031c682d37e27f6b20ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-05 19:53:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4ee8a7b194764ad082f58277a9ca9d67\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8fc27857b05b4613a2293fbc36a05da5 : SUCCESS in 6m 34s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/c8ce42d929df408182b3487d36546104 : FAILURE in 16m 06s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/a4cc6595b45c4612806611cdc50bd040 : SUCCESS in 1h 14m 53s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/cfd197f4d39d410789153bcdd6d6dbb6 : SUCCESS in 1h 16m 06s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/cc006e1c9c3544bf843983c9942b1bcb : SUCCESS in 1h 21m 56s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/0dab0f42572145489afbb618f689e77c : SUCCESS in 1h 16m 56s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/bb59fc41351942ac8305c95647f90ffb : SUCCESS in 1h 09m 27s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/27d0eec96bea488ea07d8acbd1957122 : SUCCESS in 1h 09m 01s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ec3179876c6f42a7b92cc6c54e1b075f : SUCCESS in 1h 20m 06s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/bac6e3d2109249d1a1ce68b672e829a0 : SUCCESS in 1h 25m 57s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/47f3de7089954cad93ff6e0f948f4549 : SUCCESS in 1h 08m 05s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/69c5aade808e4c838e8fd824b6492f42 : SUCCESS in 1h 34m 08s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/16eb7b8273ab49e5ae19fec77ea19925 : SUCCESS in 2h 15m 26s\n- openstack-ansible-upgrade_2024.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6b239376a7c649578b8d70f6da69105c : SUCCESS in 1h 47m 25s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7dc35b95e1d84a0a9f72d119c43a5606 : SUCCESS in 1h 50m 04s\n- openstack-ansible-upgrade_2024.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/04b7cbaac7524126948d082da8abe705 : SUCCESS in 1h 55m 36s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/45923f56212144f4960ea567f4d6d37f : SUCCESS in 2h 08m 26s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e98033e50952467f942355abb5b08b76 : SUCCESS in 1h 49m 06s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0a6fca059ba74bb6b31e19abd794d91d : SUCCESS in 51m 27s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/c3d4e76dba7f450fa647e238ab22c3d1 : FAILURE in 13m 35s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5c5b069c65f348909740bd765332caad : SUCCESS in 1h 07m 53s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/0dcaaf4e6cca485486b0d495ca1a8bb5 : SUCCESS in 1h 17m 14s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/87d8bd1b583e4f548990f1a2ebb7ee0d : SUCCESS in 1h 16m 46s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/fc4dfc9a07514d72a519679940db8084 : SUCCESS in 1h 28m 12s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/03ec1e50a42d4537b7d9a1fb2cf33723 : SUCCESS in 1h 23m 39s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/9d337a0d6bf24dd38292b7bd21814569 : SUCCESS in 48m 40s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/af392fd2a0e24dc28e216349107e4448 : SUCCESS in 50m 13s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/ac3d857ecf92484e85227b2f0dc064d0 : SUCCESS in 47m 26s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fd6b6e0889ea42d9a0383a41dc556469 : SUCCESS in 1h 12m 10s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/059f62b90a88479dbeeabc1fc6fef151 : SUCCESS in 1h 18m 42s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f5966bdf10f54bada34b2c34bec9774d : SUCCESS in 1h 13m 49s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bb1aefdc90384081a1a997d8b64abdda : SUCCESS in 48m 11s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b2773704a34140099f4fe835ebd0de4a : SUCCESS in 1h 13m 25s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/89d06d54adfc4271b90135670848615c : SUCCESS in 1h 19m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1837c39e22284164817ef1f663debf00 : SUCCESS in 6m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9569383e9abc46bc9818d3d4c0c1b8d2 : SUCCESS in 4m 04s","accounts_in_message":[],"_revision_number":1},{"id":"0f0258371b48ca20fac10dd3cd71b541795314bd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-11-06 12:46:02.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"b13820f5d189c8699df465844daae2cf8f1908de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-06 15:11:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/304b5f43b512421b9b8f026b088596c3\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5a57c93b22e847a5bd4b45ed115b294c : SUCCESS in 7m 17s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/f74c227d00994773b6f1720c9618ae48 : SUCCESS in 1h 12m 03s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/1f6e2f4912634956b4d7219b18cd453f : SUCCESS in 1h 17m 17s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/62e2247d98394086ab98e7399c5ef8d8 : SUCCESS in 1h 20m 23s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e732f3612608476bb11aa7ce465e4700 : SUCCESS in 1h 18m 27s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/3799887c22b24a3b93041fd552ca4d35 : SUCCESS in 1h 24m 08s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/00727c6663c2429f9166416b25c8b607 : SUCCESS in 1h 31m 10s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b57fd8f091104dafb25f4597bda2abff : SUCCESS in 1h 27m 42s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9ec122211ceb49ff91e429c3ac7dbef7 : SUCCESS in 1h 29m 13s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/27a87f8f67c64a92a5c516c57e450546 : SUCCESS in 1h 18m 50s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2868a360a4234b5182e4012fe057e288 : SUCCESS in 1h 27m 22s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/09257d361cf54539b3fc7f6f5e6e6dc1 : SUCCESS in 1h 29m 17s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f728b13aa05e4dd29b080672c6ff4c57 : SUCCESS in 1h 59m 35s\n- openstack-ansible-upgrade_2024.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/df68848c4da94df19935ad349f2a73b0 : SUCCESS in 2h 02m 04s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2ede1843f4ea4962be1439f5d2b0028d : SUCCESS in 1h 45m 46s\n- openstack-ansible-upgrade_2024.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2f80c5a7c53b487aaaf30a8e38fb5248 : SUCCESS in 1h 51m 43s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f8863e0ebcb040cbafafafa156887612 : SUCCESS in 2h 07m 32s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c1486ad9b2504762ae0780a0c1a528cb : SUCCESS in 2h 16m 28s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5f177b0006ba43bb8ee3c2eca94d120c : SUCCESS in 1h 02m 22s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/65fad42e016c44b7a1bb0c4a88f042f0 : FAILURE in 11m 08s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8b3d9f751649417ea8fea2cb83d1552d : SUCCESS in 1h 08m 51s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/c1c7b4c408f64353ad3d3b9ced1ca6c9 : SUCCESS in 1h 16m 56s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ff2c01c6a7a944b6a190cd32ac251737 : SUCCESS in 51m 23s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e48ea2d771994b97bb9ba943d52c1c4d : SUCCESS in 1h 18m 46s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/71bb6743515d4a85b7548b67a1601b77 : SUCCESS in 1h 24m 44s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/59e3cd51808f492b82955469f34776e1 : SUCCESS in 1h 13m 35s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/9f69f486280a475880c2e0ccd7b7ecd9 : SUCCESS in 1h 04m 35s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/79b8ec3b87d34286bfc4262fbfd7b281 : SUCCESS in 1h 09m 08s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6deb4c3e37f44363aa16a5e8e29971a0 : SUCCESS in 1h 09m 18s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0817af4ab4c249ffaecfdddd8c09908f : SUCCESS in 1h 12m 19s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c81839973e784be9af246b5769d7ebc8 : SUCCESS in 1h 16m 53s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f87091f6b27144efac296beae489b725 : SUCCESS in 1h 25m 59s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aa625d7dcc664a408fe529a14b1dae79 : SUCCESS in 1h 18m 34s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/96e81b59db8d4ee08ebc8357d9f77e7e : SUCCESS in 1h 23m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96bd54eb81d24d53aad7326922510e8f : SUCCESS in 4m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/42125e5718ca4ac6b4c53300f7168332 : SUCCESS in 2m 58s","accounts_in_message":[],"_revision_number":2},{"id":"8806311e93ffbe420cc79f7d964b815d6a1b1213","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-11-06 18:42:53.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"7e7d771692283f1a78399c5fbc16a8f7b4f67f2e","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2025-11-07 10:55:48.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"2a96ad78e2c8b3aca6f3dc19f06adb6a8fc806fc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-07 10:57:42.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6c1ef429e0597e14ac53bed849963eae01ad079d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-07 12:47:40.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/953ffb23be024cc6a359ae00a77f8cd5\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/70a13be48d69413a93e0364d87a0adbe : SUCCESS in 6m 30s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/af7d90e475294cea87d11388e4a06450 : SUCCESS in 1h 09m 55s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/3d94229405644dd0876bf606525eee1f : SUCCESS in 1h 12m 11s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e97b335294d44d36b1a20e13b4236997 : SUCCESS in 1h 24m 34s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/0da92b42edaa40f690c62f3b59a4e42e : SUCCESS in 1h 30m 03s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/06cce47a7a8c40e99f70aad810ccbbcf : SUCCESS in 1h 09m 30s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0e135957723047f99837641ea12dc4d7 : SUCCESS in 1h 06m 57s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c9c68ce61cc24a9cbe620a81a7b54803 : SUCCESS in 1h 18m 10s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/74327581542d46289870854229bdfb3d : SUCCESS in 1h 19m 09s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dc2f4a4a5828447f80a12f6a75057067 : SUCCESS in 1h 22m 55s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/585a511263ff4b9d99c2d8594a0667b0 : SUCCESS in 1h 14m 12s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5b49e279338f4a90aba1219d5657c260 : SUCCESS in 1h 19m 13s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/344e55f0087c4919932d748d675e51e4 : SUCCESS in 1h 14m 59s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/a8e3d06ca81a470199e7e98a3fb1eaec : SUCCESS in 1h 06m 23s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/c8d47154881f4908856fedc6862e36c4 : SUCCESS in 1h 31m 58s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/4c6d2922fca14a62956824da38f7380b : SUCCESS in 1h 06m 15s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/5e0265e37fd7459880e9500ba67f00fd : SUCCESS in 47m 39s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/970f53c519bc42e2b04b410a403611fc : SUCCESS in 1h 09m 34s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6d468c87a09a45b2a8ed86cfa9a7ece2 : SUCCESS in 1h 01m 58s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/de4cb3903baa44f5a6c7abe796944980 : SUCCESS in 1h 17m 24s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b3b8c0dceebb4d46b8b8c3e25af3fd88 : SUCCESS in 47m 26s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8f5648b32f404f00a4808c874106ebf1 : SUCCESS in 1h 16m 43s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5d212e203f814c3f98e7bdc719c0e6d0 : SUCCESS in 1h 20m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae4c239c972c4e1dbbad9af2e137191e : SUCCESS in 5m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bbc1cabafd794d5f8a0bde0940070175 : SUCCESS in 4m 28s","accounts_in_message":[],"_revision_number":2},{"id":"875780640819887f03368ffb3d339d4c6cd1534b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-07 12:47:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"dc705bd7a5687dd8cf83fdc6f7e4d0eee1238041","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-07 12:48:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c8b016ba986b46dd83eb04d38e44ba7e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/929d467342c34685b78fbd43c48573ad : SUCCESS in 38s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0c1ca81e4e3aa33d310cc40247ab9fa9321462a8","revisions":{"6a842287cafd05d7082f61c65fc98c4f668124ea":{"kind":"REWORK","_number":1,"created":"2025-11-05 17:30:22.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/06/966206/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/06/966206/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/06/966206/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/06/966206/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/06/966206/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/06/966206/1"}}},"commit":{"parents":[{"commit":"59570166d7d93c5134a98cdaacd937d30fbdc662","subject":"Merge \"Fix package name for mod_auth_openidc\" into stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/59570166d7d93c5134a98cdaacd937d30fbdc662"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-10-29 09:14:20.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-11-05 17:30:22.000000000","tz":0},"subject":"Fix URI for Shibboleth on EL","message":"Fix URI for Shibboleth on EL\n\nWe use Rocky repos, as modern Shibboleth is available\nfor Rocky 8/9/10 only. It should be compatible with other\nEL-based distros as well though.\n\nThis is alternative patch to [1]\n\n[1] https://review.opendev.org/c/openstack/openstack-ansible-os_keystone/+/963474\n\nChange-Id: I57f9423e69e7353f89b5571e7eb540a31c9bce6e\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n(cherry picked from commit 334882bb46a3e73ccbef7bcf15afd54e04ad5035)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/6a842287cafd05d7082f61c65fc98c4f668124ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/6a842287cafd05d7082f61c65fc98c4f668124ea"}]},"branch":"refs/heads/stable/2025.1"},"0c1ca81e4e3aa33d310cc40247ab9fa9321462a8":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-11-06 12:46:02.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/06/966206/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/06/966206/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/06/966206/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/06/966206/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/06/966206/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/06/966206/2"}}},"commit":{"parents":[{"commit":"59570166d7d93c5134a98cdaacd937d30fbdc662","subject":"Merge \"Fix package name for mod_auth_openidc\" into stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/59570166d7d93c5134a98cdaacd937d30fbdc662"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-10-29 09:14:20.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-11-06 12:46:02.000000000","tz":0},"subject":"Fix URI for Shibboleth on EL","message":"Fix URI for Shibboleth on EL\n\nWe use Rocky repos, as modern Shibboleth is available\nfor Rocky 8/9/10 only. It should be compatible with other\nEL-based distros as well though.\n\nThis is alternative patch to [1]\n\n[1] https://review.opendev.org/c/openstack/openstack-ansible-os_keystone/+/963474\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-plugins/+/965941\nChange-Id: I57f9423e69e7353f89b5571e7eb540a31c9bce6e\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n(cherry picked from commit 334882bb46a3e73ccbef7bcf15afd54e04ad5035)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/0c1ca81e4e3aa33d310cc40247ab9fa9321462a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/0c1ca81e4e3aa33d310cc40247ab9fa9321462a8"}]},"branch":"refs/heads/stable/2025.1","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
