)]}'
{"id":"openstack%2Fopenstack-ansible-os_keystone~970811","triplet_id":"openstack%2Fopenstack-ansible-os_keystone~master~Ide2bda79fe1dc6d8f908579f3e136951d331ab10","project":"openstack/openstack-ansible-os_keystone","branch":"master","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2025-12-25 15:34:25.000000000","reason":"removed on reply"},"31542":{"account":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"last_update":"2026-01-24 18:43:29.000000000","reason":"Change was submitted"},"32666":{"account":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"last_update":"2026-01-24 18:43:29.000000000","reason":"Change was submitted"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2025-12-24 00:41:46.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-01-24 14:15:31.000000000","reason":"\u003cGERRIT_ACCOUNT_37598\u003e replied on the change","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2025-12-27 20:54:02.000000000","reason":"\u003cGERRIT_ACCOUNT_25023\u003e replied on the change","reason_account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}}},"hashtags":[],"change_id":"Ide2bda79fe1dc6d8f908579f3e136951d331ab10","subject":"Remove unused variable keystone_admin_description","status":"MERGED","created":"2025-12-12 10:36:38.000000000","updated":"2026-01-24 18:44:26.000000000","submitted":"2026-01-24 18:43:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"970811","meta_rev_id":"27c98774b217ec44e6577a6ead956071fdab4998","_number":970811,"virtual_id_number":970811,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-24 18:43:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"recommended":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"all":[{"value":1,"date":"2025-12-25 15:34:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":2,"date":"2025-12-24 00:41:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-12-27 20:54:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-12-27 20:54:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-12 12:44:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-12-12 15:16:53.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2025-12-12 15:16:53.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2025-12-23 18:11:35.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"state":"REVIEWER"},{"updated":"2025-12-23 18:11:35.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2025-12-23 18:11:53.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"state":"REVIEWER"}],"messages":[{"id":"1a528e99711d069db21293fd065ffee635520b21","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-12 10:36:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"555d7bcaa3a81794c8a3da4dc1eb5eecf04f9b7a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-12 10:40:11.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"39d3ed723339ccd4eb10fb267d68a9128eb5b2c7","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-12 10:40:46.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"f2d66fe5fdedfbf4a730dc2b0601b8082d8f7228","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-12 10:40:49.000000000","message":"Patch Set 3:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":3},{"id":"9112337e97617641deb0149cf9acb04a7078a4f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-12 12:44:53.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/84b79971d7cc423e9f2d8dff60906279\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/27d17455cc134a17b2c9d0dceb02b37e : SUCCESS in 6m 28s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/5cc117162d7d4ec49957caeac405721a : SUCCESS in 42m 20s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/a116104988154ec18014872ed14844eb : SUCCESS in 1h 24m 53s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/3a64fed6ff4a4c7681824fadc4b3f50d : SUCCESS in 1h 08m 40s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/3516faaec4b547a78b786874ee051093 : SUCCESS in 1h 29m 26s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/9f95d3d1cbfc4914a381b7cde3e9a286 : SUCCESS in 1h 11m 11s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/07acee1264d14758a38eb627fe783c60 : SUCCESS in 1h 26m 57s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ea249980ffe84b229f77fd702398d3eb : SUCCESS in 1h 09m 00s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1fb695915b024dcbbce1d045202c6481 : FAILURE in 2h 00m 30s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e33838c2ff024ce7b088adc9e7edf7f8 : SUCCESS in 1h 01m 23s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/8e3611d12d634490885c85a1dafd89e8 : FAILURE in 10m 06s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/749179d569c54f2f9fa114c874456159 : FAILURE in 46m 39s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/9a60e31e318f456f8d16a6f5ad00d456 : SUCCESS in 1h 11m 07s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/acdb118fbe1245b68235394a68c75844 : SUCCESS in 1h 16m 02s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/a35659460bdf4df2a79c382c54c80ea7 : SUCCESS in 1h 08m 50s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/3a89273244fd4bbaaebeaec37c126950 : SUCCESS in 1h 07m 49s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/b3c4698f094e46b98b681ee6e663122a : SUCCESS in 45m 31s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/af6cd44e807b4851a72a8e3d6d1d9923 : SUCCESS in 1h 03m 30s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d65f6cc4445142cc93c86741f655829e : SUCCESS in 1h 16m 19s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cfb905265b4448308b5e0f1350568e23 : SUCCESS in 1h 11m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a64fe9758e745329a94f25c53394313 : SUCCESS in 3m 14s","accounts_in_message":[],"_revision_number":3},{"id":"d59c2c80221f3221d97a9db37168e799e033f4f9","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-12 12:47:35.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"193648456b1ade8b13d9ef52a61e2f1b02a7ae02","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-12 12:48:23.000000000","message":"Patch Set 4: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"1a535806f48c4f2e8ed1f2f969ea2268664fb07b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-12 15:10:52.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe60d2d311064d6a930f4423bb2faf36\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6d4ca52c872e4d3f87aa71dfa84a590e : SUCCESS in 5m 50s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/604fe921ae4348f4ab18e4463358ac08 : SUCCESS in 42m 48s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/03bd3ef165fc40d3bbca46fbf972c19e : SUCCESS in 1h 14m 37s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/eb2fabe415e34eea8240c5837748a6aa : SUCCESS in 43m 05s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/65796a8432c248da97b0786d0f688f57 : SUCCESS in 1h 12m 35s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/23500005b162453284c68a9c00e2b57d : SUCCESS in 1h 08m 52s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e371ce936174472a95a5fac67462cd41 : SUCCESS in 1h 27m 48s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a1cdcb5f55564587b3b287caf37af738 : SUCCESS in 1h 25m 38s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e50d9f76681f48ae9009c7b781e5e8e7 : SUCCESS in 2h 10m 28s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3729da2172fd442d801bb3e315e3c9df : SUCCESS in 2h 18m 52s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/f92d88dac3f54633853670f403142522 : FAILURE in 7m 28s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ebc298a9ee124cda9d7d03de18639401 : FAILURE in 48m 46s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/29b8952b05fb4a4e9b3ef1084b24b5f6 : SUCCESS in 1h 18m 55s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/50bc924cabbc4f2282c8f037eba4ea81 : SUCCESS in 43m 18s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/8fd41f1ff7c8491b925f5c50cf0fa3ee : SUCCESS in 1h 18m 14s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/21ee4381d112467ea3204ba16817c00d : SUCCESS in 47m 20s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/3272a92188d04b33a3c77bfed6db8f0c : SUCCESS in 44m 24s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cd8fcb6471764739866a177986f80894 : SUCCESS in 1h 09m 07s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d329060b1edc481fbb8e06a5b328879f : SUCCESS in 1h 21m 01s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8daaf13271a44dbfaf6860e327fac4eb : SUCCESS in 42m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4262000228914fb59dd7406a370a5565 : SUCCESS in 5m 58s","accounts_in_message":[],"_revision_number":4},{"id":"b7ecdc4b010bc39ade46afc6beab0f6f0d573496","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-12-23 11:17:35.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"3921d35d8c6c9a674f5e14161ab5626c24543d2c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-23 18:10:45.000000000","message":"Patch Set 5: Commit message was updated.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"bf0fa3710673946f7bbfa898f504c105697e8b67","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-23 18:11:11.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"6e9cd9c0a871d467c9e9d3c345d50c8de405650a","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-23 18:11:35.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"fd66b334b9361386b0f1bb71e3fca64c9b9fdf68","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-23 20:31:00.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ab108e0b08240f089d428edaddb6755\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a2135321725046b0800327c472f1bd07 : SUCCESS in 6m 45s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/a4d7d9c62fbd4fd48a4e2395ae8a8c01 : SUCCESS in 1h 01m 08s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/3e7b38ee576c419baff7bd0c7e2e759c : SUCCESS in 46m 12s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/cc6de494e035430a934ba82e4416210f : SUCCESS in 43m 59s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/3cb4f13aa25a4288afa0dad0ce303b92 : SUCCESS in 1h 28m 12s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/eef18a12600c4dc9938ed79d3bc861c1 : SUCCESS in 1h 29m 23s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b81d70db71f34815997df5a7c61e0479 : SUCCESS in 1h 25m 07s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6e405dc40d0e440aa8f46515a9aeeebf : SUCCESS in 1h 13m 44s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/c1a2df868240445bba9ea5c1a7c0485f : SUCCESS in 1h 36m 56s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/ee7c59e0274e46fabdc07ea8f8495e1f : SUCCESS in 2h 11m 18s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/abbda957103746ce84edd2dc61168746 : SUCCESS in 2h 09m 58s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/10bb6716cded4819859442ade82bcc6a : SUCCESS in 1h 59m 03s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f2924e741c5e4546a48170329cc01dcf : SUCCESS in 1h 01m 39s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/ddfb2dd232244301bdaf0a49ecbd2afd : FAILURE in 7m 08s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0e160f16d508416ab259ca51c4667f62 : FAILURE in 34m 45s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/1b698ce539f147578052a2477af2b4d0 : SUCCESS in 1h 21m 33s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/a7c134b936134eaea35fcbb303c6073d : SUCCESS in 1h 12m 06s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/edc9e985996041e29ebc8641a839ab5a : SUCCESS in 1h 14m 56s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/05b4945da7ba41dcb22fe2f3bdca4b4f : SUCCESS in 45m 01s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/4a4de341594342448a95fab003d39462 : SUCCESS in 1h 26m 26s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0b25d8e0b8234306add2aacd2cb311b3 : SUCCESS in 41m 26s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/72efdfa7e72c40f0ba43f4d18673daff : SUCCESS in 1h 15m 43s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bfe97533cc9a4cf4bf13b6bf4674a008 : SUCCESS in 1h 20m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73cd3eb9a0e24d3b959aca0d21a3a0de : SUCCESS in 6m 51s","accounts_in_message":[],"_revision_number":6},{"id":"d617f252a6bb2af4c319732faa2f31e2e654bde9","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-12-24 00:41:46.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"f74d1d81ef0d9065f13703896434fc45af290010","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2025-12-25 15:34:25.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"132c1e1caf13b966fee33c057d3d5728dd097684","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2025-12-27 20:54:02.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"c9312e9ee20a9bc9724e475ba92c30480e786562","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-27 20:55:55.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"0db26fbb250d20582242328c8809b1ca24d9f971","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-27 22:33:23.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9f894603ebf140a0a0c54996542a336d\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9fa3445947ff4efe9de14309c9d4aade : SUCCESS in 6m 29s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/ce40523720364ce68eb2c33c5565f024 : SUCCESS in 41m 43s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/f2479b5b5e8b4686801674c72a97fb2f : SUCCESS in 1h 25m 22s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/5aa5ddf2b2d24d67beccf0d2ee1fc41a : SUCCESS in 1h 22m 27s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/1934bcff532b42578fb2b1be6b370104 : SUCCESS in 43m 02s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/a552a2a65bc24e29b844db53c19831a4 : SUCCESS in 42m 52s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6e4b9337b1a34df985221012ebdeeb1e : SUCCESS in 1h 20m 02s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2ce49fb60376486e892abe798ee9c851 : SUCCESS in 1h 32m 36s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/a70559007ba640c4a022f8a2283fc300 : FAILURE in 25m 05s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/69a06ed3b63c4338b2001fd23b555ced : SUCCESS in 1h 10m 52s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/4c2f5d99ce0447cf9b0d3ab921b45161 : SUCCESS in 1h 12m 56s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/8d086df2b6d84396a17279e4a217ab21 : SUCCESS in 42m 23s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/57a622cdece641bcb91df736302f213d : SUCCESS in 1h 27m 48s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1a0d85aa315848d3abec9c555afcf374 : SUCCESS in 1h 15m 27s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ff7ac40c72694418bbaec3fc17fece77 : SUCCESS in 1h 17m 26s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8bdef8536242452d8ae210802cce99d8 : SUCCESS in 1h 15m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f8d7d6002b644088560f134ec591eef : SUCCESS in 3m 22s","accounts_in_message":[],"_revision_number":6},{"id":"aa909774414abd33abd9aa90e077d70a509fac9d","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-22 12:27:41.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"980ac41d72d30490ddb829ce765de077e91e65ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-22 16:30:32.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9bce08f488224e30aee0fe2a975a0745\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8033cde9345641a4a51b32c88d83b6c1 : SUCCESS in 12m 22s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/ccf8fab8b0374106b1251c38a3e4553c : SUCCESS in 48m 05s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/d34d56865761423ca42bc05ef36550b0 : FAILURE in 58m 29s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/8237b043f37b469cb68be616ef2e4a3c : SUCCESS in 1h 22m 25s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/76d98f49d0784c06ac0920953deac37c : SUCCESS in 48m 17s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/67e04190be0b4244be18b7314b7e543b : SUCCESS in 1h 42m 27s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/274d11ee8a70414696426901e868b3e0 : SUCCESS in 1h 34m 32s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/11d7bd1ee29742de88215bfb081954a1 : SUCCESS in 1h 48m 54s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/b4ed3813fff44f31a4da7742ab336110 : SUCCESS in 1h 07m 29s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/e641f8a360f34bb297f6d88467fe0048 : FAILURE in 1h 04m 43s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f39b0f0be4764f0183f7d1feccf975cc : SUCCESS in 2h 10m 36s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c5d22b6b9f484034bd7badf24fc1eb14 : SUCCESS in 1h 47m 12s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/eb3a5805d6da4c4cae92ca3035e0e1a1 : TIMED_OUT in 3h 10m 01s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/6bd0470474f94c53aaedb403563345e5 : FAILURE in 11m 55s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/837a25f28a1a43a09bce0d171518e4c9 : FAILURE in 33m 11s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/bc7d1bb84d5647dfb505c75a71d15dc5 : SUCCESS in 1h 10m 39s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/9c1e061337e64625b61d7e1eaa23db47 : SUCCESS in 1h 05m 24s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/06cac5fcecab4e0f849f9e6acbd69229 : SUCCESS in 1h 05m 57s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/25164d7184e74e668f8b4955687a92c1 : SUCCESS in 1h 19m 42s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/a9afef3f689d4dcf92330f8be1a9683b : SUCCESS in 1h 15m 47s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e53a3c2b77d741879250fa4f6f80700d : SUCCESS in 1h 16m 00s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b7dbf7751a3244ccaebaefffdb505452 : SUCCESS in 1h 16m 50s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bc4ac6611c974943ad1e959af0016159 : SUCCESS in 1h 18m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a31a38691d544de4a3adcf1cd87a8b5b : SUCCESS in 7m 01s","accounts_in_message":[],"_revision_number":6},{"id":"21e43a9f814ef24ed7f049d8e267ad050f2fe563","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-24 14:15:31.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"5f8033a321906058da09a31bf540c8c636a5b934","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-24 17:01:02.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/98bc10951cd8453990e858fdbdb27de5\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6f8f81bcc1f445e58fe7314e3146cfd2 : SUCCESS in 14m 35s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/2bc13c700318485d811b8cf8f41ce783 : SUCCESS in 1h 12m 24s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/b0f7912d07924bd88a6012bfa7920726 : SUCCESS in 1h 29m 25s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/5516b08b0046465b8af95c4125c7b7b1 : SUCCESS in 1h 18m 14s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/18cefc8c249c4aafa72b666bd344a5ab : SUCCESS in 53m 17s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/876f3c3626634b699c2ba3aea6c0c5f3 : SUCCESS in 1h 35m 43s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ae2398b2a6c64f75a3ce06e7ab25e378 : SUCCESS in 1h 08m 39s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2d0e653b3ee6471ea9b6518973db0002 : SUCCESS in 1h 27m 37s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/f338a32983bb4cd795fa038cbc006372 : SUCCESS in 1h 16m 29s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/e86967a050334a73802814b4c3950cdd : SUCCESS in 2h 14m 20s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aa99530cd39e44b6aeab26ad383061a3 : SUCCESS in 2h 06m 27s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/238b0f2da1004993b506edac17e84591 : SUCCESS in 1h 58m 04s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9f2eade23e9b4a7bb504a22448bd50c6 : SUCCESS in 2h 05m 02s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/ed929a7eb1cb4f3680c8e5befe74b927 : FAILURE in 11m 17s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/97688bf9a4c54400a7987bb2167cfb7e : FAILURE in 48m 35s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/2f369c1ca1cd440eb2e9ab61f7538cab : SUCCESS in 45m 40s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/9df8238b499247799ff2f812fb6d5c48 : SUCCESS in 1h 13m 12s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/639c5f0bf615470fa3dcea166cbadf58 : SUCCESS in 1h 12m 13s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/c3442311ed0d4032992416495ee65eb2 : SUCCESS in 57m 00s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/4098fd9bd35e422f8e05513c405ccaa7 : SUCCESS in 1h 15m 36s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4d1c9c3706eb4558b7149267ac879e79 : SUCCESS in 1h 02m 48s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f8ccb7106d9a4443a9ad3bb60683ca5c : SUCCESS in 1h 17m 14s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f129f22fbd084aa1811e9033c2cd379d : SUCCESS in 1h 21m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/caa3a058d8ea42e3a43a7ad04fd5f7cc : SUCCESS in 6m 13s","accounts_in_message":[],"_revision_number":6},{"id":"80d493d4b78a60f901bffc982053e8280498cca1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-24 17:02:57.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"214ed7520d956f3483fc22eea24458bea7e1a519","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-24 18:43:29.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c19e31402a5444cafb38a65e89b3f19\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/519cdb1f1aa94afbb76385979fcf5b67 : SUCCESS in 8m 32s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/a46dff8ab58740b0bcadce2eb6e44ce9 : SUCCESS in 45m 20s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/cb284b52c63d4ae699335d35c3fecf79 : SUCCESS in 45m 35s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/df2a7a85cbb045ab95fe54a56afa1599 : SUCCESS in 1h 22m 26s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/cb1ee81e53d04c37927ecdb0e5306420 : SUCCESS in 1h 31m 44s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/9433a5234ac6483d84870ce2772e4287 : SUCCESS in 1h 31m 55s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b64ad774622343ea85f795b8bab20650 : SUCCESS in 1h 12m 07s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/79b4fc644b744d7597679e35d4a1b359 : SUCCESS in 1h 29m 54s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/dea30cfee23549a08eda244637af362c : SUCCESS in 1h 09m 37s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/7f65c95f50184397a60afce08ee9f3b3 : SUCCESS in 1h 18m 51s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/3c19bfcdcef44f18befd9c8e31a99ab7 : SUCCESS in 1h 22m 46s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/c6778aa6f316419bbff67503a6bd4cd2 : SUCCESS in 1h 10m 52s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/2bd0abc945934a74a509951d7b3bbefa : SUCCESS in 1h 12m 54s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/accdbe52720e40c3a3270eb9743d6cb8 : SUCCESS in 1h 13m 23s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e28ea31bc51041e78f7181a9349f357f : SUCCESS in 1h 15m 06s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0846b510cab34c139c39607f4166d093 : SUCCESS in 1h 29m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95cdc515e8c2414a84e26c8dade80e97 : SUCCESS in 5m 55s","accounts_in_message":[],"_revision_number":6},{"id":"b298f2e36acdd8af3f9e09c3271ab8e521429abf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-24 18:43:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"27c98774b217ec44e6577a6ead956071fdab4998","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-24 18:44:26.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/636a56b139a64c29aa6fac26a9010822\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20430b690d60458e8de3ab0ce1b18941 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"8cbdfa535dcd139f83e6c6631548db8287ee20f0","revisions":{"6ec0c5eb4ce581656c169d21f1461f7f97e07800":{"kind":"REWORK","_number":1,"created":"2025-12-12 10:36:38.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/11/970811/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/11/970811/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/11/970811/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/11/970811/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/11/970811/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/11/970811/1"}}},"commit":{"parents":[{"commit":"9cb8df62e5225ae3a85400557cd1a36210fd37b7","subject":"Merge \"docs: fix for OSA naming\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/9cb8df62e5225ae3a85400557cd1a36210fd37b7"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-12 10:36:38.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-12 10:36:38.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: Ide2bda79fe1dc6d8f908579f3e136951d331ab10\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/6ec0c5eb4ce581656c169d21f1461f7f97e07800"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/6ec0c5eb4ce581656c169d21f1461f7f97e07800"}]},"branch":"refs/heads/master"},"c08448c00ca1a9c423c5b473a131d3fba21f2231":{"kind":"REWORK","_number":2,"created":"2025-12-12 10:40:11.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/11/970811/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/11/970811/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/11/970811/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/11/970811/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/11/970811/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/11/970811/2"}}},"commit":{"parents":[{"commit":"9cb8df62e5225ae3a85400557cd1a36210fd37b7","subject":"Merge \"docs: fix for OSA naming\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/9cb8df62e5225ae3a85400557cd1a36210fd37b7"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-12 10:36:38.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-12 10:40:11.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: Ide2bda79fe1dc6d8f908579f3e136951d331ab10\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/c08448c00ca1a9c423c5b473a131d3fba21f2231"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/c08448c00ca1a9c423c5b473a131d3fba21f2231"}]},"branch":"refs/heads/master"},"6cfec635f86b075baf4aa009a9cbf1d1f857f115":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2025-12-12 10:40:46.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/11/970811/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/11/970811/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/11/970811/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/11/970811/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/11/970811/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/11/970811/3"}}},"commit":{"parents":[{"commit":"9cb8df62e5225ae3a85400557cd1a36210fd37b7","subject":"Merge \"docs: fix for OSA naming\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/9cb8df62e5225ae3a85400557cd1a36210fd37b7"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-12 10:36:38.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-12 10:40:46.000000000","tz":0},"subject":"Admin Tenant renamed in Admin Project","message":"Admin Tenant renamed in Admin Project\n\ntrivial fix\n\nChange-Id: Ide2bda79fe1dc6d8f908579f3e136951d331ab10\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/6cfec635f86b075baf4aa009a9cbf1d1f857f115"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/6cfec635f86b075baf4aa009a9cbf1d1f857f115"}]},"branch":"refs/heads/master","description":"Edit commit message"},"360904aa3a205d65d25e60ed9124a62939c670c2":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2025-12-12 12:48:23.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/11/970811/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/11/970811/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/11/970811/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/11/970811/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/11/970811/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/11/970811/4"}}},"commit":{"parents":[{"commit":"9cb8df62e5225ae3a85400557cd1a36210fd37b7","subject":"Merge \"docs: fix for OSA naming\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/9cb8df62e5225ae3a85400557cd1a36210fd37b7"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-12 10:36:38.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-12 12:48:23.000000000","tz":0},"subject":"Admin Tenant renamed in Admin Project for keystone project","message":"Admin Tenant renamed in Admin Project for keystone project\n\ntrivial fix\n\nChange-Id: Ide2bda79fe1dc6d8f908579f3e136951d331ab10\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/360904aa3a205d65d25e60ed9124a62939c670c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/360904aa3a205d65d25e60ed9124a62939c670c2"}]},"branch":"refs/heads/master","description":"Edit commit message"},"ffa9829173414fde9dbab068734757ae4e49ce7d":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2025-12-23 18:10:45.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/11/970811/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/11/970811/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/11/970811/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/11/970811/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/11/970811/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/11/970811/5"}}},"commit":{"parents":[{"commit":"9cb8df62e5225ae3a85400557cd1a36210fd37b7","subject":"Merge \"docs: fix for OSA naming\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/9cb8df62e5225ae3a85400557cd1a36210fd37b7"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-12 10:36:38.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-23 18:10:45.000000000","tz":0},"subject":"Remove unused variable keystone_admin_description","message":"Remove unused variable keystone_admin_description\n\nChange-Id: Ide2bda79fe1dc6d8f908579f3e136951d331ab10\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/ffa9829173414fde9dbab068734757ae4e49ce7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/ffa9829173414fde9dbab068734757ae4e49ce7d"}]},"branch":"refs/heads/master","description":"Edit commit message"},"8cbdfa535dcd139f83e6c6631548db8287ee20f0":{"kind":"REWORK","_number":6,"created":"2025-12-23 18:11:11.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/11/970811/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/11/970811/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/11/970811/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/11/970811/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/11/970811/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/11/970811/6"}}},"commit":{"parents":[{"commit":"9cb8df62e5225ae3a85400557cd1a36210fd37b7","subject":"Merge \"docs: fix for OSA naming\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/9cb8df62e5225ae3a85400557cd1a36210fd37b7"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-12 10:36:38.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-23 18:11:10.000000000","tz":0},"subject":"Remove unused variable keystone_admin_description","message":"Remove unused variable keystone_admin_description\n\nChange-Id: Ide2bda79fe1dc6d8f908579f3e136951d331ab10\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/8cbdfa535dcd139f83e6c6631548db8287ee20f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/8cbdfa535dcd139f83e6c6631548db8287ee20f0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
