)]}'
{"id":"openstack%2Fopenstack-ansible-os_keystone~998694","triplet_id":"openstack%2Fopenstack-ansible-os_keystone~master~I18b21021174ab2cf8ca3a502982f810f479a9b84","project":"openstack/openstack-ansible-os_keystone","branch":"master","topic":"osa/ansible_managed","attention_set":{},"removed_from_attention_set":{"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-08-26 09:05:15.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I18b21021174ab2cf8ca3a502982f810f479a9b84","subject":"Update ansible_managed variable","status":"MERGED","created":"2026-07-25 06:06:22.000000000","updated":"2026-08-26 09:06:33.000000000","submitted":"2026-08-26 09:05:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"998694-osa/ansible_managed","meta_rev_id":"14ab26084e95aae765949324fe8a9dfabc8ab921","_number":998694,"virtual_id_number":998694,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-26 09:05:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2026-08-20 11:18:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-08-26 06:56:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-08-26 06:56:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-25 09:15:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-20 11:18:34.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2026-08-26 06:56:32.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"a4c85f36cc3f556354e0917d0768536f46b18099","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-07-25 06:06:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"05e2680d755102e87ae950d507c41c507f510db7","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-07-25 06:07:57.000000000","message":"Topic set to osa/ansible_managed","accounts_in_message":[],"_revision_number":1},{"id":"42edc4a7f316d6256d9087b1cabdef40559ffaa2","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-07-25 06:09:52.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"b4e0c2489d7e751c32675ed657ee730e07d20ed7","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-07-25 06:09:55.000000000","message":"Patch Set 2:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"92bf9864989ad9aa13b8a72f529d3c507b34df51","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-25 09:15:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7469b586224f49f9909ba4fa6a00a979\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7b9aaf1908d84a00aba5193ff067f988 : SUCCESS in 11m 39s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/2492300670324591be04feefde70f3f9 : SUCCESS in 2h 05m 11s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/c611ebdf86e746edb1adcec06ae4e560 : SUCCESS in 1h 24m 31s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/de3f478a62a84433bf70e36779e3b92d : SUCCESS in 51m 33s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/70040de811f443d48631b71cda519f90 : SUCCESS in 1h 34m 19s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/58bbc44461264f7dbfdf26569cc74420 : SUCCESS in 1h 29m 13s\n- openstack-ansible-deploy-aio_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/9e448b3f43a84e56aadc76f7a24f8d17 : SUCCESS in 1h 24m 21s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/83b9f7233f5248d1a873b8f54ac32338 : SUCCESS in 1h 36m 54s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/d43ee98a8a4649d7ad33456b61e0b2e8 : SUCCESS in 1h 31m 59s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9a61573bc64943dca26d9b21a508135b : SUCCESS in 1h 18m 59s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/080635b54616464dbf8c67d22e378133 : SUCCESS in 2h 29m 27s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e59078f9369448fd9db20756e0c0a220 : SUCCESS in 48m 20s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/62ebd72c93044343ab26051d3cbcaee1 : SUCCESS in 45m 16s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/09286c7ed37247ea910a0741cfe606c3 : SUCCESS in 3h 02m 42s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/23462b2c82544925bc4d2f5fcc23c82e : SUCCESS in 50m 48s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/2735bf0bf8b94ebf8df44f41eb3ff1cc : SUCCESS in 1h 42m 52s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/a2cab632d68448a48d9f1103da3a3da7 : SUCCESS in 1h 23m 25s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a7734e39f22745feb5a59d0e6f141288 : SUCCESS in 1h 28m 36s\n- openstack-ansible-deploy-infra_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/9b42238b28d44cfaab2f40f4c27d0f08 : SUCCESS in 1h 44m 37s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6dbbb9339d354a2e80170143fa9c8dd8 : SUCCESS in 1h 33m 37s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/71fb9dc9d7bb49e38dc2033c2f3b6607 : SUCCESS in 1h 30m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/475dfdca71404425ada37d7dd2922b5a : SUCCESS in 5m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b0e7ce866bd04396b512913b1c93d4bd : SUCCESS in 3m 48s","accounts_in_message":[],"_revision_number":2},{"id":"448ceffda8edb145f6a9cef7a45975c4839407ee","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-08-20 11:18:34.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"c066f225925852cfc4726d61f11f6498c515c1f9","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2026-08-26 06:56:32.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"1b8d01734c1991380034b8516e7cd3c86631b0aa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 06:58:37.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f59ee86899b33ac4febe4977159437ca4aa99d34","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 09:05:14.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d7173c80ce146a5aeb162f4d9d58b13\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/351348ce70d5417c82ac66b8911e30ce : SUCCESS in 14m 31s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/056b29d7bcda45dc95e21135be8622d8 : SUCCESS in 1h 11m 11s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/c4f7ee040e8f4b44b5cd32292260c708 : SUCCESS in 1h 20m 06s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/fd62192e7a4b44d69fbc7b41260de839 : SUCCESS in 1h 12m 46s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/9556184b0f674900a8eb41c73b0247dd : SUCCESS in 1h 20m 18s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9f4e0237720249bf9af6a33bd3bf431f : SUCCESS in 1h 18m 11s\n- openstack-ansible-deploy-aio_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/ec4b1681efe34fb6b42fc4b55a171040 : SUCCESS in 1h 24m 42s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f1401bdde84947f18c44deedfc7eac15 : SUCCESS in 1h 18m 47s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3a6e6ded9c214552b241e5b0f66df7d8 : SUCCESS in 1h 03m 47s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/71191079a7ac4642a2edc952b99f6e6f : SUCCESS in 1h 12m 47s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/5956a23118be46fab45f52206d86a09b : SUCCESS in 1h 15m 41s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/f065230f3ea5418390ae42b829dd145b : SUCCESS in 1h 12m 24s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/76fdc0402f3e4ef4a15762b974fbe0cd : SUCCESS in 1h 20m 50s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/02a35261790040bea39520645da2cf42 : SUCCESS in 1h 18m 29s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2fe66ea1ed484c219840df033f136df4 : SUCCESS in 1h 52m 04s\n- openstack-ansible-deploy-infra_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/5049a8973de6439ca843e0900d1f925b : SUCCESS in 1h 36m 18s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/434def5e5fdf410eaa0c1a3abda6b002 : SUCCESS in 1h 42m 05s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ba84df3c54674b3da65bcf60226e1e56 : SUCCESS in 1h 14m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64d9a0d552b4418cb4e8d09bcfd837a8 : SUCCESS in 6m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/db52399f9f034c349ef766376799576f : SUCCESS in 3m 58s","accounts_in_message":[],"_revision_number":2},{"id":"9fd0137b8ad3de3ea96cacae995ce5ac97a5527a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 09:05:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"14ab26084e95aae765949324fe8a9dfabc8ab921","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 09:06:33.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a98374fd9377427cadced6d7f687a0cc\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6b574dd4a6c4b15a457bdfd34078534 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"559b784db224e5b48fe7e85bd84e0a4454ce75a8","revisions":{"6fd161fccf8e78db7a93a4d71059e95e6ee00358":{"kind":"REWORK","_number":1,"created":"2026-07-25 06:06:22.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/94/998694/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/94/998694/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/94/998694/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/94/998694/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/94/998694/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/94/998694/1"}}},"commit":{"parents":[{"commit":"23dfb373656de5ba4a39bc5653334a7421431df0","subject":"Fix shibboleth defenition for http role","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/23dfb373656de5ba4a39bc5653334a7421431df0"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-25 06:06:22.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-25 06:06:22.000000000","tz":0},"subject":"Update ansible_managed variable","message":"Update ansible_managed variable\n\nDeprecated in Ansible Core 2.23 [0]\n\n[0] https://docs.ansible.com/projects/ansible-core/devel/reference_appendices/config.html#default-managed-str\n\nChange-Id: I18b21021174ab2cf8ca3a502982f810f479a9b84\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/6fd161fccf8e78db7a93a4d71059e95e6ee00358"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/6fd161fccf8e78db7a93a4d71059e95e6ee00358"}]},"branch":"refs/heads/master","conflicts":{"no_base_reason":"NO_MERGE_PERFORMED","contains_conflicts":false}},"559b784db224e5b48fe7e85bd84e0a4454ce75a8":{"kind":"REWORK","_number":2,"created":"2026-07-25 06:09:52.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/94/998694/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/94/998694/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/94/998694/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/94/998694/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/94/998694/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/94/998694/2"}}},"commit":{"parents":[{"commit":"23dfb373656de5ba4a39bc5653334a7421431df0","subject":"Fix shibboleth defenition for http role","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/23dfb373656de5ba4a39bc5653334a7421431df0"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-25 06:06:22.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-25 06:09:14.000000000","tz":0},"subject":"Update ansible_managed variable","message":"Update ansible_managed variable\n\nDeprecated in Ansible Core 2.23 [0]\n\n[0] https://docs.ansible.com/projects/ansible-core/devel/reference_appendices/config.html#default-managed-str\n\nChange-Id: I18b21021174ab2cf8ca3a502982f810f479a9b84\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/559b784db224e5b48fe7e85bd84e0a4454ce75a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/559b784db224e5b48fe7e85bd84e0a4454ce75a8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
