)]}'
{"id":"openstack%2Fopenstack-ansible-os_manila~705019","triplet_id":"openstack%2Fopenstack-ansible-os_manila~master~I2515dc4abf93f2d30157a376c8a9a497c865aaf5","project":"openstack/openstack-ansible-os_manila","branch":"master","topic":"osa/manila-generic","hashtags":[],"change_id":"I2515dc4abf93f2d30157a376c8a9a497c865aaf5","subject":"Add Manila key generation and distribution","status":"MERGED","created":"2020-01-30 14:14:02.000000000","updated":"2020-11-18 16:27:46.000000000","submitted":"2020-11-18 16:26:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"705019-1605716778939-195fdec3","meta_rev_id":"47c5e765e1aa9fbc4c981f685907ac4b68affac3","_number":705019,"virtual_id_number":705019,"owner":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-18 16:26:18.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2020-11-17 16:41:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-11-18 13:23:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-11-18 13:23:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-17 16:41:34.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2020-11-18 13:23:12.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-11-18 16:26:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"988372b7237d5ad48ee14f475520b1203e069712","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-01-30 14:14:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7b6f1cefbdf2a81cf779bbb365d555958c9dbfa9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 14:14:22.000000000","message":"Patch Set 1: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"2e1bf67284959543b952a54178c03bd50cd64617","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-01-30 14:20:57.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"efb616dde524527b2edef816e02771e3cde06412","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 14:21:20.000000000","message":"Patch Set 2: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":2},{"id":"bda20b78bf1d2602d0a36211057ef9827b6decb4","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-01-30 14:38:25.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"8b6841ee390347bc64341f7d1c99d4b43a8616d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 14:38:45.000000000","message":"Patch Set 3: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":3},{"id":"746ddfbdc8841308105de845760fbbb637fb7f03","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-01-30 17:08:06.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"1f3cca35d41756b9394c7495af342a6171aa1aac","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-01-30 18:31:45.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"e53cbf8ba5ed8315c72b17b817951bf1e66174c8","author":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},"date":"2020-01-30 19:05:32.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"e9abc677445057645c39f19e8a1481862cd37e20","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 21:46:29.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6cd2e69fda7d4b3c8a016dd95e167a6d : SUCCESS in 11m 10s\n- openstack-ansible-linters-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/908132f8321446228e3ff91c6f5b316c : SUCCESS in 14m 08s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dbcca2aea7d84d0981e780139182bda5 : FAILURE in 1h 13m 47s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/2c6502d68a4b4936aa45fa8fa97816ef : FAILURE in 1h 48m 12s\n- openstack-ansible-deploy-aio_lxc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b60c362acabf448bbc1728958de97b39 : FAILURE in 2h 33m 00s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"64970a99f469007c798c7720902907ff6b7fcb16","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-01-31 15:15:03.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"60eca0024e80aaf63fdc6814c76653f0fd4ee7e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-31 17:12:53.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d1591db2f9143459dfcc44dbfa93656 : SUCCESS in 4m 47s\n- openstack-ansible-linters-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/17aad65560a444c285327b0357694517 : SUCCESS in 10m 09s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/485e3b315e9a410aa16c8ca2be3a1673 : FAILURE in 1h 16m 29s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/988898fa51164c5897b93d4b7e0df8c2 : FAILURE in 1h 52m 40s\n- openstack-ansible-deploy-aio_lxc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9f747f1a104f41d4a2bb0fe9eafa9a7c : FAILURE in 22m 03s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"9397be4729b1387bc97c5fcc6cd3d7d6b2959111","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-02-11 08:04:25.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"30052724a035b4d1798878f9935116df319b34c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-11 09:47:21.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4360a262735440ca9871e139c9ce27b : SUCCESS in 4m 24s\n- openstack-ansible-linters-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/302defa8063949d8b9b390204bbc6605 : SUCCESS in 10m 30s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6c4fd2529df243298527769e1a18ccf2 : FAILURE in 14m 47s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/ff27b3282f6f487ca777a52837f36488 : FAILURE in 17m 43s\n- openstack-ansible-deploy-aio_lxc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d42f91ed30144721a95142124063041a : FAILURE in 1h 37m 51s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"e38d71d8e7cbf7ff4ee7251a4574f9b42a204484","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-02-15 13:39:15.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"9cae675e8f51fa69bbb9ae91a9a11c6584395cef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-15 15:50:16.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3aca29ed78984b37b30aa1c96a281068 : SUCCESS in 5m 39s\n- openstack-ansible-linters-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e79ac8b15b5a4f6e859d1504a97dd431 : SUCCESS in 10m 40s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4b50f54b91df41acb344240d9c50101c : FAILURE in 1h 13m 09s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/318ad760fa5d40468bd8c1dd44aa5c84 : FAILURE in 1h 51m 14s\n- openstack-ansible-deploy-aio_lxc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1e2ae58ac10b43b293471b5172b85fb9 : FAILURE in 2h 06m 07s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"97bc9c5dae01075f74789afee1ae6b666ecc8d81","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-02-20 14:32:45.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"65f2ee1fcbdebe3248f759ba1c97e26c2e893e7e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-20 17:12:36.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91822fe045084bcb908efbff3194799f : SUCCESS in 8m 42s\n- openstack-ansible-linters-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/41e32ea125284a5a90527cb47d5ef620 : SUCCESS in 14m 04s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bae41013a7044335919be5033cc270be : FAILURE in 1h 25m 05s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/d75ce3ebeab04fda961927e74c96565f : FAILURE in 1h 48m 03s\n- openstack-ansible-deploy-aio_lxc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/abb2b4f29c3648779155bf37e2639f51 : FAILURE in 2h 36m 18s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"e1e492079bda808cd50ce7a09bcb109e0372ba65","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-07-30 20:09:50.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":9},{"id":"d7f9bf45dd991bef39e67a4df9bef190380426f8","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-11-17 09:09:58.000000000","message":"Restored","accounts_in_message":[],"_revision_number":9},{"id":"9e11f32510f1dc67c99da45353f1592fe4b9ce2c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-17 09:10:10.000000000","message":"Patch Set 9:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":9},{"id":"efd807e7b2577ac75698775e938625c49ea78caa","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-11-17 09:13:17.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"dcfe6d978337a97ca93e6d60da1b8ade3a151a1f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-17 11:43:28.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d2168a89d99443193b28f20bafcdf28 : SUCCESS in 6m 44s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/84571eeb7aa14f908bce9574ac1400b7 : SUCCESS in 20m 05s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/153fe9bb4ade493c8894c452371eb2ee : SUCCESS in 1h 48m 21s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/958088043318424c9140f9bbb1700f77 : SUCCESS in 1h 34m 26s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/0a69eeee9727478f99c30320bc92013c : SUCCESS in 2h 01m 58s\n- openstack-ansible-deploy-aio_lxc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f28867b1a69a4d31b085ec2794f3d0e5 : FAILURE in 2h 19m 04s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"c058d2a72e74e09eceacf608b8965f7600b00e91","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-11-17 16:41:34.000000000","message":"Patch Set 10: Code-Review+2","accounts_in_message":[],"_revision_number":10},{"id":"67c4e4d51c980b90e1aab5d46157a6f923815091","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-11-18 13:23:12.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"3112764acc54e699d8ad1ba9df8dde130d1c975b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-18 13:23:23.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"e02a0bd9351efcda2180269c816b7865153db5a7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-18 16:26:18.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00458ce4c91642e0ab0c637bb3a274ab : SUCCESS in 4m 55s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/314a8735757b4e0e94f77cf793d53261 : SUCCESS in 15m 54s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3b273e803ba14242b8103e878a5dc3df : SUCCESS in 2h 58m 15s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/333af5cf82ff4df2a8b216a8417cf40a : SUCCESS in 1h 25m 12s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/7949067e7dda43b18517216b3ccd3129 : SUCCESS in 2h 13m 16s","accounts_in_message":[],"_revision_number":10},{"id":"cccd0d08a1f96006132587a92f36d89d448a248c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-18 16:26:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":10},{"id":"edcbcd51c0ea40e1cdea6eeb781b63e1e087ea1e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-18 16:27:46.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d87c98724124cdfb76674774586b74b : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"a045397c57ea2fd0edd470e7b798f881730079bd","revisions":{"315f14a42349e00c209ad9552c694fa263d71035":{"kind":"REWORK","_number":1,"created":"2020-01-30 14:14:02.000000000","uploader":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"ref":"refs/changes/19/705019/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_manila","ref":"refs/changes/19/705019/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_manila refs/changes/19/705019/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_manila refs/changes/19/705019/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_manila refs/changes/19/705019/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_manila refs/changes/19/705019/1"}}},"commit":{"parents":[{"commit":"ff74834fd5ba82ddc55d664ac91cba13e38c5d95","subject":"Add image download for manila","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_manila/commit/ff74834fd5ba82ddc55d664ac91cba13e38c5d95"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2020-01-30 14:13:40.000000000","tz":60},"committer":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2020-01-30 14:13:40.000000000","tz":60},"subject":"Add Manila key generation and distribution","message":"Add Manila key generation and distribution\n\nChange-Id: I2515dc4abf93f2d30157a376c8a9a497c865aaf5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_manila/commit/315f14a42349e00c209ad9552c694fa263d71035"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_manila/commit/315f14a42349e00c209ad9552c694fa263d71035"}]},"branch":"refs/heads/master"},"4fb95353e143ab5e4a47006b8b5f71b2231c2baf":{"kind":"REWORK","_number":2,"created":"2020-01-30 14:20:57.000000000","uploader":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"ref":"refs/changes/19/705019/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_manila","ref":"refs/changes/19/705019/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_manila refs/changes/19/705019/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_manila refs/changes/19/705019/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_manila refs/changes/19/705019/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_manila refs/changes/19/705019/2"}}},"commit":{"parents":[{"commit":"ff74834fd5ba82ddc55d664ac91cba13e38c5d95","subject":"Add image download for manila","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_manila/commit/ff74834fd5ba82ddc55d664ac91cba13e38c5d95"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2020-01-30 14:13:40.000000000","tz":60},"committer":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2020-01-30 14:20:44.000000000","tz":60},"subject":"Add Manila key generation and distribution","message":"Add Manila key generation and distribution\n\nChange-Id: I2515dc4abf93f2d30157a376c8a9a497c865aaf5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_manila/commit/4fb95353e143ab5e4a47006b8b5f71b2231c2baf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_manila/commit/4fb95353e143ab5e4a47006b8b5f71b2231c2baf"}]},"branch":"refs/heads/master"},"eeb104bbc4c23d04dadf4933d64303bb5aa06951":{"kind":"REWORK","_number":3,"created":"2020-01-30 14:38:25.000000000","uploader":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"ref":"refs/changes/19/705019/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_manila","ref":"refs/changes/19/705019/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_manila refs/changes/19/705019/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_manila refs/changes/19/705019/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_manila refs/changes/19/705019/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_manila refs/changes/19/705019/3"}}},"commit":{"parents":[{"commit":"ff74834fd5ba82ddc55d664ac91cba13e38c5d95","subject":"Add image download for manila","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_manila/commit/ff74834fd5ba82ddc55d664ac91cba13e38c5d95"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2020-01-30 14:13:40.000000000","tz":60},"committer":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2020-01-30 14:38:13.000000000","tz":60},"subject":"Add Manila key generation and distribution","message":"Add Manila key generation and distribution\n\nChange-Id: I2515dc4abf93f2d30157a376c8a9a497c865aaf5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_manila/commit/eeb104bbc4c23d04dadf4933d64303bb5aa06951"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_manila/commit/eeb104bbc4c23d04dadf4933d64303bb5aa06951"}]},"branch":"refs/heads/master"},"ab1da31b9d5da96b0912a82e9365904d858b9579":{"kind":"NO_CHANGE","_number":4,"created":"2020-01-30 17:08:06.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/19/705019/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_manila","ref":"refs/changes/19/705019/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_manila refs/changes/19/705019/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_manila refs/changes/19/705019/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_manila refs/changes/19/705019/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_manila refs/changes/19/705019/4"}}},"commit":{"parents":[{"commit":"859dcd1d2116237cce846e16e6774e171d7101b0","subject":"Add image download for manila","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_manila/commit/859dcd1d2116237cce846e16e6774e171d7101b0"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2020-01-30 14:13:40.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2020-01-30 17:08:06.000000000","tz":0},"subject":"Add Manila key generation and distribution","message":"Add Manila key generation and distribution\n\nChange-Id: I2515dc4abf93f2d30157a376c8a9a497c865aaf5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_manila/commit/ab1da31b9d5da96b0912a82e9365904d858b9579"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_manila/commit/ab1da31b9d5da96b0912a82e9365904d858b9579"}]},"branch":"refs/heads/master"},"9e85717c873570ccb55704b2e7739e5337f87f72":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2020-01-30 18:31:45.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/19/705019/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_manila","ref":"refs/changes/19/705019/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_manila refs/changes/19/705019/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_manila refs/changes/19/705019/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_manila refs/changes/19/705019/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_manila refs/changes/19/705019/5"}}},"commit":{"parents":[{"commit":"0745f4a4131f06ddefd8fe5675d3047890ee2115","subject":"Add image download for manila","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_manila/commit/0745f4a4131f06ddefd8fe5675d3047890ee2115"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2020-01-30 14:13:40.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2020-01-30 18:31:45.000000000","tz":0},"subject":"Add Manila key generation and distribution","message":"Add Manila key generation and distribution\n\nChange-Id: I2515dc4abf93f2d30157a376c8a9a497c865aaf5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_manila/commit/9e85717c873570ccb55704b2e7739e5337f87f72"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_manila/commit/9e85717c873570ccb55704b2e7739e5337f87f72"}]},"branch":"refs/heads/master"},"b4dca970171a454888da29b73e6eb94686278b57":{"kind":"REWORK","_number":6,"created":"2020-01-30 19:05:32.000000000","uploader":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},"ref":"refs/changes/19/705019/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_manila","ref":"refs/changes/19/705019/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_manila refs/changes/19/705019/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_manila refs/changes/19/705019/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_manila refs/changes/19/705019/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_manila refs/changes/19/705019/6"}}},"commit":{"parents":[{"commit":"0745f4a4131f06ddefd8fe5675d3047890ee2115","subject":"Add image download for manila","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_manila/commit/0745f4a4131f06ddefd8fe5675d3047890ee2115"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2020-01-30 14:13:40.000000000","tz":60},"committer":{"name":"Amy Marrich (spotz)","email":"amy@demarco.com","date":"2020-01-30 19:04:20.000000000","tz":0},"subject":"Add Manila key generation and distribution","message":"Add Manila key generation and distribution\n\nChange-Id: I2515dc4abf93f2d30157a376c8a9a497c865aaf5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_manila/commit/b4dca970171a454888da29b73e6eb94686278b57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_manila/commit/b4dca970171a454888da29b73e6eb94686278b57"}]},"branch":"refs/heads/master"},"0da070e77ffa9a9bc8bbfd78cd9df5cfe0c964e8":{"kind":"REWORK","_number":7,"created":"2020-01-31 15:15:03.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/19/705019/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_manila","ref":"refs/changes/19/705019/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_manila refs/changes/19/705019/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_manila refs/changes/19/705019/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_manila refs/changes/19/705019/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_manila refs/changes/19/705019/7"}}},"commit":{"parents":[{"commit":"0745f4a4131f06ddefd8fe5675d3047890ee2115","subject":"Add image download for manila","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_manila/commit/0745f4a4131f06ddefd8fe5675d3047890ee2115"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2020-01-30 14:13:40.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2020-01-31 15:14:57.000000000","tz":0},"subject":"Add Manila key generation and distribution","message":"Add Manila key generation and distribution\n\nChange-Id: I2515dc4abf93f2d30157a376c8a9a497c865aaf5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_manila/commit/0da070e77ffa9a9bc8bbfd78cd9df5cfe0c964e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_manila/commit/0da070e77ffa9a9bc8bbfd78cd9df5cfe0c964e8"}]},"branch":"refs/heads/master"},"507ba551e5c4826b4c60b2a07228ecaa42abd5c7":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2020-02-11 08:04:25.000000000","uploader":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"ref":"refs/changes/19/705019/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_manila","ref":"refs/changes/19/705019/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_manila refs/changes/19/705019/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_manila refs/changes/19/705019/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_manila refs/changes/19/705019/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_manila refs/changes/19/705019/8"}}},"commit":{"parents":[{"commit":"2b29c87527b5b6eadada8e77df0c206d6aca5d47","subject":"Add image download for manila","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_manila/commit/2b29c87527b5b6eadada8e77df0c206d6aca5d47"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2020-01-30 14:13:40.000000000","tz":60},"committer":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2020-02-11 08:04:25.000000000","tz":0},"subject":"Add Manila key generation and distribution","message":"Add Manila key generation and distribution\n\nChange-Id: I2515dc4abf93f2d30157a376c8a9a497c865aaf5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_manila/commit/507ba551e5c4826b4c60b2a07228ecaa42abd5c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_manila/commit/507ba551e5c4826b4c60b2a07228ecaa42abd5c7"}]},"branch":"refs/heads/master"},"0e63ff32153f038edb15a7bd851ac2ce1a0a758a":{"kind":"REWORK","_number":9,"created":"2020-02-20 14:32:45.000000000","uploader":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"ref":"refs/changes/19/705019/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_manila","ref":"refs/changes/19/705019/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_manila refs/changes/19/705019/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_manila refs/changes/19/705019/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_manila refs/changes/19/705019/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_manila refs/changes/19/705019/9"}}},"commit":{"parents":[{"commit":"d8922402d68590b889e9f9a7ce6df8d22d831021","subject":"Add image download for manila","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_manila/commit/d8922402d68590b889e9f9a7ce6df8d22d831021"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2020-02-20 14:32:30.000000000","tz":60},"committer":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2020-02-20 14:32:30.000000000","tz":60},"subject":"Add Manila key generation and distribution","message":"Add Manila key generation and distribution\n\nChange-Id: I2515dc4abf93f2d30157a376c8a9a497c865aaf5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_manila/commit/0e63ff32153f038edb15a7bd851ac2ce1a0a758a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_manila/commit/0e63ff32153f038edb15a7bd851ac2ce1a0a758a"}]},"branch":"refs/heads/master"},"a045397c57ea2fd0edd470e7b798f881730079bd":{"kind":"REWORK","_number":10,"created":"2020-11-17 09:13:17.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/19/705019/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_manila","ref":"refs/changes/19/705019/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_manila refs/changes/19/705019/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_manila refs/changes/19/705019/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_manila refs/changes/19/705019/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_manila refs/changes/19/705019/10"}}},"commit":{"parents":[{"commit":"3a7d20720a32db032e9feb222eeea0580e506d84","subject":"Replace deprecated UPPER_CONSTRAINTS_FILE variable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_manila/commit/3a7d20720a32db032e9feb222eeea0580e506d84"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2020-02-20 14:32:30.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2020-11-17 09:13:01.000000000","tz":120},"subject":"Add Manila key generation and distribution","message":"Add Manila key generation and distribution\n\nChange-Id: I2515dc4abf93f2d30157a376c8a9a497c865aaf5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_manila/commit/a045397c57ea2fd0edd470e7b798f881730079bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_manila/commit/a045397c57ea2fd0edd470e7b798f881730079bd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
