)]}'
{"id":"openstack%2Fopenstack-ansible-os_manila~729513","triplet_id":"openstack%2Fopenstack-ansible-os_manila~master~I8a77e5e6d7f178f1f6926b0c1e04c8167cb7c86f","project":"openstack/openstack-ansible-os_manila","branch":"master","hashtags":[],"change_id":"I8a77e5e6d7f178f1f6926b0c1e04c8167cb7c86f","subject":"Fix pygments style","status":"ABANDONED","created":"2020-05-20 06:14:00.000000000","updated":"2020-11-06 12:57:34.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ff277e29008980bddb3c017e83af051dafe2fd4e","_number":729513,"virtual_id_number":729513,"owner":{"_account_id":31446,"name":"fuzihao","email":"fuzihao@inspur.com","username":"fuzihao"},"actions":{},"labels":{"Verified":{"all":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":31446,"name":"fuzihao","email":"fuzihao@inspur.com","username":"fuzihao"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":31446,"name":"fuzihao","email":"fuzihao@inspur.com","username":"fuzihao"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":31446,"name":"fuzihao","email":"fuzihao@inspur.com","username":"fuzihao"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":31446,"name":"fuzihao","email":"fuzihao@inspur.com","username":"fuzihao"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":31446,"name":"fuzihao","email":"fuzihao@inspur.com","username":"fuzihao"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-20 06:27:45.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-05-20 06:29:14.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-10-19 11:25:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"56b459b16b5ccc611c8db95bdfe06fa308c2bd42","author":{"_account_id":31446,"name":"fuzihao","email":"fuzihao@inspur.com","username":"fuzihao"},"date":"2020-05-20 06:14:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"154911a8ad64b88dc000e803f86dbe9899f22f6c","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-20 06:19:34.000000000","message":"Patch Set 1:\n\nWhy is this needed? What problem does it address?","accounts_in_message":[],"_revision_number":1},{"id":"18e296a2770560ed7ff56edccbc595b86c2404d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-20 06:25:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f5551cc093e4b1abc124503fc5b4cac : SUCCESS in 8m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bbe535b7d754450d8d6a20cf78b15b68 : SUCCESS in 5m 29s","accounts_in_message":[],"_revision_number":1},{"id":"4ff688f0bdadb9456cf871650812098033fa8dbd","author":{"_account_id":31446,"name":"fuzihao","email":"fuzihao@inspur.com","username":"fuzihao"},"date":"2020-05-20 06:26:05.000000000","message":"Patch Set 1:\n\n\u003e Why is this needed? What problem does it address?\n\nNew theme of docs (Victoria+) respects pygments_style.\nSince we starts using Victoria reqs while being on Ussuri,\nthis patch ensures proper rendering both in Ussuri and Victoria.","accounts_in_message":[],"_revision_number":1},{"id":"89c5232fcd879d96f85bc5fbba970d0bf6424c49","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-20 06:27:45.000000000","message":"Patch Set 1:\n\n\u003e \u003e Why is this needed? What problem does it address?\n \u003e \n \u003e New theme of docs (Victoria+) respects pygments_style.\n \u003e Since we starts using Victoria reqs while being on Ussuri,\n \u003e this patch ensures proper rendering both in Ussuri and Victoria.\n\nDid you see that the project needs this? IS it fixing anything?","accounts_in_message":[],"_revision_number":1},{"id":"1360ed4e2764631dabf387bcfe3d4968beaf53e8","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-05-20 06:29:14.000000000","message":"Patch Set 1: Code-Review-1\n\nCan you please provide commit message explaining why this is needed (maybe reference a relevant mailing list post?) There is documentation of good commit messages here https://wiki.openstack.org/wiki/GitCommitMessages.\n\nI also looked with codesearch.openstack.org and it seems the whole of openstack with the exception of Kolla is using the sphinx style so a better explanation of this patch is required.","accounts_in_message":[],"_revision_number":1},{"id":"dd7dc8f0a206991c4254c3a37c9011df853f8a4a","author":{"_account_id":31446,"name":"fuzihao","email":"fuzihao@inspur.com","username":"fuzihao"},"date":"2020-05-20 06:45:36.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"2e2132f3c541c40728613d425b4d921e9c2fe765","author":{"_account_id":31446,"name":"fuzihao","email":"fuzihao@inspur.com","username":"fuzihao"},"date":"2020-05-20 06:46:09.000000000","message":"Topic f307563283c3b37da968446c3f923005ff3852bc removed","accounts_in_message":[],"_revision_number":2},{"id":"9a13d037b900d6ed408220479bbd3663dbadde82","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-20 07:06:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b000f9531e04449e9ee67da5dccc1ed5 : SUCCESS in 11m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c26f7683cc4f40c586fe795a04991e61 : SUCCESS in 6m 49s","accounts_in_message":[],"_revision_number":2},{"id":"22f95ebd1a6a3e54d0971b4633dc5500abcb48b7","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-19 08:47:29.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"4ffdde3bd26a9de4782ab14f7e98fdbad8cdbd95","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 11:25:22.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/80d211f121d647619608a4f4018a430e : SUCCESS in 5m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67c588100eb54cffb0beed29d72a03b0 : SUCCESS in 6m 37s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d63452891cb2448096fc4510e0165d5c : SUCCESS in 18m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d066b2a9a274450eb048b4bc39295cb5 : SUCCESS in 5m 50s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b8ab3ff1ab514bd9afa348b54200915e : FAILURE in 1h 56m 36s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/d96b35e6232942b79b6b3bc8e9b1ab16 : RETRY_LIMIT in 13m 25s\n- openstack-ansible-deploy-aio_lxc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/889e048fd64242e28ff27ff002ff2a43 : FAILURE in 2h 28m 41s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"fb202b2a025c315a9a4a930a33f79caac286e6e8","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-11-06 12:57:16.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"ff277e29008980bddb3c017e83af051dafe2fd4e","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-11-06 12:57:34.000000000","message":"Abandoned\n\nhas been already implemented, sorry","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"1ecd1dba6c56fa530b85258a7184dd3d83c7974f","revisions":{"f307563283c3b37da968446c3f923005ff3852bc":{"kind":"REWORK","_number":1,"created":"2020-05-20 06:14:00.000000000","uploader":{"_account_id":31446,"name":"fuzihao","email":"fuzihao@inspur.com","username":"fuzihao"},"ref":"refs/changes/13/729513/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_manila","ref":"refs/changes/13/729513/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_manila refs/changes/13/729513/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_manila refs/changes/13/729513/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_manila refs/changes/13/729513/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_manila refs/changes/13/729513/1"}}},"commit":{"parents":[{"commit":"add53dbf8730d1f2ef3193125db3bb4b89b5faa8","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_manila/commit/add53dbf8730d1f2ef3193125db3bb4b89b5faa8"}]}],"author":{"name":"fuzihao","email":"fuzihao@inspur.com","date":"2020-05-20 06:13:27.000000000","tz":480},"committer":{"name":"fuzihao","email":"fuzihao@inspur.com","date":"2020-05-20 06:13:27.000000000","tz":480},"subject":"Fix pygments style","message":"Fix pygments style\n\nChange-Id: I8a77e5e6d7f178f1f6926b0c1e04c8167cb7c86f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_manila/commit/f307563283c3b37da968446c3f923005ff3852bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_manila/commit/f307563283c3b37da968446c3f923005ff3852bc"}]},"branch":"refs/heads/master"},"96db540b17a9f05267b274a84ee71312f72a6cd1":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-05-20 06:45:36.000000000","uploader":{"_account_id":31446,"name":"fuzihao","email":"fuzihao@inspur.com","username":"fuzihao"},"ref":"refs/changes/13/729513/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_manila","ref":"refs/changes/13/729513/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_manila refs/changes/13/729513/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_manila refs/changes/13/729513/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_manila refs/changes/13/729513/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_manila refs/changes/13/729513/2"}}},"commit":{"parents":[{"commit":"add53dbf8730d1f2ef3193125db3bb4b89b5faa8","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_manila/commit/add53dbf8730d1f2ef3193125db3bb4b89b5faa8"}]}],"author":{"name":"fuzihao","email":"fuzihao@inspur.com","date":"2020-05-20 06:13:27.000000000","tz":480},"committer":{"name":"fuzihao","email":"fuzihao@inspur.com","date":"2020-05-20 06:44:55.000000000","tz":480},"subject":"Fix pygments style","message":"Fix pygments style\n\nNew theme of docs (Victoria+) respects pygments_style.\nSince we starts using Victoria reqs while being on Ussuri,\nthis patch ensures proper rendering both in Ussuri and Victoria.\n\nChange-Id: I8a77e5e6d7f178f1f6926b0c1e04c8167cb7c86f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_manila/commit/96db540b17a9f05267b274a84ee71312f72a6cd1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_manila/commit/96db540b17a9f05267b274a84ee71312f72a6cd1"}]},"branch":"refs/heads/master"},"dba66d8bedfa201f3e47f2d653f5af676d438325":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-10-19 08:47:29.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/13/729513/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_manila","ref":"refs/changes/13/729513/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_manila refs/changes/13/729513/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_manila refs/changes/13/729513/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_manila refs/changes/13/729513/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_manila refs/changes/13/729513/3"}}},"commit":{"parents":[{"commit":"4e6dbd907dafbef511ba8572808bcd1748994dc9","subject":"Fix linter errors","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_manila/commit/4e6dbd907dafbef511ba8572808bcd1748994dc9"}]}],"author":{"name":"fuzihao","email":"fuzihao@inspur.com","date":"2020-05-20 06:13:27.000000000","tz":480},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2020-10-19 08:47:29.000000000","tz":0},"subject":"Fix pygments style","message":"Fix pygments style\n\nNew theme of docs (Victoria+) respects pygments_style.\nSince we starts using Victoria reqs while being on Ussuri,\nthis patch ensures proper rendering both in Ussuri and Victoria.\n\nChange-Id: I8a77e5e6d7f178f1f6926b0c1e04c8167cb7c86f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_manila/commit/dba66d8bedfa201f3e47f2d653f5af676d438325"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_manila/commit/dba66d8bedfa201f3e47f2d653f5af676d438325"}]},"branch":"refs/heads/master"},"1ecd1dba6c56fa530b85258a7184dd3d83c7974f":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-11-06 12:57:16.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/13/729513/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_manila","ref":"refs/changes/13/729513/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_manila refs/changes/13/729513/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_manila refs/changes/13/729513/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_manila refs/changes/13/729513/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_manila refs/changes/13/729513/4"}}},"commit":{"parents":[{"commit":"1838f95d3389684f04dc71cf9ad112b988f3307d","subject":"Merge \"Add Ubuntu Focal CI jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_manila/commit/1838f95d3389684f04dc71cf9ad112b988f3307d"}]}],"author":{"name":"fuzihao","email":"fuzihao@inspur.com","date":"2020-05-20 06:13:27.000000000","tz":480},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2020-11-06 12:57:16.000000000","tz":0},"subject":"Fix pygments style","message":"Fix pygments style\n\nNew theme of docs (Victoria+) respects pygments_style.\nSince we starts using Victoria reqs while being on Ussuri,\nthis patch ensures proper rendering both in Ussuri and Victoria.\n\nChange-Id: I8a77e5e6d7f178f1f6926b0c1e04c8167cb7c86f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_manila/commit/1ecd1dba6c56fa530b85258a7184dd3d83c7974f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_manila/commit/1ecd1dba6c56fa530b85258a7184dd3d83c7974f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
