)]}'
{"id":"openstack%2Fopenstack-ansible-os_manila~754653","triplet_id":"openstack%2Fopenstack-ansible-os_manila~master~I1cee8475e679b021185e94ba5c7d5f3faa92eb31","project":"openstack/openstack-ansible-os_manila","branch":"master","hashtags":[],"change_id":"I1cee8475e679b021185e94ba5c7d5f3faa92eb31","subject":"Add Ubuntu Focal CI jobs","status":"MERGED","created":"2020-09-28 08:25:54.000000000","updated":"2020-11-06 10:59:29.000000000","submitted":"2020-11-06 10:55:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"754653-1604660100844-3f776333","meta_rev_id":"72668f746c2a714dd24cc4d25db73584165e3081","_number":754653,"virtual_id_number":754653,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-06 10:55:00.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2020-11-02 10:22:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":2,"date":"2020-11-02 16:04:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":1,"date":"2020-11-06 08:15:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-02 10:22:45.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2020-11-06 08:15:38.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2020-11-06 10:55:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3b57bad18a25d8ec9ddf1557a2d78934dd18c2c1","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-09-28 08:25:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c769ab91e647c417f671189396bb799a00d4ae38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-28 11:06:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93f7598808f84d798ab1980ee32ccb05 : SUCCESS in 5m 15s\n- openstack-ansible-linters-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/11d3cc5a7bb441d2a778bf7d0c8fd8d1 : SUCCESS in 14m 47s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9773906aa3604a9fa778258d992c076b : FAILURE in 1h 28m 22s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9fd76ac658074d14a4840ff0388378d6 : FAILURE in 32m 40s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/3a7903f760964c2aac6b4db34266fe88 : FAILURE in 1h 43m 33s\n- openstack-ansible-deploy-aio_lxc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f231fa19790a469ab8ea7dec7acff608 : FAILURE in 2h 22m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1e3428e69ea4c2c64dd4b6f616d2eb452aec13bf","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-10-13 19:12:48.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"77ca3290827bd02e4cc0337482e0d47c264d24c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-13 21:43:20.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1036e0d6fc9d4eef88afeec37c1e4a22 : SUCCESS in 6m 16s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e6d3654f85fe4d1990863d924baeda2a : SUCCESS in 15m 28s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/95e8ad2d38dd42eba8ab0a962c9924d3 : FAILURE in 1h 39m 38s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e68209aae3dc4329bbd0cc8133bff332 : FAILURE in 1h 26m 50s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/3eacb19f72af4dd797eee2ce075ef164 : FAILURE in 1h 55m 12s\n- openstack-ansible-deploy-aio_lxc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/48cdb7969ae34bdb8b87a72751e1b5a3 : FAILURE in 2h 16m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"77a1382502698d5c61a6e9dfd1710dfa04194078","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-14 07:34:29.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"c55444cd76a0fa632361e8e30b72a888875dd352","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-14 07:34:59.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"d9049f4ed1a915f13603fb9418aa32c02710fa29","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-14 10:18:16.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8756b14e62654fcc923890ad51a4a613 : SUCCESS in 6m 32s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/55539cc4a8d1421fbead0af37b449d1e : SUCCESS in 22m 41s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a99be84eae614088bf44b0947b352fa6 : SUCCESS in 1h 39m 28s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/711e0a9ce40b44418f264bb60cceef38 : FAILURE in 1h 27m 25s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/5e06ebd0bc9a459584d9423c32892d8f : FAILURE in 1h 58m 57s\n- openstack-ansible-deploy-aio_lxc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/aa375428e24d4efdbdfe108394119d18 : FAILURE in 2h 30m 55s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d30414a0d5e8689e87e27498e28671aa99ce64fa","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-23 11:18:01.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"d2099bf06b2db60745fba658efbf1b8db38695fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-23 13:33:17.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/879025e475e74499b81ae325adb27ed9 : SUCCESS in 7m 39s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/74b8f32147dc4aad804f464a4998c98b : SUCCESS in 18m 36s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7d8cb69e331146cb8434d5a34fa0bc14 : SUCCESS in 1h 52m 09s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/23777dd25db6472ba9d0803496e65c3a : FAILURE in 1h 44m 04s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/4fe7bea750264c38945206381afb8b06 : SUCCESS in 2h 09m 24s\n- openstack-ansible-deploy-aio_lxc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/454a3cc0881b41e9beaec814abe45970 : FAILURE in 20m 34s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8f8a1d728b9ecc39add10cd871308595f2be5a1b","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-23 13:36:36.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"8b40309b4c048fc16abe5fe6e2fcec6b9f21f06e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-23 15:57:52.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4326038405a14af8b3d5382b0e685692 : SUCCESS in 7m 11s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dad7395990f54b8693e16f6345fc1c70 : SUCCESS in 18m 00s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4953eba236074e9e80b4b765a876e3d9 : SUCCESS in 2h 04m 23s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4eb82a10d57b41ab86f12e9e912afbd7 : FAILURE in 1h 41m 25s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/88999c9db72046268cc33b2bb4a25d03 : FAILURE in 2h 12m 58s\n- openstack-ansible-deploy-aio_lxc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/70992fdc83384bc996d7991807497a0a : FAILURE in 23m 02s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"9cf420473c93c00df8b9156ad9d01fdcda7c4870","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-23 17:31:58.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"e659872e5b86f26999f57e9630275f802423984c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-23 20:48:16.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f1a7c57758f4fb19f6c173e4913742e : SUCCESS in 6m 59s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/97784ff2bb1d4972a1022fbf89f6b602 : SUCCESS in 20m 26s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/924652ea90044971b9a609e886549d9f : SUCCESS in 1h 53m 48s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5cbf7eece2484ed28e43b3b133346687 : TIMED_OUT in 3h 05m 02s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/dd13dc8e727b4ef4a995b59d4df7c3de : SUCCESS in 2h 18m 39s\n- openstack-ansible-deploy-aio_lxc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e24fa71c91254071a700a6467b3875ad : FAILURE in 21m 50s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"502d6c0823171b5e7e6ff474009490967b9158dc","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-26 08:31:30.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"eca84a9bf89e2de377a308b3ecc99f31f1f955e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-26 10:53:34.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86b0041bceed47369455cf3be1fdfa45 : SUCCESS in 4m 57s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4fb9e23964d643de8b5c5d471c55c005 : SUCCESS in 17m 12s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b0c073d34b03480b822bae4c3b848b1e : FAILURE in 2h 08m 37s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dedbc7602de24402ba79d44b8110cb9a : SUCCESS in 2h 00m 04s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/4124ddc06e224ca587e755bf4e13e064 : SUCCESS in 2h 02m 54s\n- openstack-ansible-deploy-aio_lxc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9bf7e80ee9044631923e4d9990bb201b : FAILURE in 18m 46s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ae7b95afe7a9184bbf4e81b0dfef18096977a3c7","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-26 11:11:14.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"ad04f3899494731345d54a81fdabd2e806440b5c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-26 14:28:37.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5dab58041c1d4950bfdaa3d0cc2c337c : SUCCESS in 5m 17s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/de3605514d9347bbb69eb3bb22938c07 : SUCCESS in 13m 36s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/631ab4d5cd80413a9edf0b2dd097d87b : TIMED_OUT in 3h 10m 28s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/56928ec682f147499692b296d3838fb6 : TIMED_OUT in 3h 11m 45s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/8ca7b81e07934c009baf346162695c9c : FAILURE in 2h 12m 34s\n- openstack-ansible-deploy-aio_lxc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/eb4d792858de4dd6ae41e6ec74b51c28 : FAILURE in 16m 59s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f8959aad26eecf79861ccedea7731fee5e477034","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-28 09:03:57.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"069cbb904ea7d7582dd414690e772cdbb5c48726","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-28 11:44:58.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2611acda54c4d57ae146383e46c65a6 : SUCCESS in 5m 32s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/93e614a67a5742e68fc0f9f8127a4e93 : SUCCESS in 30m 42s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dd7d3af8a3884b6a9653da0e54e057bd : SUCCESS in 2h 00m 56s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/98fabfe2453c431da9e7e420c17f485d : SUCCESS in 1h 38m 45s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/132f2a5b5bc64f31aee7d3158926513e : FAILURE in 2h 04m 41s\n- openstack-ansible-deploy-aio_lxc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e955d4134635487cbe0acf8183388d7c : FAILURE in 2h 27m 42s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"045501a8f56747e6f132046fd47e559cf50ce48a","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-11-02 07:27:33.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"d9e912594460f4d0e3bdef40bcf8d6b0ed9bbb18","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-02 10:19:36.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f07328389a194419b78ea0a0771b78cc : SUCCESS in 5m 59s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e0ec52bcdbe24bfaa28fa43c8bd3a6ee : SUCCESS in 21m 21s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/eaaa74fed7ee41cd9b54ba6e82cd3f9e : SUCCESS in 1h 51m 02s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/95a6b577559d490e8a17fde5839c8566 : SUCCESS in 1h 34m 57s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/641558fb95664bfd98905510de2d7a65 : SUCCESS in 1h 55m 44s\n- openstack-ansible-deploy-aio_lxc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ad43d4a0f88b42b9ac5a93bf07312479 : FAILURE in 2h 37m 31s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"74a283d85ac22c826bd1425783664e86040792b5","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2020-11-02 10:22:45.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"d7e8800eaa23db5adc741f26afb5b1c79e02d3f5","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-11-02 16:04:21.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"05c6317290e63cca7cbba4174c2e68f09872e66f","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-11-06 08:15:35.000000000","message":"Removed Workflow+1 by Dmitriy Rabotyagov (noonedeadpunk) \u003cnoonedeadpunk@ya.ru\u003e\n","accounts_in_message":[],"_revision_number":4},{"id":"9c99f6508699186dce40a2c1afe30223c5c75fd1","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-11-06 08:15:38.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"7eb49d88b4f0dd172c2e2a8525bf93888ae68100","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-06 08:15:52.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"aacc18bcba9661ea09ecf45249cd48695d3210b6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-06 10:55:00.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a31d92d0d1a549ae9c9c88d76f92c91e : SUCCESS in 6m 04s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c65debdf40c64230ad44683b294a55a6 : SUCCESS in 16m 29s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7a68493afd2040db9a70c16b1f54bc4b : SUCCESS in 1h 45m 07s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/98ecdc165d774fc0a1a225ffe92d0f1a : SUCCESS in 1h 40m 04s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/331cad9743484e108e1872910b3419a2 : SUCCESS in 2h 23m 43s","accounts_in_message":[],"_revision_number":4},{"id":"b12a5648d48dd7b31486b97cc7a9e4704b470a09","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-06 10:55:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"72668f746c2a714dd24cc4d25db73584165e3081","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-06 10:59:29.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3620c9b89db54e32ae6d6c8ff3210db2 : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"105a4f79cd058b99df1cb549b515701a5b5307cf","revisions":{"d386d7c12832e4ff363986bbc917d93cb3534e98":{"kind":"REWORK","_number":1,"created":"2020-09-28 08:25:54.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/53/754653/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_manila","ref":"refs/changes/53/754653/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_manila refs/changes/53/754653/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_manila refs/changes/53/754653/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_manila refs/changes/53/754653/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_manila refs/changes/53/754653/1"}}},"commit":{"parents":[{"commit":"90920ec7d86ec2e15e79204eea2cead8413623ff","subject":"Switch CI job for centos to centos-8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_manila/commit/90920ec7d86ec2e15e79204eea2cead8413623ff"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-09-28 08:25:27.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-09-28 08:25:27.000000000","tz":60},"subject":"Add Ubuntu Focal CI jobs","message":"Add Ubuntu Focal CI jobs\n\nChange-Id: I1cee8475e679b021185e94ba5c7d5f3faa92eb31\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_manila/commit/d386d7c12832e4ff363986bbc917d93cb3534e98"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_manila/commit/d386d7c12832e4ff363986bbc917d93cb3534e98"}]},"branch":"refs/heads/master"},"a221d1f161d666056501c90d352441dc763c4b1f":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-10-14 07:34:29.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/53/754653/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_manila","ref":"refs/changes/53/754653/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_manila refs/changes/53/754653/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_manila refs/changes/53/754653/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_manila refs/changes/53/754653/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_manila refs/changes/53/754653/2"}}},"commit":{"parents":[{"commit":"80ea46f075935bd890fd0dc76d3ad475e1d60aed","subject":"Add centos-8 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_manila/commit/80ea46f075935bd890fd0dc76d3ad475e1d60aed"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-09-28 08:25:27.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2020-10-14 07:34:29.000000000","tz":0},"subject":"Add Ubuntu Focal CI jobs","message":"Add Ubuntu Focal CI jobs\n\nChange-Id: I1cee8475e679b021185e94ba5c7d5f3faa92eb31\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_manila/commit/a221d1f161d666056501c90d352441dc763c4b1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_manila/commit/a221d1f161d666056501c90d352441dc763c4b1f"}]},"branch":"refs/heads/master"},"fd76027b162b67589df580a62d04e4cbc7efdd11":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-10-23 11:18:01.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/53/754653/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_manila","ref":"refs/changes/53/754653/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_manila refs/changes/53/754653/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_manila refs/changes/53/754653/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_manila refs/changes/53/754653/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_manila refs/changes/53/754653/3"}}},"commit":{"parents":[{"commit":"9d6e0100060c014aa60f50df78f4aafb0ae6eef7","subject":"Merge \"Add centos-8 support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_manila/commit/9d6e0100060c014aa60f50df78f4aafb0ae6eef7"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-09-28 08:25:27.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2020-10-23 11:18:01.000000000","tz":0},"subject":"Add Ubuntu Focal CI jobs","message":"Add Ubuntu Focal CI jobs\n\nChange-Id: I1cee8475e679b021185e94ba5c7d5f3faa92eb31\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_manila/commit/fd76027b162b67589df580a62d04e4cbc7efdd11"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_manila/commit/fd76027b162b67589df580a62d04e4cbc7efdd11"}]},"branch":"refs/heads/master"},"105a4f79cd058b99df1cb549b515701a5b5307cf":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2020-11-02 07:27:33.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/53/754653/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_manila","ref":"refs/changes/53/754653/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_manila refs/changes/53/754653/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_manila refs/changes/53/754653/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_manila refs/changes/53/754653/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_manila refs/changes/53/754653/4"}}},"commit":{"parents":[{"commit":"9d6e0100060c014aa60f50df78f4aafb0ae6eef7","subject":"Merge \"Add centos-8 support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_manila/commit/9d6e0100060c014aa60f50df78f4aafb0ae6eef7"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-09-28 08:25:27.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2020-11-02 07:27:30.000000000","tz":0},"subject":"Add Ubuntu Focal CI jobs","message":"Add Ubuntu Focal CI jobs\n\nDepends-On: https://review.opendev.org/760771\nChange-Id: I1cee8475e679b021185e94ba5c7d5f3faa92eb31\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_manila/commit/105a4f79cd058b99df1cb549b515701a5b5307cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_manila/commit/105a4f79cd058b99df1cb549b515701a5b5307cf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
