)]}'
{"id":"openstack%2Fopenstack-ansible-os_manila~769944","triplet_id":"openstack%2Fopenstack-ansible-os_manila~master~Ie2c30dcc3640361b91e602125e98996c897f0f06","project":"openstack/openstack-ansible-os_manila","branch":"master","topic":"osa/service_variables","hashtags":[],"change_id":"Ie2c30dcc3640361b91e602125e98996c897f0f06","subject":"Use global service variables","status":"MERGED","created":"2021-01-08 15:35:41.000000000","updated":"2021-01-21 16:52:11.000000000","submitted":"2021-01-21 16:45:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"769944-osa/service_variables","meta_rev_id":"f538aa0e449fb46ffd78134422828365f6491f2e","_number":769944,"virtual_id_number":769944,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-21 16:45:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2021-01-21 08:55:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-01-18 22:35:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2021-01-21 08:55:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-08 18:25:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-08 18:42:07.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2021-01-11 08:31:36.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"CC"},{"updated":"2021-01-21 08:55:22.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"b5b69c14880551cd8d635859abd4ec9125dbbca6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-01-08 15:35:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"691f7dfe74177e0039946546060fd2f57192e4e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-08 18:25:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18b39aa86e94409e953f23ebd0244b6d : SUCCESS in 6m 48s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b0234f4897884b19b97dc0549eaf9298 : SUCCESS in 14m 38s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b18edf1f81d94d7182f4a9b42101f96f : SUCCESS in 1h 49m 09s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d74516ba616143f4834a00dd6a935723 : SUCCESS in 1h 28m 21s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/503ff279398549b6a20913491fdcf034 : SUCCESS in 2h 02m 25s\n- openstack-ansible-deploy-aio_lxc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9e2ce6adee2246048480083229c7bed7 : FAILURE in 2h 37m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1a4a2f8a719bd46d833f697fb0429bca7d8c596c","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-01-08 18:42:07.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0a0f285225836a023f5004cb9b40387ed5a17095","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2021-01-11 08:31:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6893e4e0e22658a68cce607099d3dc483a2d761c","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-01-18 09:20:54.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"de0d9edb702fd3f48a7cde3dfc280e7223d15de1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-01-18 09:36:52.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5728a4b2e9133e3ec36fecb8e300428789c86bef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-18 10:17:45.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f83cd8ebc98472dad2e42065b02286d : SUCCESS in 5m 08s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/00716d2265ba4d4891f06a910481bc2f : SUCCESS in 16m 45s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/543d43fc6acb4e75bbf8e05cc544d08c : FAILURE in 29m 42s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/36041e7fef6e425dbc556156bf7eddd4 : FAILURE in 26m 50s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/44025fa7334f4719b0cf65e7a3fc0635 : FAILURE in 29m 04s\n- openstack-ansible-deploy-aio_lxc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ea251a53f62b4787a8f36efef79afb43 : FAILURE in 28m 36s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ccb353f2de30027e6617035a715849fd9ea5928a","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-01-18 17:24:14.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"a1b5ed88ba4a23eb3c9cd2d70f7527e003bfa408","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-18 20:15:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21faa057514e420ba51df521ebbca15f : SUCCESS in 7m 40s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/61622c45ff2548b899ac211890a8ba0a : SUCCESS in 15m 59s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c9e8d64f74384090bf5f4eb9da0a2afb : SUCCESS in 1h 45m 13s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e7b93c71797f468ea435e6fa22855d7b : SUCCESS in 1h 33m 38s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/7410549dc30a4014a7b25ca5876916f7 : SUCCESS in 2h 11m 25s\n- openstack-ansible-deploy-aio_lxc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c9939dbdc5394cb4bac203aac99640c1 : FAILURE in 2h 33m 49s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"da824d0a2a32039a05fcb93876d2fac23e087cdc","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-01-18 22:35:12.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"91d0078fd925f8585b3fc0572f844cdcaf298780","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2021-01-21 08:55:22.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"253206f922e074aeedd3c227014ab0a836931d60","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 08:55:46.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"75aba5dae7823b78b69a20070d863ff5098f5e1c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 11:16:16.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c597804d51f487a9d9beb9d9153f0bb : SUCCESS in 5m 48s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4968aa7acf0146e1835046f3a7a8557c : SUCCESS in 17m 03s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/69e15d84b0bd4d41832a49f2245b0b23 : FAILURE in 38m 29s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f17421bf45ce46ac92ee421db17d4234 : FAILURE in 1h 00m 26s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/afaeb9db0fd042bf915041a3d90b8d94 : SUCCESS in 2h 10m 58s","accounts_in_message":[],"_revision_number":2},{"id":"10d64f979a7cd571d0abfde3fe09922ad8992dab","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-01-21 11:35:39.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"984ed36f17c15c7b85f06b657013ab60ad73ba3e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 14:33:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b649d9d3e257433c9b07d22c25992c27 : SUCCESS in 5m 51s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8997489e08754200a7291071a6877f8d : SUCCESS in 15m 10s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/70dc78f322b3490cbec049358410493b : SUCCESS in 2h 06m 42s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d8337eaaa5ea4d3187388ebbae393b2c : SUCCESS in 1h 42m 20s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/3643895810034da38cf7fb3a4b662f28 : SUCCESS in 2h 12m 45s\n- openstack-ansible-deploy-aio_lxc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e3349779f38c4eefbeaaceb2f062f7ab : FAILURE in 2h 39m 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6b6ef4b29d5efe98074cedddebf9ea62c177f803","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 14:33:52.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c2af297872194d23ebf5b46e0ca5f324c29b3fd6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 16:45:06.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43e22aceb46644349def062fd1ed232d : SUCCESS in 8m 24s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4cba4d85e37a4a648843e7bcdebb88e5 : SUCCESS in 22m 24s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d7f180dba5b948daba711ffff52c63be : SUCCESS in 1h 54m 24s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cdaf930ea6ff4b6d810e0944bb1e831b : SUCCESS in 1h 31m 18s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/8a642fcd3dce4b50a2f41d8b057a8894 : SUCCESS in 1h 59m 45s","accounts_in_message":[],"_revision_number":2},{"id":"0d57869dbe1b73486b55a376f2df7af3fb027c4a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 16:45:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"f538aa0e449fb46ffd78134422828365f6491f2e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 16:52:11.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10598f8a3c1446a1a09a6a8d552dc4d6 : SUCCESS in 2m 28s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b5aed5c1301f60aea8fe34daefed8a6cb9572f14","revisions":{"87496d95de7b569f96b3eff0e5a099342372bd1f":{"kind":"REWORK","_number":1,"created":"2021-01-08 15:35:41.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/44/769944/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_manila","ref":"refs/changes/44/769944/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_manila refs/changes/44/769944/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_manila refs/changes/44/769944/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_manila refs/changes/44/769944/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_manila refs/changes/44/769944/1"}}},"commit":{"parents":[{"commit":"dfc29c7bac76f4523e6fb29138629d4212d40266","subject":"Merge \"Add Manila key generation and distribution\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_manila/commit/dfc29c7bac76f4523e6fb29138629d4212d40266"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2021-01-08 15:35:29.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2021-01-08 15:35:29.000000000","tz":120},"subject":"Use global service variables","message":"Use global service variables\n\nInstead of overriding each service separatelly it might make\nsense for deployers to define some higher level variable that\nwill be used first or fallback to default variable.\n\nChange-Id: Ie2c30dcc3640361b91e602125e98996c897f0f06\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_manila/commit/87496d95de7b569f96b3eff0e5a099342372bd1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_manila/commit/87496d95de7b569f96b3eff0e5a099342372bd1f"}]},"branch":"refs/heads/master"},"b5aed5c1301f60aea8fe34daefed8a6cb9572f14":{"kind":"REWORK","_number":2,"created":"2021-01-18 09:36:52.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/44/769944/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_manila","ref":"refs/changes/44/769944/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_manila refs/changes/44/769944/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_manila refs/changes/44/769944/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_manila refs/changes/44/769944/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_manila refs/changes/44/769944/2"}}},"commit":{"parents":[{"commit":"dfc29c7bac76f4523e6fb29138629d4212d40266","subject":"Merge \"Add Manila key generation and distribution\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_manila/commit/dfc29c7bac76f4523e6fb29138629d4212d40266"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2021-01-08 15:35:29.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2021-01-18 09:36:42.000000000","tz":120},"subject":"Use global service variables","message":"Use global service variables\n\nInstead of overriding each service separatelly it might make\nsense for deployers to define some higher level variable that\nwill be used first or fallback to default variable.\n\nChange-Id: Ie2c30dcc3640361b91e602125e98996c897f0f06\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_manila/commit/b5aed5c1301f60aea8fe34daefed8a6cb9572f14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_manila/commit/b5aed5c1301f60aea8fe34daefed8a6cb9572f14"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
