)]}'
{"id":"openstack%2Fopenstack-ansible-os_manila~810237","triplet_id":"openstack%2Fopenstack-ansible-os_manila~master~I94b3f8ba5116cdfb94e9d0dc575bd7edb1d27b3c","project":"openstack/openstack-ansible-os_manila","branch":"master","topic":"osa/galera_pki","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2021-12-06 18:53:07.000000000","reason":"Dmitriy Rabotyagov replied on the change"}},"hashtags":[],"change_id":"I94b3f8ba5116cdfb94e9d0dc575bd7edb1d27b3c","subject":"Refactor galera_use_ssl behaviour","status":"MERGED","created":"2021-09-21 13:27:05.000000000","updated":"2021-12-06 23:30:57.000000000","submitted":"2021-12-06 23:29:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"810237-osa/galera_pki","meta_rev_id":"12735a2fad2fb522758c1228575ec9babb71c0dd","_number":810237,"virtual_id_number":810237,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-06 23:29:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":2,"date":"2021-12-06 18:53:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":1,"date":"2021-12-06 18:53:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-21 16:01:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-06 11:38:44.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2021-12-06 12:22:47.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"95f1daa00066c73cd31c413b3eb67e8f1ab19110","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-09-21 13:27:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9fdfa0c24a82734f4e2d13c6159604fd5a65a8a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-21 16:01:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/030ee20f540b4fddb49d9da34f79593f : SUCCESS in 7m 11s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ed26fabdda8d499ebd97ceada9f21162 : SUCCESS in 15m 44s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eb27480ca85041d5a369c4deabb8dd1d : FAILURE in 1h 43m 32s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/513b4b7cf2b8477baf13aedb77421ce6 : FAILURE in 46m 49s\n- openstack-ansible-deploy-aio_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b3c0598aecd149a7a8f923b668784bcb : FAILURE in 2h 26m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1b47268f2a2b5f4f1e0b7288ea6fd6e6206be0e2","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-09-24 10:11:26.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d4a9543be5f5cf2c75afbb2b139df5d964543c74","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-24 12:35:58.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d666f0b7cf047db8ca336bc4b32ff58 : SUCCESS in 8m 44s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ff860667c70847aba942fdb04c2a1157 : SUCCESS in 21m 31s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5ef58c5e94d2467d81906801573d3396 : FAILURE in 1h 35m 59s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/43a0e757ff904e0a9796200fbfc075b0 : FAILURE in 52m 13s\n- openstack-ansible-deploy-aio_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eaf7833a082d47f9b92c0166fd44f179 : FAILURE in 2h 16m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7da1473da665c80698834f35ae56cb8f83da1f40","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-11-30 13:29:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"424adcea21df57327a59fb8d889526579a9ae2d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-30 17:26:56.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65cc5343693a411f815816056a7467b8 : SUCCESS in 6m 06s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b544f63df51f4140b5f3fee48d6aec46 : SUCCESS in 26m 05s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7f66e948365a40bab20dd1103d66f983 : FAILURE in 2h 13m 53s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/39c11b0bd7ec482ebd8911f057f268f4 : FAILURE in 52m 24s\n- openstack-ansible-deploy-aio_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fea5cc8fc4dd4c158f550463a11748ed : FAILURE in 3h 00m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8a02d7b442974e61cb6659aeb148e82ae0c76dbe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-12-06 08:59:53.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"8ff5da89f8a5c7c0b02f92762b013ffce82501d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-06 11:36:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ebe82c422274f64845def3faa387a41 : SUCCESS in 6m 01s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/119a77f168ca4da9b6a5803ca04c00d7 : SUCCESS in 15m 27s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a9b67f47f9ce4a2fb062a0e064aa0035 : SUCCESS in 1h 44m 39s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/5cc1c0027417461995d0b8d5929024e9 : FAILURE in 1h 56m 04s (non-voting)\n- openstack-ansible-deploy-aio_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/312520b2a9e544fab5796670f25d9101 : FAILURE in 2h 33m 12s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4b783f26c8f93ed5f4e6592645cf8a0e06a5d7ce","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2021-12-06 11:38:44.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"bcdecf0e7458f579d1b83a48e3beffe7b7e822f5","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-12-06 12:22:47.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"6a306d2658c702ee52b84810440e72c27bd00ef0","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-12-06 16:10:48.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"927ac57b0cdfa563e1b00144fa40a0f785d37f4a","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-12-06 16:54:50.000000000","message":"Removed Workflow+1 by Dmitriy Rabotyagov \u003cnoonedeadpunk@ya.ru\u003e\n","accounts_in_message":[],"_revision_number":2},{"id":"dbf6ae4bb4746a6cc49e864c9ca926e04cea0445","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-12-06 16:55:06.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"01b10fa3ef92027d489ac82af709d8545eaec2ec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-06 18:41:09.000000000","message":"Patch Set 2: Verified-2\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":2},{"id":"1bc2f7d37d515845527259a79b66c84791fc5f4d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-12-06 18:52:27.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"098032d594318fec565cc6d0746a58a39bce020e","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-12-06 18:53:07.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f1cbc2917cb60a5fd47b8be52e412ce6e2df5a0c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-06 21:33:35.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/348534a0299f496f89e5d9221fba6b06 : SUCCESS in 6m 14s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ca4964737ed448a1aee71561b6cccc48 : SUCCESS in 20m 29s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f3650c9902524d409e60b16f7ac7529b : SUCCESS in 1h 47m 23s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/a82106f61a3b49f395d8f34dce9047b9 : FAILURE in 1h 56m 13s (non-voting)\n- openstack-ansible-deploy-aio_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/91fc3206e61246fbbb92569d96836590 : FAILURE in 2h 37m 25s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5f7d93510e9894a056db2027df3f73c4d58d041a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-06 21:35:50.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"b1a2463178db1ba067b46c65b86d0c8224c49baf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-06 23:29:17.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03ec021300754ec7b124e1ea0a6c832c : SUCCESS in 5m 11s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e47ec4e6c5084fe9a7489d8a4ed0bbb5 : SUCCESS in 15m 28s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7746c6d33bc84f738f99dcf22d66906b : SUCCESS in 1h 52m 19s","accounts_in_message":[],"_revision_number":3},{"id":"118d7257c93b8b9f3f682a4dc4d49b2d8d77dcfd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-06 23:29:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"12735a2fad2fb522758c1228575ec9babb71c0dd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-06 23:30:57.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99001071f8e44efda006fb783f6acc9e : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"a422c42665f987dc4e5be2a907e814e14d097d3c","revisions":{"b0414d9d526b2b41ee53367d63bd76eb43d3815e":{"kind":"REWORK","_number":1,"created":"2021-09-21 13:27:05.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/37/810237/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_manila","ref":"refs/changes/37/810237/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_manila refs/changes/37/810237/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_manila refs/changes/37/810237/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_manila refs/changes/37/810237/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_manila refs/changes/37/810237/1"}}},"commit":{"parents":[{"commit":"415b5ad7a88c3a08e367cb5c83ea0cb7a9ece3ba","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_manila/commit/415b5ad7a88c3a08e367cb5c83ea0cb7a9ece3ba"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2021-09-21 13:26:57.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2021-09-21 13:26:57.000000000","tz":180},"subject":"Refactor galera_use_ssl behaviour","message":"Refactor galera_use_ssl behaviour\n\nWith PKI role in place in most cases you don\u0027t need to explicitly\nprovide path to the CA file because PKI role ensures that CA is trusted\nby the system overall. In the meanwhile in PyMySQL [1] you must either\nprovide CA file or cert/key or enable verify.\n\nSince current behaviour is to provide path to the custom CA we expect\ncertificate being trusted overall. Thus we enable cert verification when\ngalera_use_ssl is True.\n\n[1] https://github.com/PyMySQL/PyMySQL/blob/78f0cf99e5d5351df0821442e4dc35c49a6390c6/pymysql/connections.py#L267\n\nChange-Id: I94b3f8ba5116cdfb94e9d0dc575bd7edb1d27b3c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_manila/commit/b0414d9d526b2b41ee53367d63bd76eb43d3815e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_manila/commit/b0414d9d526b2b41ee53367d63bd76eb43d3815e"}]},"branch":"refs/heads/master"},"91fd5f84e134fcf89061331e44a30517822533df":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-12-06 08:59:53.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/37/810237/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_manila","ref":"refs/changes/37/810237/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_manila refs/changes/37/810237/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_manila refs/changes/37/810237/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_manila refs/changes/37/810237/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_manila refs/changes/37/810237/2"}}},"commit":{"parents":[{"commit":"0ccd0c036066daf4a9fad013ddef668def7a1ffb","subject":"Use config_template as a collection","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_manila/commit/0ccd0c036066daf4a9fad013ddef668def7a1ffb"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2021-09-21 13:26:57.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-12-06 08:59:53.000000000","tz":0},"subject":"Refactor galera_use_ssl behaviour","message":"Refactor galera_use_ssl behaviour\n\nWith PKI role in place in most cases you don\u0027t need to explicitly\nprovide path to the CA file because PKI role ensures that CA is trusted\nby the system overall. In the meanwhile in PyMySQL [1] you must either\nprovide CA file or cert/key or enable verify.\n\nSince current behaviour is to provide path to the custom CA we expect\ncertificate being trusted overall. Thus we enable cert verification when\ngalera_use_ssl is True.\n\n[1] https://github.com/PyMySQL/PyMySQL/blob/78f0cf99e5d5351df0821442e4dc35c49a6390c6/pymysql/connections.py#L267\n\nChange-Id: I94b3f8ba5116cdfb94e9d0dc575bd7edb1d27b3c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_manila/commit/91fd5f84e134fcf89061331e44a30517822533df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_manila/commit/91fd5f84e134fcf89061331e44a30517822533df"}]},"branch":"refs/heads/master","description":"Rebase"},"a422c42665f987dc4e5be2a907e814e14d097d3c":{"kind":"REWORK","_number":3,"created":"2021-12-06 18:52:27.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/37/810237/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_manila","ref":"refs/changes/37/810237/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_manila refs/changes/37/810237/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_manila refs/changes/37/810237/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_manila refs/changes/37/810237/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_manila refs/changes/37/810237/3"}}},"commit":{"parents":[{"commit":"83fba27a1436a6ce678a30f05b88df866d5e579c","subject":"Merge \"Updated from OpenStack Ansible Tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_manila/commit/83fba27a1436a6ce678a30f05b88df866d5e579c"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2021-09-21 13:26:57.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2021-12-06 18:52:14.000000000","tz":120},"subject":"Refactor galera_use_ssl behaviour","message":"Refactor galera_use_ssl behaviour\n\nWith PKI role in place in most cases you don\u0027t need to explicitly\nprovide path to the CA file because PKI role ensures that CA is trusted\nby the system overall. In the meanwhile in PyMySQL [1] you must either\nprovide CA file or cert/key or enable verify.\n\nSince current behaviour is to provide path to the custom CA we expect\ncertificate being trusted overall. Thus we enable cert verification when\ngalera_use_ssl is True.\n\n[1] https://github.com/PyMySQL/PyMySQL/blob/78f0cf99e5d5351df0821442e4dc35c49a6390c6/pymysql/connections.py#L267\n\nChange-Id: I94b3f8ba5116cdfb94e9d0dc575bd7edb1d27b3c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_manila/commit/a422c42665f987dc4e5be2a907e814e14d097d3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_manila/commit/a422c42665f987dc4e5be2a907e814e14d097d3c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
