)]}'
{"id":"openstack%2Fopenstack-ansible-os_manila~890202","triplet_id":"openstack%2Fopenstack-ansible-os_manila~master~If989f3f5f8630af331e46f07f7df20852272188f","project":"openstack/openstack-ansible-os_manila","branch":"master","topic":"kstrenko-branch","attention_set":{},"removed_from_attention_set":{"30674":{"account":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"last_update":"2023-08-17 14:47:14.000000000","reason":"Change was abandoned"},"36227":{"account":{"_account_id":36227,"name":"Katarina Strenkova","email":"kstrenko@redhat.com","username":"kstrenko"},"last_update":"2023-08-17 14:47:14.000000000","reason":"Change was abandoned"},"22873":{"account":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"last_update":"2023-08-17 14:47:14.000000000","reason":"Change was abandoned"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2023-08-17 14:47:14.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"If989f3f5f8630af331e46f07f7df20852272188f","subject":"Replace deprecated terms","status":"ABANDONED","created":"2023-08-01 12:37:33.000000000","updated":"2023-08-17 14:47:14.000000000","total_comment_count":18,"unresolved_comment_count":4,"has_review_started":true,"meta_rev_id":"11deea439dd45217f3714aff7cc452d553ebb294","_number":890202,"virtual_id_number":890202,"owner":{"_account_id":36227,"name":"Katarina Strenkova","email":"kstrenko@redhat.com","username":"kstrenko"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-08-17 10:36:45.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"}],"CC":[{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-01 12:38:12.000000000","updated_by":{"_account_id":36227,"name":"Katarina Strenkova","email":"kstrenko@redhat.com","username":"kstrenko"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"},{"updated":"2023-08-01 12:38:12.000000000","updated_by":{"_account_id":36227,"name":"Katarina Strenkova","email":"kstrenko@redhat.com","username":"kstrenko"},"reviewer":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"state":"REVIEWER"},{"updated":"2023-08-01 15:05:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-15 10:05:28.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"CC"}],"messages":[{"id":"f010843860d41bda18a3501813003979c758a8fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36227,"name":"Katarina Strenkova","email":"kstrenko@redhat.com","username":"kstrenko"},"date":"2023-08-01 12:37:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"21c355bf7e45926f9a2e5c29d4d66765bc6d0693","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-01 15:05:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1c6d9c9efee24b1d94d2c3e593508a9e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aac804200b064f738eb98e1996a76220 : SUCCESS in 5m 03s\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c0029834d48b43b2a1541a1e2172fbac : SUCCESS in 13m 28s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1586ce3609334acdae0e04ce13f2579f : FAILURE in 26m 45s\n- openstack-ansible-deploy-aio_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1ce880f2a8ad4454b4dbb26f0a1f64a3 : FAILURE in 2h 25m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3b65d93146e2522c8766a30bc6ce36f2564279f4","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2023-08-02 08:17:17.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5f78416dcb76e857537c9776c8aa4dd53052c93d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-02 10:52:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/abeee03f38d7480b94df04f46f05884f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5fc986d1936541fcbafed78a3830959b : SUCCESS in 4m 30s\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/acf5e2c4131442118b69b610a6dfa9f8 : SUCCESS in 10m 33s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dea433bd673c4ad0b35c0f1a27b9f576 : SUCCESS in 1h 52m 20s\n- openstack-ansible-deploy-aio_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c17b8298dd98403db8eb892b8c318332 : FAILURE in 2h 28m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a3bd925b7f8d7e4948cbad1ae274030ab8e9db61","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2023-08-03 06:46:46.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5ed41735be0b746730994031f72d950ce9c53a9f","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2023-08-12 08:07:08.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"e238d4d1fc6d9142973d844164f64e7b1ba00305","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2023-08-14 07:59:55.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3aabd5891c0def8f6089d7a449be0d09169a147d","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2023-08-14 09:05:03.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"13c447d114624d89647db39264397705de30e085","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2023-08-14 10:29:23.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"dfe7286bd203347be5c2fcbd633479dbbbc83666","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36227,"name":"Katarina Strenkova","email":"kstrenko@redhat.com","username":"kstrenko"},"date":"2023-08-15 09:13:40.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"1e70a2ef5945f97cca69c08b9b1e56de93d9bae0","author":{"_account_id":36227,"name":"Katarina Strenkova","email":"kstrenko@redhat.com","username":"kstrenko"},"date":"2023-08-15 09:17:38.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"699832331a2a59f84fe59d72955bb48664eb4042","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-08-15 10:05:28.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c582017b817086d00e1d04b70689133a0e8f6204","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-08-15 10:06:03.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c9fcccc704dac86f60b93450018eb467a9580e4e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-15 11:49:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db91bd8d0b2f4b6e82ccbdb277ed5199\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/134fb1fd06db4408a3a11fb1fe0f44d2 : SUCCESS in 4m 09s\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4bbd43a3f7d44f0caa3908adee595a9a : SUCCESS in 10m 22s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9925d2318c5448f0bdad158d63f66fde : SUCCESS in 1h 28m 08s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/92dd2588cdd540ab8f356e0f2e7ad4d4 : FAILURE in 27m 14s (non-voting)\n- openstack-ansible-deploy-aio_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/db02d8c20ccc480a89c1ec2c4eedcbcc : FAILURE in 2h 27m 00s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"94313cbe782ede3341eec44bbabc8b242d16374c","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2023-08-16 09:27:30.000000000","message":"Patch Set 2: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"53f2e94cdbf1574f34776b21b8141576038e9662","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36227,"name":"Katarina Strenkova","email":"kstrenko@redhat.com","username":"kstrenko"},"date":"2023-08-17 07:40:57.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"a02b83a90e14205d76577218af5f1d0da817f087","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36227,"name":"Katarina Strenkova","email":"kstrenko@redhat.com","username":"kstrenko"},"date":"2023-08-17 07:41:38.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"4560c7e7cf7e139e2b307e92777d181bde8d8132","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-08-17 08:00:21.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ea8ca39ca982b6b8bb2f02284ca530718faf477a","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-08-17 08:14:03.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"c23f06e91a0e0cdcceb1e7c8d907e0ca0536c1fe","author":{"_account_id":36227,"name":"Katarina Strenkova","email":"kstrenko@redhat.com","username":"kstrenko"},"date":"2023-08-17 08:20:43.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"0c370485be8bee1cba883100ec62364a84b334a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 10:36:45.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4acb10d97bb4482980409debd9f99676\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/842110f62e184069a1ef9e96b2c63fbc : FAILURE in 4m 39s\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cdd338d393dd4cbabe2863ac0ed61141 : SUCCESS in 13m 29s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/749fee0bcebd43e1a28f9a4cfb50341b : SUCCESS in 2h 10m 40s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/f921994838844b8993bd60233e6196fe : FAILURE in 35m 09s (non-voting)\n- openstack-ansible-deploy-aio_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cfbbdda2c01d4049982379e62c74aaf8 : FAILURE in 2h 36m 59s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"11deea439dd45217f3714aff7cc452d553ebb294","tag":"autogenerated:gerrit:abandon","author":{"_account_id":36227,"name":"Katarina Strenkova","email":"kstrenko@redhat.com","username":"kstrenko"},"date":"2023-08-17 14:47:14.000000000","message":"Abandoned\n\nThis file is set up for clean-up","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"35dceea23ee8c2ea0faf03fd0d8d27ac4425e1dd","revisions":{"4ba37cb674e91a4479cd6f0629a20fd52e8a63fe":{"kind":"REWORK","_number":1,"created":"2023-08-01 12:37:33.000000000","uploader":{"_account_id":36227,"name":"Katarina Strenkova","email":"kstrenko@redhat.com","username":"kstrenko"},"ref":"refs/changes/02/890202/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_manila","ref":"refs/changes/02/890202/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_manila refs/changes/02/890202/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_manila refs/changes/02/890202/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_manila refs/changes/02/890202/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_manila refs/changes/02/890202/1"}}},"commit":{"parents":[{"commit":"93beaf8c4ac15a34566f1ffa312a817d759cab4d","subject":"Switch jobs to Jammy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_manila/commit/93beaf8c4ac15a34566f1ffa312a817d759cab4d"}]}],"author":{"name":"Katarina Strenkova","email":"kstrenko@redhat.com","date":"2023-08-01 12:37:00.000000000","tz":120},"committer":{"name":"Katarina Strenkova","email":"kstrenko@redhat.com","date":"2023-08-01 12:37:31.000000000","tz":120},"subject":"Replace\tdeprecated terms","message":"Replace\tdeprecated terms\n\nFollowing up on\tthe usage of inclusive jargon [1] and to ensure\nconsistency in code, there is a need to replace deprecated term\ntempest_test_blacklist with a new term tempest_test_exclude_list.\n\n[1] https://review.opendev.org/c/openstack/tempest/+/768583\n\nChange-Id: If989f3f5f8630af331e46f07f7df20852272188f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_manila/commit/4ba37cb674e91a4479cd6f0629a20fd52e8a63fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_manila/commit/4ba37cb674e91a4479cd6f0629a20fd52e8a63fe"}]},"branch":"refs/heads/master"},"e241be1d0e4ca4f0aaace6a3d817d651fc2fcc56":{"kind":"REWORK","_number":2,"created":"2023-08-15 09:13:40.000000000","uploader":{"_account_id":36227,"name":"Katarina Strenkova","email":"kstrenko@redhat.com","username":"kstrenko"},"ref":"refs/changes/02/890202/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_manila","ref":"refs/changes/02/890202/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_manila refs/changes/02/890202/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_manila refs/changes/02/890202/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_manila refs/changes/02/890202/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_manila refs/changes/02/890202/2"}}},"commit":{"parents":[{"commit":"93beaf8c4ac15a34566f1ffa312a817d759cab4d","subject":"Switch jobs to Jammy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_manila/commit/93beaf8c4ac15a34566f1ffa312a817d759cab4d"}]}],"author":{"name":"Katarina Strenkova","email":"kstrenko@redhat.com","date":"2023-08-01 12:37:00.000000000","tz":120},"committer":{"name":"Katarina Strenkova","email":"kstrenko@redhat.com","date":"2023-08-15 09:12:02.000000000","tz":120},"subject":"Replace\tdeprecated terms","message":"Replace\tdeprecated terms\n\nFollowing up on\tthe usage of inclusive jargon [1] and to ensure\nconsistency in code, there is a need to replace deprecated term\ntempest_test_blacklist with a new term tempest_test_exclude_list.\n\nAdditionally, term tempest_test_whitelist was replaced with a new term\ntempest_test_include_list.\n\n[1] https://review.opendev.org/c/openstack/tempest/+/768583\n\nChange-Id: If989f3f5f8630af331e46f07f7df20852272188f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_manila/commit/e241be1d0e4ca4f0aaace6a3d817d651fc2fcc56"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_manila/commit/e241be1d0e4ca4f0aaace6a3d817d651fc2fcc56"}]},"branch":"refs/heads/master"},"fa19ba228960432f4d98653bcc5a381e78496a67":{"kind":"REWORK","_number":3,"created":"2023-08-17 07:40:57.000000000","uploader":{"_account_id":36227,"name":"Katarina Strenkova","email":"kstrenko@redhat.com","username":"kstrenko"},"ref":"refs/changes/02/890202/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_manila","ref":"refs/changes/02/890202/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_manila refs/changes/02/890202/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_manila refs/changes/02/890202/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_manila refs/changes/02/890202/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_manila refs/changes/02/890202/3"}}},"commit":{"parents":[{"commit":"93beaf8c4ac15a34566f1ffa312a817d759cab4d","subject":"Switch jobs to Jammy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_manila/commit/93beaf8c4ac15a34566f1ffa312a817d759cab4d"}]}],"author":{"name":"Katarina Strenkova","email":"kstrenko@redhat.com","date":"2023-08-01 12:37:00.000000000","tz":120},"committer":{"name":"Katarina Strenkova","email":"kstrenko@redhat.com","date":"2023-08-17 07:40:25.000000000","tz":120},"subject":"Replace deprecated terms","message":"Replace deprecated terms\n\nFollowing up on\tthe usage of inclusive jargon [1] and to ensure\nconsistency in code, there is a need to replace deprecated term\ntempest_test_blacklist with a new term tempest_test_exclude_list.\n\nAdditionally, term tempest_test_whitelist was replaced with a new term\ntempest_test_includelist.\n\n[1] https://review.opendev.org/c/openstack/tempest/+/768583\n\nChange-Id: If989f3f5f8630af331e46f07f7df20852272188f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_manila/commit/fa19ba228960432f4d98653bcc5a381e78496a67"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_manila/commit/fa19ba228960432f4d98653bcc5a381e78496a67"}]},"branch":"refs/heads/master"},"35dceea23ee8c2ea0faf03fd0d8d27ac4425e1dd":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2023-08-17 07:41:38.000000000","uploader":{"_account_id":36227,"name":"Katarina Strenkova","email":"kstrenko@redhat.com","username":"kstrenko"},"ref":"refs/changes/02/890202/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_manila","ref":"refs/changes/02/890202/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_manila refs/changes/02/890202/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_manila refs/changes/02/890202/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_manila refs/changes/02/890202/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_manila refs/changes/02/890202/4"}}},"commit":{"parents":[{"commit":"93beaf8c4ac15a34566f1ffa312a817d759cab4d","subject":"Switch jobs to Jammy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_manila/commit/93beaf8c4ac15a34566f1ffa312a817d759cab4d"}]}],"author":{"name":"Katarina Strenkova","email":"kstrenko@redhat.com","date":"2023-08-01 12:37:00.000000000","tz":120},"committer":{"name":"Katarina Strenkova","email":"kstrenko@redhat.com","date":"2023-08-17 07:41:38.000000000","tz":0},"subject":"Replace deprecated terms","message":"Replace deprecated terms\n\nFollowing up on\tthe usage of inclusive jargon [1] and to ensure\nconsistency in code, there is a need to replace deprecated term\ntempest_test_blacklist with a new term tempest_test_excludelist.\n\nAdditionally, term tempest_test_whitelist was replaced with a new term\ntempest_test_includelist.\n\n[1] https://review.opendev.org/c/openstack/tempest/+/768583\n\nChange-Id: If989f3f5f8630af331e46f07f7df20852272188f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_manila/commit/35dceea23ee8c2ea0faf03fd0d8d27ac4425e1dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_manila/commit/35dceea23ee8c2ea0faf03fd0d8d27ac4425e1dd"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
