)]}'
{"id":"openstack%2Fopenstack-ansible-os_masakari~655613","triplet_id":"openstack%2Fopenstack-ansible-os_masakari~stable%2Fstein~I28d5b2329ad1371cd192421a5d784016087a0f56","project":"openstack/openstack-ansible-os_masakari","branch":"stable/stein","hashtags":[],"change_id":"I28d5b2329ad1371cd192421a5d784016087a0f56","subject":"Fixed nova-compute path for dist installed","status":"MERGED","created":"2019-04-25 04:58:12.000000000","updated":"2019-04-25 15:33:40.000000000","submitted":"2019-04-25 15:33:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"655613-1556206420566-cdd0ea5d","meta_rev_id":"db17e92526a209762985e4ed197b9bc5fa5a7770","_number":655613,"virtual_id_number":655613,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-04-25 15:33:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"all":[{"value":2,"date":"2019-04-25 08:18:00.000000000","_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":2,"date":"2019-04-25 13:45:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":1,"date":"2019-04-25 13:45:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-25 08:18:00.000000000","updated_by":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"reviewer":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"state":"REVIEWER"},{"updated":"2019-04-25 13:45:24.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-04-25 15:33:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"66aa336fc2befeb4a0847413740fe393019335d4","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-04-25 04:58:12.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"8d5ae797f77d611f5eea9b228d760366aae5d087","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 06:35:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/13/655613/1/check/openstack-ansible-linters/056f98f/ : SUCCESS in 5m 30s\n- openstack-tox-docs http://logs.openstack.org/13/655613/1/check/openstack-tox-docs/a4e5641/html/ : SUCCESS in 2m 44s\n- openstack-ansible-deploy-aio_metal_masakari-centos-7 http://logs.openstack.org/13/655613/1/check/openstack-ansible-deploy-aio_metal_masakari-centos-7/449922a/ : SUCCESS in 1h 31m 19s\n- openstack-ansible-deploy-aio_metal_masakari-debian-stable http://logs.openstack.org/13/655613/1/check/openstack-ansible-deploy-aio_metal_masakari-debian-stable/b31a021/ : SUCCESS in 1h 11m 11s\n- openstack-ansible-deploy-aio_metal_masakari-ubuntu-bionic http://logs.openstack.org/13/655613/1/check/openstack-ansible-deploy-aio_metal_masakari-ubuntu-bionic/95a6073/ : SUCCESS in 1h 03m 42s","accounts_in_message":[],"_revision_number":1},{"id":"d571678e9efd33ecc92d9a83b95511666c05fec4","author":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"date":"2019-04-25 08:18:00.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ed364359ae9b2bfe640484fccaa837ca7793db70","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-04-25 13:45:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7734408af4ad02b2930e7ccd26a81b316a860c52","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 13:45:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"69f2d98dff24c0b7764334ae8fddabf03e96487a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 15:33:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"db17e92526a209762985e4ed197b9bc5fa5a7770","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 15:33:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/13/655613/1/gate/openstack-ansible-linters/0289796/ : SUCCESS in 5m 11s\n- openstack-tox-docs http://logs.openstack.org/13/655613/1/gate/openstack-tox-docs/0a857eb/html/ : SUCCESS in 2m 21s\n- openstack-ansible-deploy-aio_metal_masakari-centos-7 http://logs.openstack.org/13/655613/1/gate/openstack-ansible-deploy-aio_metal_masakari-centos-7/f5623dc/ : SUCCESS in 1h 34m 01s\n- openstack-ansible-deploy-aio_metal_masakari-debian-stable http://logs.openstack.org/13/655613/1/gate/openstack-ansible-deploy-aio_metal_masakari-debian-stable/8fcb19e/ : SUCCESS in 1h 15m 29s\n- openstack-ansible-deploy-aio_metal_masakari-ubuntu-bionic http://logs.openstack.org/13/655613/1/gate/openstack-ansible-deploy-aio_metal_masakari-ubuntu-bionic/d137b1e/ : SUCCESS in 1h 01m 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"68124d9ea377d0fd191d95de293dfa42aaa3c23a","revisions":{"68124d9ea377d0fd191d95de293dfa42aaa3c23a":{"kind":"REWORK","_number":1,"created":"2019-04-25 04:58:12.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/13/655613/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_masakari","ref":"refs/changes/13/655613/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_masakari refs/changes/13/655613/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_masakari refs/changes/13/655613/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_masakari refs/changes/13/655613/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_masakari refs/changes/13/655613/1"}}},"commit":{"parents":[{"commit":"d2f9cd04705d8d540b60ed61af3329185f813cfb","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_masakari/commit/d2f9cd04705d8d540b60ed61af3329185f813cfb"}]}],"author":{"name":"Dmitriy Rabotjagov","email":"dmitriy.r@sitevalley.com","date":"2019-04-18 10:53:22.000000000","tz":180},"committer":{"name":"Dmitriy Rabotjagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2019-04-25 04:58:12.000000000","tz":0},"subject":"Fixed nova-compute path for dist installed","message":"Fixed nova-compute path for dist installed\n\nFixing path for nova which was installed from dist packages.\n\nChange-Id: I28d5b2329ad1371cd192421a5d784016087a0f56\n(cherry picked from commit 03bdc05de00925f0aefe351f81d93cd9d0ebb301)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_masakari/commit/68124d9ea377d0fd191d95de293dfa42aaa3c23a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_masakari/commit/68124d9ea377d0fd191d95de293dfa42aaa3c23a"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
