)]}'
{"id":"openstack%2Fopenstack-ansible-os_mistral~655561","triplet_id":"openstack%2Fopenstack-ansible-os_mistral~master~Ie2ad65c2645e3f9363ac96ab2f3d61ece38acde8","project":"openstack/openstack-ansible-os_mistral","branch":"master","topic":"bug/1826250","hashtags":[],"change_id":"Ie2ad65c2645e3f9363ac96ab2f3d61ece38acde8","subject":"Adding mistral_service_timeout","status":"ABANDONED","created":"2019-04-24 20:30:30.000000000","updated":"2019-07-30 19:46:14.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"617a9ea720292b756103d09c53ea61fd167358b7","_number":655561,"virtual_id_number":655561,"owner":{"_account_id":14552,"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","username":"bjoern"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"date":"2019-04-26 16:37:03.000000000","_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":1,"date":"2019-04-25 00:47:35.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":-1,"date":"2019-04-26 17:34:44.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-25 00:47:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-04-26 16:37:03.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2019-04-26 17:34:44.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"}],"messages":[{"id":"ca315e3e8a4e04bba41e478353348dfe9e92e622","author":{"_account_id":14552,"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","username":"bjoern"},"date":"2019-04-24 20:30:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"79dca578b7e368aa5865debe83639cdd6f07dc83","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 00:47:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/61/655561/1/check/openstack-ansible-linters/f5eab91/ : SUCCESS in 5m 24s\n- openstack-tox-docs http://logs.openstack.org/61/655561/1/check/openstack-tox-docs/343a443/html/ : SUCCESS in 2m 09s\n- build-openstack-releasenotes http://logs.openstack.org/61/655561/1/check/build-openstack-releasenotes/72032f8/html/ : SUCCESS in 2m 32s\n- openstack-ansible-deploy-aio_metal_mistral-centos-7 http://logs.openstack.org/61/655561/1/check/openstack-ansible-deploy-aio_metal_mistral-centos-7/b98d819/ : SUCCESS in 1h 28m 09s\n- openstack-ansible-deploy-aio_metal_mistral-debian-stable http://logs.openstack.org/61/655561/1/check/openstack-ansible-deploy-aio_metal_mistral-debian-stable/b7408f4/ : SUCCESS in 1h 00m 30s\n- openstack-ansible-deploy-aio_metal_mistral-ubuntu-bionic http://logs.openstack.org/61/655561/1/check/openstack-ansible-deploy-aio_metal_mistral-ubuntu-bionic/5c9335c/ : SUCCESS in 1h 15m 39s\n- openstack-ansible-deploy-aio_distro_metal_mistral-centos-7 http://logs.openstack.org/61/655561/1/check/openstack-ansible-deploy-aio_distro_metal_mistral-centos-7/3757a61/ : SUCCESS in 1h 28m 36s\n- openstack-ansible-deploy-aio_distro_metal_mistral-ubuntu-bionic http://logs.openstack.org/61/655561/1/check/openstack-ansible-deploy-aio_distro_metal_mistral-ubuntu-bionic/f191f4a/ : SUCCESS in 1h 18m 14s","accounts_in_message":[],"_revision_number":1},{"id":"da4487d6c9d0a1e1c0933c0f33bb5d04316bba81","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-04-26 16:37:03.000000000","message":"Patch Set 1:\n\nsee comment in https://review.opendev.org/#/c/655526/2","accounts_in_message":[],"_revision_number":1},{"id":"1ef142f3d6773fc8df2ef7a2d5d43f305edd5a61","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-04-26 17:34:44.000000000","message":"Patch Set 1: Code-Review-1\n\nplease review feedback on https://review.opendev.org/#/c/655559/1","accounts_in_message":[],"_revision_number":1},{"id":"ab346daa798563f980f822419551daadf20c2b70","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-07-30 19:46:14.000000000","message":"Abandoned\n\nno follow up patch and adds an unnecessary variable which can just be used by an override","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fb8505a647e4c2438616a57b09d7da93ef0f39d5","revisions":{"fb8505a647e4c2438616a57b09d7da93ef0f39d5":{"kind":"REWORK","_number":1,"created":"2019-04-24 20:30:30.000000000","uploader":{"_account_id":14552,"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","username":"bjoern"},"ref":"refs/changes/61/655561/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_mistral","ref":"refs/changes/61/655561/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/61/655561/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/61/655561/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/61/655561/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/61/655561/1"}}},"commit":{"parents":[{"commit":"0f8ecc36b28651c824c8833ea882b17fe40d7c44","subject":"Merge \"Update master for stable/stein\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/0f8ecc36b28651c824c8833ea882b17fe40d7c44"}]}],"author":{"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","date":"2019-04-24 20:28:39.000000000","tz":-300},"committer":{"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","date":"2019-04-24 20:30:30.000000000","tz":-300},"subject":"Adding mistral_service_timeout","message":"Adding mistral_service_timeout\n\nThe override `mistral_service_timeout` is introduced to allow\nmultiple service components to use the same frontend/backend\ntimeouts. The default value is 60 (seconds).\n\nChange-Id: Ie2ad65c2645e3f9363ac96ab2f3d61ece38acde8\nCloses-Bug: 1826250\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/fb8505a647e4c2438616a57b09d7da93ef0f39d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/fb8505a647e4c2438616a57b09d7da93ef0f39d5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
