)]}'
{"id":"openstack%2Fopenstack-ansible-os_mistral~656635","triplet_id":"openstack%2Fopenstack-ansible-os_mistral~stable%2Frocky~If10d6b8cb05da4e2824903002b1cc8adcb07f3f8","project":"openstack/openstack-ansible-os_mistral","branch":"stable/rocky","hashtags":[],"change_id":"If10d6b8cb05da4e2824903002b1cc8adcb07f3f8","subject":"Revert \"Remove deprecated auth_uri\"","status":"ABANDONED","created":"2019-05-01 13:51:38.000000000","updated":"2019-05-10 19:07:26.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"bc18c0acd82cb9f14f1fdfe35e0c350d94722d7c","_number":656635,"virtual_id_number":656635,"owner":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":-1,"date":"2019-05-01 14:42:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-01 14:42:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4ebb5eab39a1b884747e5ad951acd39d549dde92","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2019-05-01 13:51:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a78bedd60b166fcbbc23f463570811f04eeb34fa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-01 14:42:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/35/656635/1/check/openstack-tox-docs/0d1f337/html/ : SUCCESS in 3m 20s\n- openstack-ansible-linters http://logs.openstack.org/35/656635/1/check/openstack-ansible-linters/a67ca9f/ : SUCCESS in 9m 23s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/35/656635/1/check/openstack-ansible-functional-centos-7/ba6e3df/ : FAILURE in 9m 07s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/35/656635/1/check/openstack-ansible-functional-opensuse-423/3a97729/ : FAILURE in 10m 06s (non-voting)\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/35/656635/1/check/openstack-ansible-functional-opensuse-150/977f5eb/ : FAILURE in 10m 20s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/35/656635/1/check/openstack-ansible-functional-ubuntu-bionic/2cef6bc/ : FAILURE in 28m 40s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/35/656635/1/check/openstack-ansible-functional-ubuntu-xenial/d94d46c/ : FAILURE in 38m 56s","accounts_in_message":[],"_revision_number":1},{"id":"bc18c0acd82cb9f14f1fdfe35e0c350d94722d7c","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2019-05-10 19:07:26.000000000","message":"Abandoned\n\ndo not need to revert that in the end.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"763538cdfba5d175127dce86a4a6741f98df5fc2","revisions":{"763538cdfba5d175127dce86a4a6741f98df5fc2":{"kind":"REWORK","_number":1,"created":"2019-05-01 13:51:38.000000000","uploader":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"ref":"refs/changes/35/656635/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_mistral","ref":"refs/changes/35/656635/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/35/656635/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/35/656635/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/35/656635/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/35/656635/1"}}},"commit":{"parents":[{"commit":"1ae1c896bcdfe17f2076ca29b960fa1affabd4ed","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/1ae1c896bcdfe17f2076ca29b960fa1affabd4ed"}]}],"author":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2019-04-30 19:29:40.000000000","tz":-240},"committer":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2019-05-01 13:51:07.000000000","tz":-240},"subject":"Revert \"Remove deprecated auth_uri\"","message":"Revert \"Remove deprecated auth_uri\"\n\nThis reverts commit 36daa97e1b241590f1ae34b0ecb5b71e65cbdbd3.\nthis commit is not needed on Rocky.\n\nChange-Id: If10d6b8cb05da4e2824903002b1cc8adcb07f3f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/763538cdfba5d175127dce86a4a6741f98df5fc2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/763538cdfba5d175127dce86a4a6741f98df5fc2"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
