)]}'
{"id":"openstack%2Fopenstack-ansible-os_mistral~658029","triplet_id":"openstack%2Fopenstack-ansible-os_mistral~master~If09a871a48915a27c20fdf29f707e49f07f9f6ba","project":"openstack/openstack-ansible-os_mistral","branch":"master","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"If09a871a48915a27c20fdf29f707e49f07f9f6ba","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-05-09 11:34:59.000000000","updated":"2019-05-09 20:57:47.000000000","submitted":"2019-05-09 20:57:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"658029-1557435467127-d3510572","meta_rev_id":"54c4ebf861d13b0da015cf78957cbe2eea0ae9c0","_number":658029,"virtual_id_number":658029,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":2,"date":"2019-05-09 20:57:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"all":[{"value":2,"date":"2019-05-09 12:57:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"all":[{"value":1,"date":"2019-05-09 12:57:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-09 12:57:30.000000000","updated_by":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"reviewer":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"state":"REVIEWER"},{"updated":"2019-05-09 20:57:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6120d22b56d38ae011defa76a81982a5ce7ab250","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-05-09 11:34:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e5146dbbdc70d6d8105263e15cc7e59dd124556e","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-05-09 12:57:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0952f713addb546644e8564b68a7930f961be827","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 13:22:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/29/658029/1/check/openstack-ansible-linters/65f7d01/ : SUCCESS in 5m 52s\n- openstack-tox-docs http://logs.openstack.org/29/658029/1/check/openstack-tox-docs/aa39450/html/ : SUCCESS in 2m 50s\n- build-openstack-releasenotes http://logs.openstack.org/29/658029/1/check/build-openstack-releasenotes/4b2c4e5/html/ : SUCCESS in 2m 38s\n- openstack-ansible-deploy-aio_metal_mistral-centos-7 http://logs.openstack.org/29/658029/1/check/openstack-ansible-deploy-aio_metal_mistral-centos-7/f301d44/ : SUCCESS in 1h 29m 21s\n- openstack-ansible-deploy-aio_metal_mistral-debian-stable http://logs.openstack.org/29/658029/1/check/openstack-ansible-deploy-aio_metal_mistral-debian-stable/6a5937a/ : SUCCESS in 58m 49s\n- openstack-ansible-deploy-aio_metal_mistral-ubuntu-bionic http://logs.openstack.org/29/658029/1/check/openstack-ansible-deploy-aio_metal_mistral-ubuntu-bionic/0587751/ : SUCCESS in 1h 00m 04s\n- openstack-ansible-deploy-aio_distro_metal_mistral-centos-7 http://logs.openstack.org/29/658029/1/check/openstack-ansible-deploy-aio_distro_metal_mistral-centos-7/e8a8cf8/ : SUCCESS in 1h 30m 02s\n- openstack-ansible-deploy-aio_distro_metal_mistral-ubuntu-bionic http://logs.openstack.org/29/658029/1/check/openstack-ansible-deploy-aio_distro_metal_mistral-ubuntu-bionic/c8c4c33/ : SUCCESS in 59m 21s","accounts_in_message":[],"_revision_number":1},{"id":"e01df28becb5238b10e69335b8fe3fa83fe39866","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 13:22:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6c4fc8f26288026b6846b9cd90e80c425548fd86","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 16:49:25.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/29/658029/1/gate/openstack-ansible-linters/4ee8e38/ : SUCCESS in 5m 15s\n- openstack-tox-docs http://logs.openstack.org/29/658029/1/gate/openstack-tox-docs/992d07e/html/ : SUCCESS in 2m 51s\n- build-openstack-releasenotes http://logs.openstack.org/29/658029/1/gate/build-openstack-releasenotes/1cbac5a/html/ : SUCCESS in 2m 08s\n- openstack-ansible-deploy-aio_metal_mistral-centos-7 http://logs.openstack.org/29/658029/1/gate/openstack-ansible-deploy-aio_metal_mistral-centos-7/29f11fc/ : SUCCESS in 1h 29m 46s\n- openstack-ansible-deploy-aio_metal_mistral-debian-stable http://logs.openstack.org/29/658029/1/gate/openstack-ansible-deploy-aio_metal_mistral-debian-stable/01ee99c/ : SUCCESS in 1h 03m 24s\n- openstack-ansible-deploy-aio_metal_mistral-ubuntu-bionic http://logs.openstack.org/29/658029/1/gate/openstack-ansible-deploy-aio_metal_mistral-ubuntu-bionic/79d18e6/ : SUCCESS in 1h 08m 19s\n- openstack-ansible-deploy-aio_distro_metal_mistral-centos-7 http://logs.openstack.org/29/658029/1/gate/openstack-ansible-deploy-aio_distro_metal_mistral-centos-7/7ec9542/ : FAILURE in 2h 46m 26s\n- openstack-ansible-deploy-aio_distro_metal_mistral-ubuntu-bionic http://logs.openstack.org/29/658029/1/gate/openstack-ansible-deploy-aio_distro_metal_mistral-ubuntu-bionic/c9a9076/ : SUCCESS in 59m 04s","accounts_in_message":[],"_revision_number":1},{"id":"e3bb5608387dcd611848c77657c621e6ec8f4de1","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-05-09 17:38:11.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"392eefdfe21ac72fd5e09a7bdd865c28b7e5d479","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 19:19:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/29/658029/1/check/openstack-ansible-linters/c16a1f7/ : SUCCESS in 7m 51s\n- openstack-tox-docs http://logs.openstack.org/29/658029/1/check/openstack-tox-docs/77bb04c/html/ : SUCCESS in 3m 05s\n- build-openstack-releasenotes http://logs.openstack.org/29/658029/1/check/build-openstack-releasenotes/fd9a83f/html/ : SUCCESS in 2m 10s\n- openstack-ansible-deploy-aio_metal_mistral-centos-7 http://logs.openstack.org/29/658029/1/check/openstack-ansible-deploy-aio_metal_mistral-centos-7/9094249/ : SUCCESS in 1h 27m 49s\n- openstack-ansible-deploy-aio_metal_mistral-debian-stable http://logs.openstack.org/29/658029/1/check/openstack-ansible-deploy-aio_metal_mistral-debian-stable/aa340b3/ : SUCCESS in 1h 05m 21s\n- openstack-ansible-deploy-aio_metal_mistral-ubuntu-bionic http://logs.openstack.org/29/658029/1/check/openstack-ansible-deploy-aio_metal_mistral-ubuntu-bionic/f586be4/ : SUCCESS in 1h 00m 51s\n- openstack-ansible-deploy-aio_distro_metal_mistral-centos-7 http://logs.openstack.org/29/658029/1/check/openstack-ansible-deploy-aio_distro_metal_mistral-centos-7/95348dc/ : SUCCESS in 1h 27m 11s\n- openstack-ansible-deploy-aio_distro_metal_mistral-ubuntu-bionic http://logs.openstack.org/29/658029/1/check/openstack-ansible-deploy-aio_distro_metal_mistral-ubuntu-bionic/92bcb79/ : SUCCESS in 1h 39m 23s","accounts_in_message":[],"_revision_number":1},{"id":"723116e55437d7fdd0f72b9e74d2cb67a4c302bc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 19:20:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"87204e639d9770f4c1f3e8897b53febb1a9ede43","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 20:57:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/29/658029/1/gate/openstack-ansible-linters/dac7576/ : SUCCESS in 4m 49s\n- openstack-tox-docs http://logs.openstack.org/29/658029/1/gate/openstack-tox-docs/822edcd/html/ : SUCCESS in 2m 27s\n- build-openstack-releasenotes http://logs.openstack.org/29/658029/1/gate/build-openstack-releasenotes/7144abc/html/ : SUCCESS in 2m 31s\n- openstack-ansible-deploy-aio_metal_mistral-centos-7 http://logs.openstack.org/29/658029/1/gate/openstack-ansible-deploy-aio_metal_mistral-centos-7/8c35535/ : SUCCESS in 1h 27m 37s\n- openstack-ansible-deploy-aio_metal_mistral-debian-stable http://logs.openstack.org/29/658029/1/gate/openstack-ansible-deploy-aio_metal_mistral-debian-stable/a4c6b91/ : SUCCESS in 1h 04m 13s\n- openstack-ansible-deploy-aio_metal_mistral-ubuntu-bionic http://logs.openstack.org/29/658029/1/gate/openstack-ansible-deploy-aio_metal_mistral-ubuntu-bionic/f3a4679/ : SUCCESS in 1h 05m 06s\n- openstack-ansible-deploy-aio_distro_metal_mistral-centos-7 http://logs.openstack.org/29/658029/1/gate/openstack-ansible-deploy-aio_distro_metal_mistral-centos-7/79ff187/ : SUCCESS in 1h 30m 33s\n- openstack-ansible-deploy-aio_distro_metal_mistral-ubuntu-bionic http://logs.openstack.org/29/658029/1/gate/openstack-ansible-deploy-aio_distro_metal_mistral-ubuntu-bionic/6569dce/ : SUCCESS in 1h 07m 05s","accounts_in_message":[],"_revision_number":1},{"id":"05619393a0fcc1c37213be97c7f72959b599f16a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 20:57:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"efb9bc7a245d2795246928993ccb39d5a9f0febe","revisions":{"efb9bc7a245d2795246928993ccb39d5a9f0febe":{"kind":"REWORK","_number":1,"created":"2019-05-09 11:34:59.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/29/658029/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_mistral","ref":"refs/changes/29/658029/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/29/658029/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/29/658029/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/29/658029/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/29/658029/1"}}},"commit":{"parents":[{"commit":"0f8ecc36b28651c824c8833ea882b17fe40d7c44","subject":"Merge \"Update master for stable/stein\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/0f8ecc36b28651c824c8833ea882b17fe40d7c44"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-09 11:34:58.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-09 11:34:58.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: If09a871a48915a27c20fdf29f707e49f07f9f6ba\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/efb9bc7a245d2795246928993ccb39d5a9f0febe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/efb9bc7a245d2795246928993ccb39d5a9f0febe"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
