)]}'
{"id":"openstack%2Fopenstack-ansible-os_mistral~659299","triplet_id":"openstack%2Fopenstack-ansible-os_mistral~stable%2Frocky~I061eccac155958171097ba3633433fc0b9b6a5b2","project":"openstack/openstack-ansible-os_mistral","branch":"stable/rocky","topic":"fix-rocky-deploy","hashtags":[],"change_id":"I061eccac155958171097ba3633433fc0b9b6a5b2","subject":"Fix rocky config","status":"MERGED","created":"2019-05-15 14:28:01.000000000","updated":"2019-05-27 19:08:58.000000000","submitted":"2019-05-27 19:08:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"659299-1558984138408-72d99f86","meta_rev_id":"50c842bd0f7b49ce93ca53fbc97374941a7de846","_number":659299,"virtual_id_number":659299,"owner":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-05-27 13:50:02.000000000","_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-05-27 19:08:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":2,"date":"2019-05-27 17:45:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":2,"date":"2019-05-27 18:16:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":1,"date":"2019-05-27 18:16:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-27 17:45:53.000000000","updated_by":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"reviewer":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"state":"REVIEWER"},{"updated":"2019-05-27 18:16:58.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-05-27 19:08:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"14c9b4c1d4d680e65663cadf1ab9ff63c91bb541","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2019-05-15 14:28:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9e2e5e5a7ff254d158273509c5e41e831e11d2ab","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 15:06:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/99/659299/1/check/openstack-tox-docs/2d2e9c0/html/ : SUCCESS in 2m 58s\n- openstack-ansible-linters http://logs.openstack.org/99/659299/1/check/openstack-ansible-linters/743d2f3/ : SUCCESS in 5m 13s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/99/659299/1/check/openstack-ansible-functional-centos-7/5417270/ : FAILURE in 9m 26s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/99/659299/1/check/openstack-ansible-functional-opensuse-423/4240ecd/ : FAILURE in 10m 58s (non-voting)\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/99/659299/1/check/openstack-ansible-functional-opensuse-150/6e44184/ : FAILURE in 10m 21s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/99/659299/1/check/openstack-ansible-functional-ubuntu-bionic/a64f4c9/ : SUCCESS in 33m 31s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/99/659299/1/check/openstack-ansible-functional-ubuntu-xenial/feb4f41/ : SUCCESS in 30m 19s","accounts_in_message":[],"_revision_number":1},{"id":"9fc0bfe06b9bb4deec91efda64a7ad0c4b747959","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2019-05-27 13:50:02.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7b40d4e3144cefddc64e6eaa71e338d15ff383c2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-27 14:41:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/99/659299/1/check/openstack-tox-docs/7cabd43/html/ : SUCCESS in 3m 13s\n- openstack-ansible-linters http://logs.openstack.org/99/659299/1/check/openstack-ansible-linters/0d366fd/ : SUCCESS in 5m 51s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/99/659299/1/check/openstack-ansible-functional-centos-7/9119dee/ : SUCCESS in 42m 50s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/99/659299/1/check/openstack-ansible-functional-opensuse-423/9c2d9f8/ : FAILURE in 7m 19s (non-voting)\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/99/659299/1/check/openstack-ansible-functional-opensuse-150/849c4b3/ : FAILURE in 7m 29s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/99/659299/1/check/openstack-ansible-functional-ubuntu-bionic/7f86d21/ : SUCCESS in 34m 03s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/99/659299/1/check/openstack-ansible-functional-ubuntu-xenial/e6ae7f9/ : SUCCESS in 51m 12s","accounts_in_message":[],"_revision_number":1},{"id":"1b0dd0ce7c4a1c0c3e7e551725032ef5c1753a64","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-05-27 17:45:53.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"87079ad67234044a89c3e69cd0c19b7be6a0890f","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-05-27 18:16:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"496d8b9d7c877efec67124947160af1d22dd19e5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-27 18:17:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bf9c4c7758fd6d461cfcd705520c9c6ec265f1e5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-27 19:08:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"cd367361603eac8ce602a356b7402503deeabe98","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-27 19:08:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/99/659299/1/gate/openstack-tox-docs/11b597b/html/ : SUCCESS in 2m 59s\n- openstack-ansible-linters http://logs.openstack.org/99/659299/1/gate/openstack-ansible-linters/f941c86/ : SUCCESS in 5m 14s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/99/659299/1/gate/openstack-ansible-functional-centos-7/35da64f/ : SUCCESS in 39m 17s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/99/659299/1/gate/openstack-ansible-functional-opensuse-423/097c129/ : FAILURE in 5m 47s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/99/659299/1/gate/openstack-ansible-functional-ubuntu-bionic/fb3bd18/ : SUCCESS in 27m 53s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/99/659299/1/gate/openstack-ansible-functional-ubuntu-xenial/c3067b3/ : SUCCESS in 35m 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4100d0c8e8ad25e26769d8a75f97900dcbd08749","revisions":{"4100d0c8e8ad25e26769d8a75f97900dcbd08749":{"kind":"REWORK","_number":1,"created":"2019-05-15 14:28:01.000000000","uploader":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"ref":"refs/changes/99/659299/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_mistral","ref":"refs/changes/99/659299/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/99/659299/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/99/659299/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/99/659299/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/99/659299/1"}}},"commit":{"parents":[{"commit":"1ae1c896bcdfe17f2076ca29b960fa1affabd4ed","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/1ae1c896bcdfe17f2076ca29b960fa1affabd4ed"}]}],"author":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2019-05-15 14:13:50.000000000","tz":-240},"committer":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2019-05-15 14:13:50.000000000","tz":-240},"subject":"Fix rocky config","message":"Fix rocky config\n\nthe change auth_uri \u003e www_authenticate_uri is not in Rocky\n\nChange-Id: I061eccac155958171097ba3633433fc0b9b6a5b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/4100d0c8e8ad25e26769d8a75f97900dcbd08749"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/4100d0c8e8ad25e26769d8a75f97900dcbd08749"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
