)]}'
{"id":"openstack%2Fopenstack-ansible-os_mistral~659769","triplet_id":"openstack%2Fopenstack-ansible-os_mistral~master~I0603501d55e04b4c6d5bb03510ebf74c001ae272","project":"openstack/openstack-ansible-os_mistral","branch":"master","hashtags":[],"change_id":"I0603501d55e04b4c6d5bb03510ebf74c001ae272","subject":"Cap sphinx for py2 to match global requirements","status":"MERGED","created":"2019-05-17 10:07:47.000000000","updated":"2019-05-19 07:36:05.000000000","submitted":"2019-05-19 07:36:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"659769-1558251365249-a663f06a","meta_rev_id":"cb9b76b7c2f51cc778e455cf689c02b69fd298e8","_number":659769,"virtual_id_number":659769,"owner":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-05-19 07:36:05.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":2,"date":"2019-05-18 11:20:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-05-19 07:25:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-05-19 07:25:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-18 11:20:32.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-05-19 07:25:35.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-05-19 07:36:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"638daa036b436394ce5e97cb4be814c2736fa239","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2019-05-17 10:07:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"70708dd4ac6b379997db9478ee78c35f46a6344a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 10:36:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/69/659769/1/check/requirements-check/e4b6c2e/ : SUCCESS in 2m 02s\n- openstack-ansible-linters http://logs.openstack.org/69/659769/1/check/openstack-ansible-linters/08b809c/ : SUCCESS in 4m 06s\n- openstack-tox-docs http://logs.openstack.org/69/659769/1/check/openstack-tox-docs/76b624f/html/ : SUCCESS in 2m 33s\n- build-openstack-releasenotes http://logs.openstack.org/69/659769/1/check/build-openstack-releasenotes/f61e64b/html/ : SUCCESS in 2m 11s","accounts_in_message":[],"_revision_number":1},{"id":"1828de6474e1578e2fb8b0838e779c7d3543021c","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-05-18 11:20:32.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"df265ba25ed026514f3c0b0b2a114c5ae534264c","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-05-19 07:25:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c4132ba19d4ef115167f5a375063d4a755be06f1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-19 07:25:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6e167ca54684689cfcbc650a11e46a9833d77adc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-19 07:36:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"cb9b76b7c2f51cc778e455cf689c02b69fd298e8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-19 07:36:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check http://logs.openstack.org/69/659769/1/gate/requirements-check/0b9425c/ : SUCCESS in 1m 58s\n- openstack-ansible-linters http://logs.openstack.org/69/659769/1/gate/openstack-ansible-linters/fe9df0d/ : SUCCESS in 5m 00s\n- openstack-tox-docs http://logs.openstack.org/69/659769/1/gate/openstack-tox-docs/8bd28a8/html/ : SUCCESS in 2m 34s\n- build-openstack-releasenotes http://logs.openstack.org/69/659769/1/gate/build-openstack-releasenotes/9843eb9/html/ : SUCCESS in 2m 21s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1bd38882ba9578978db7eb850b26e55f54f5d7ab","revisions":{"1bd38882ba9578978db7eb850b26e55f54f5d7ab":{"kind":"REWORK","_number":1,"created":"2019-05-17 10:07:47.000000000","uploader":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"ref":"refs/changes/69/659769/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_mistral","ref":"refs/changes/69/659769/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/69/659769/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/69/659769/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/69/659769/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/69/659769/1"}}},"commit":{"parents":[{"commit":"efb9bc7a245d2795246928993ccb39d5a9f0febe","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/efb9bc7a245d2795246928993ccb39d5a9f0febe"}]}],"author":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2019-05-17 10:07:27.000000000","tz":480},"committer":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2019-05-17 10:07:27.000000000","tz":480},"subject":"Cap sphinx for py2 to match global requirements","message":"Cap sphinx for py2 to match global requirements\n\nChange-Id: I0603501d55e04b4c6d5bb03510ebf74c001ae272\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/1bd38882ba9578978db7eb850b26e55f54f5d7ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/1bd38882ba9578978db7eb850b26e55f54f5d7ab"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
