)]}'
{"id":"openstack%2Fopenstack-ansible-os_mistral~664448","triplet_id":"openstack%2Fopenstack-ansible-os_mistral~master~I4b9846a12382f3ef4acdffce8a6f2338ed45d6ff","project":"openstack/openstack-ansible-os_mistral","branch":"master","topic":"ubuntu-distro-install","hashtags":[],"change_id":"I4b9846a12382f3ef4acdffce8a6f2338ed45d6ff","subject":"Fix distro installs on Ubuntu","status":"MERGED","created":"2019-06-10 21:50:56.000000000","updated":"2019-06-11 22:48:28.000000000","submitted":"2019-06-11 22:48:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"664448-1560293280366-b5d80b2d","meta_rev_id":"be98955d7c870543239622e0e672f78dc427d462","_number":664448,"virtual_id_number":664448,"owner":{"_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":2,"date":"2019-06-11 22:48:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"all":[{"value":2,"date":"2019-06-11 07:05:23.000000000","_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-06-11 07:22:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-11 21:17:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-11 07:05:23.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2019-06-11 21:17:42.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-06-11 22:48:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1d3a9c1e4c1c1255b7598deacda63a618298db21","author":{"_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"},"date":"2019-06-10 21:50:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cf22fa75f0a332709693cbd977cfa8ae782a1bc1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 23:27:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/48/664448/1/check/openstack-ansible-linters/682f881/ : SUCCESS in 5m 56s\n- openstack-tox-docs http://logs.openstack.org/48/664448/1/check/openstack-tox-docs/db50b9b/html/ : SUCCESS in 2m 50s\n- openstack-ansible-deploy-aio_metal_mistral-centos-7 http://logs.openstack.org/48/664448/1/check/openstack-ansible-deploy-aio_metal_mistral-centos-7/baea90b/ : SUCCESS in 1h 30m 03s\n- openstack-ansible-deploy-aio_metal_mistral-debian-stable http://logs.openstack.org/48/664448/1/check/openstack-ansible-deploy-aio_metal_mistral-debian-stable/97aa4f7/ : SUCCESS in 56m 58s\n- openstack-ansible-deploy-aio_metal_mistral-ubuntu-bionic http://logs.openstack.org/48/664448/1/check/openstack-ansible-deploy-aio_metal_mistral-ubuntu-bionic/74c4a29/ : SUCCESS in 1h 08m 10s\n- openstack-ansible-deploy-aio_distro_metal_mistral-centos-7 http://logs.openstack.org/48/664448/1/check/openstack-ansible-deploy-aio_distro_metal_mistral-centos-7/b4d25d8/ : SUCCESS in 56m 54s\n- openstack-ansible-deploy-aio_distro_metal_mistral-ubuntu-bionic http://logs.openstack.org/48/664448/1/check/openstack-ansible-deploy-aio_distro_metal_mistral-ubuntu-bionic/6836c4a/ : FAILURE in 20m 31s","accounts_in_message":[],"_revision_number":1},{"id":"4a2d226e1e27e828b13e602e2c5308cfd527c66e","author":{"_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"},"date":"2019-06-10 23:47:31.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"d8959ad2ba376c1779c644f68b9096e9018672b2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 01:15:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/48/664448/2/check/openstack-ansible-linters/e47a146/ : SUCCESS in 6m 34s\n- openstack-tox-docs http://logs.openstack.org/48/664448/2/check/openstack-tox-docs/e5a8061/html/ : SUCCESS in 3m 46s\n- openstack-ansible-deploy-aio_metal_mistral-centos-7 http://logs.openstack.org/48/664448/2/check/openstack-ansible-deploy-aio_metal_mistral-centos-7/9b47dd4/ : SUCCESS in 1h 22m 29s\n- openstack-ansible-deploy-aio_metal_mistral-debian-stable http://logs.openstack.org/48/664448/2/check/openstack-ansible-deploy-aio_metal_mistral-debian-stable/529fdc3/ : SUCCESS in 1h 08m 21s\n- openstack-ansible-deploy-aio_metal_mistral-ubuntu-bionic http://logs.openstack.org/48/664448/2/check/openstack-ansible-deploy-aio_metal_mistral-ubuntu-bionic/3de0595/ : SUCCESS in 1h 01m 39s\n- openstack-ansible-deploy-aio_distro_metal_mistral-centos-7 http://logs.openstack.org/48/664448/2/check/openstack-ansible-deploy-aio_distro_metal_mistral-centos-7/42c9e06/ : SUCCESS in 57m 59s\n- openstack-ansible-deploy-aio_distro_metal_mistral-ubuntu-bionic http://logs.openstack.org/48/664448/2/check/openstack-ansible-deploy-aio_distro_metal_mistral-ubuntu-bionic/acc3f45/ : SUCCESS in 43m 36s","accounts_in_message":[],"_revision_number":2},{"id":"86ca7559a760e7f359d68a83ac39a04a9fa2e3d2","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-06-11 07:05:23.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"fe0dc8c91fc78e79108df509c94627d6e6a95990","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-11 07:22:02.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"8e42a4c0040ccff4a99e99f76a8bbafd251df61e","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-11 21:17:37.000000000","message":"Removed Workflow+1 by Jonathan Rosser \u003cjonathan.rosser@rd.bbc.co.uk\u003e\n","accounts_in_message":[],"_revision_number":2},{"id":"283809676cdea6e90a08fe8ccecc7f3fe1282a9d","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-11 21:17:42.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"91fbb572e690071d6f86dbd24aeaf84d203ffe59","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 21:17:54.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"94c424eccaa518db9ac363f4cc5551c42a253fe4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 22:48:00.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/48/664448/2/gate/openstack-ansible-linters/be82df1/ : SUCCESS in 4m 08s\n- openstack-tox-docs http://logs.openstack.org/48/664448/2/gate/openstack-tox-docs/0046df0/html/ : SUCCESS in 3m 24s\n- openstack-ansible-deploy-aio_metal_mistral-centos-7 http://logs.openstack.org/48/664448/2/gate/openstack-ansible-deploy-aio_metal_mistral-centos-7/5a4e4b1/ : SUCCESS in 1h 23m 35s\n- openstack-ansible-deploy-aio_metal_mistral-debian-stable http://logs.openstack.org/48/664448/2/gate/openstack-ansible-deploy-aio_metal_mistral-debian-stable/ea5fab0/ : SUCCESS in 58m 18s\n- openstack-ansible-deploy-aio_metal_mistral-ubuntu-bionic http://logs.openstack.org/48/664448/2/gate/openstack-ansible-deploy-aio_metal_mistral-ubuntu-bionic/345731b/ : SUCCESS in 1h 00m 37s\n- openstack-ansible-deploy-aio_distro_metal_mistral-centos-7 http://logs.openstack.org/48/664448/2/gate/openstack-ansible-deploy-aio_distro_metal_mistral-centos-7/c01173a/ : SUCCESS in 59m 49s\n- openstack-ansible-deploy-aio_distro_metal_mistral-ubuntu-bionic http://logs.openstack.org/48/664448/2/gate/openstack-ansible-deploy-aio_distro_metal_mistral-ubuntu-bionic/bb559ca/ : SUCCESS in 45m 40s","accounts_in_message":[],"_revision_number":2},{"id":"46d6c31b8d0e16393606056fc00e4f6d8750282a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 22:48:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"1a23feb255c06e3521662626bcb380969da4881b","author":{"_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"},"date":"2019-06-11 22:48:28.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit f1e952108037b84af16d5e59fa3dbf93a997550d","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9dc10013800621a4773abaa33d9991a29e4649e7","revisions":{"09b0b6123b9748572d60afa958852a09659d0425":{"kind":"REWORK","_number":1,"created":"2019-06-10 21:50:56.000000000","uploader":{"_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"},"ref":"refs/changes/48/664448/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_mistral","ref":"refs/changes/48/664448/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/48/664448/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/48/664448/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/48/664448/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/48/664448/1"}}},"commit":{"parents":[{"commit":"1bd38882ba9578978db7eb850b26e55f54f5d7ab","subject":"Cap sphinx for py2 to match global requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/1bd38882ba9578978db7eb850b26e55f54f5d7ab"}]}],"author":{"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","date":"2019-06-10 21:49:37.000000000","tz":-420},"committer":{"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","date":"2019-06-10 21:49:37.000000000","tz":-420},"subject":"Fix distro installs on Ubuntu","message":"Fix distro installs on Ubuntu\n\nBeginning in the Stein release, Ubuntu distro packages are now using\nPython3. This requires additionally installing and using the uwsgi\npython3 plugin.\n\nInstall the \u0027python3-mistral\u0027 package instead of individual service\npackages since they provide service config files that conflict\nwith the ones OSA provides.\n\nChange-Id: I4b9846a12382f3ef4acdffce8a6f2338ed45d6ff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/09b0b6123b9748572d60afa958852a09659d0425"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/09b0b6123b9748572d60afa958852a09659d0425"}]},"branch":"refs/heads/master"},"9dc10013800621a4773abaa33d9991a29e4649e7":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-06-10 23:47:31.000000000","uploader":{"_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"},"ref":"refs/changes/48/664448/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_mistral","ref":"refs/changes/48/664448/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/48/664448/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/48/664448/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/48/664448/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/48/664448/2"}}},"commit":{"parents":[{"commit":"1bd38882ba9578978db7eb850b26e55f54f5d7ab","subject":"Cap sphinx for py2 to match global requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/1bd38882ba9578978db7eb850b26e55f54f5d7ab"}]}],"author":{"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","date":"2019-06-10 21:49:37.000000000","tz":-420},"committer":{"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","date":"2019-06-10 23:47:22.000000000","tz":0},"subject":"Fix distro installs on Ubuntu","message":"Fix distro installs on Ubuntu\n\nBeginning in the Stein release, Ubuntu distro packages are now using\nPython3. This requires additionally installing and using the uwsgi\npython3 plugin.\n\nInstall the \u0027python3-mistral\u0027 package instead of individual service\npackages since they provide service config files that conflict\nwith the ones OSA provides.\n\nDepends-On: https://review.opendev.org/#/c/662571\nChange-Id: I4b9846a12382f3ef4acdffce8a6f2338ed45d6ff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/9dc10013800621a4773abaa33d9991a29e4649e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/9dc10013800621a4773abaa33d9991a29e4649e7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
