)]}'
{"id":"openstack%2Fopenstack-ansible-os_mistral~664728","triplet_id":"openstack%2Fopenstack-ansible-os_mistral~master~I4b368558d2bbdafa7900e71a66664a94345fda0f","project":"openstack/openstack-ansible-os_mistral","branch":"master","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"I4b368558d2bbdafa7900e71a66664a94345fda0f","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-06-11 22:18:31.000000000","updated":"2019-06-12 07:42:04.000000000","submitted":"2019-06-12 07:42:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"664728-1560325324535-7fce86ad","meta_rev_id":"e4bc4beff9be9384a0da06f4b260954fe311a695","_number":664728,"virtual_id_number":664728,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":2,"date":"2019-06-12 07:42:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-06-12 06:14:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-12 06:14:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-12 06:14:35.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-06-12 07:42:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cddf3d9bd61fc6641105abe43b9e042b1c8c80ae","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-06-11 22:18:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"75f18767f7b3dff96aa61234f76fadef135a3236","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 23:56:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/28/664728/1/check/openstack-ansible-linters/751bd38/ : SUCCESS in 4m 56s\n- openstack-tox-docs http://logs.openstack.org/28/664728/1/check/openstack-tox-docs/67a530e/html/ : SUCCESS in 3m 00s\n- openstack-ansible-deploy-aio_metal_mistral-centos-7 http://logs.openstack.org/28/664728/1/check/openstack-ansible-deploy-aio_metal_mistral-centos-7/0d2854b/ : SUCCESS in 1h 15m 42s\n- openstack-ansible-deploy-aio_metal_mistral-debian-stable http://logs.openstack.org/28/664728/1/check/openstack-ansible-deploy-aio_metal_mistral-debian-stable/9ab50dc/ : SUCCESS in 1h 01m 53s\n- openstack-ansible-deploy-aio_metal_mistral-ubuntu-bionic http://logs.openstack.org/28/664728/1/check/openstack-ansible-deploy-aio_metal_mistral-ubuntu-bionic/4e63795/ : SUCCESS in 56m 02s\n- openstack-ansible-deploy-aio_distro_metal_mistral-centos-7 http://logs.openstack.org/28/664728/1/check/openstack-ansible-deploy-aio_distro_metal_mistral-centos-7/c5b9214/ : SUCCESS in 59m 36s\n- openstack-ansible-deploy-aio_distro_metal_mistral-ubuntu-bionic http://logs.openstack.org/28/664728/1/check/openstack-ansible-deploy-aio_distro_metal_mistral-ubuntu-bionic/20a1c75/ : SUCCESS in 44m 56s","accounts_in_message":[],"_revision_number":1},{"id":"54a77495b944cdbb12e0f79aa7381134eeb64a74","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-12 06:14:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"950bf5c229ed92f8d0c06db1e5b88a99691dc5c0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 06:14:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a146c5dc32c6eda4c10d8544b57fc3cdd034bc0b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 07:42:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ac5b1abe64ebb604164655ddc7bae095f35dbb1f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 07:42:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/28/664728/1/gate/openstack-ansible-linters/a11623b/ : SUCCESS in 5m 51s\n- openstack-tox-docs http://logs.openstack.org/28/664728/1/gate/openstack-tox-docs/a0224bf/html/ : SUCCESS in 2m 55s\n- openstack-ansible-deploy-aio_metal_mistral-centos-7 http://logs.openstack.org/28/664728/1/gate/openstack-ansible-deploy-aio_metal_mistral-centos-7/625cbe0/ : SUCCESS in 1h 19m 48s\n- openstack-ansible-deploy-aio_metal_mistral-debian-stable http://logs.openstack.org/28/664728/1/gate/openstack-ansible-deploy-aio_metal_mistral-debian-stable/7372631/ : SUCCESS in 1h 06m 34s\n- openstack-ansible-deploy-aio_metal_mistral-ubuntu-bionic http://logs.openstack.org/28/664728/1/gate/openstack-ansible-deploy-aio_metal_mistral-ubuntu-bionic/b4f9bb1/ : SUCCESS in 55m 51s\n- openstack-ansible-deploy-aio_distro_metal_mistral-centos-7 http://logs.openstack.org/28/664728/1/gate/openstack-ansible-deploy-aio_distro_metal_mistral-centos-7/a7ab532/ : SUCCESS in 1h 07m 55s\n- openstack-ansible-deploy-aio_distro_metal_mistral-ubuntu-bionic http://logs.openstack.org/28/664728/1/gate/openstack-ansible-deploy-aio_distro_metal_mistral-ubuntu-bionic/652b793/ : SUCCESS in 44m 25s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2d3abc0e2e89d7e898f58ef7beb853a3eede7e2b","revisions":{"2d3abc0e2e89d7e898f58ef7beb853a3eede7e2b":{"kind":"REWORK","_number":1,"created":"2019-06-11 22:18:31.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/28/664728/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_mistral","ref":"refs/changes/28/664728/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/28/664728/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/28/664728/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/28/664728/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/28/664728/1"}}},"commit":{"parents":[{"commit":"1bd38882ba9578978db7eb850b26e55f54f5d7ab","subject":"Cap sphinx for py2 to match global requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/1bd38882ba9578978db7eb850b26e55f54f5d7ab"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-11 22:18:29.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-11 22:18:29.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I4b368558d2bbdafa7900e71a66664a94345fda0f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/2d3abc0e2e89d7e898f58ef7beb853a3eede7e2b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/2d3abc0e2e89d7e898f58ef7beb853a3eede7e2b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
