)]}'
{"id":"openstack%2Fopenstack-ansible-os_mistral~664737","triplet_id":"openstack%2Fopenstack-ansible-os_mistral~stable%2Fstein~I4b9846a12382f3ef4acdffce8a6f2338ed45d6ff","project":"openstack/openstack-ansible-os_mistral","branch":"stable/stein","topic":"ubuntu-distro-install-stable/stein","hashtags":[],"change_id":"I4b9846a12382f3ef4acdffce8a6f2338ed45d6ff","subject":"Fix distro installs on Ubuntu","status":"MERGED","created":"2019-06-11 22:48:28.000000000","updated":"2019-07-13 00:52:36.000000000","submitted":"2019-07-13 00:52:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"664737-1562979156849-0a1ee91c","meta_rev_id":"78adfb6fb66533762a23e591936c6b53c816764d","_number":664737,"virtual_id_number":664737,"owner":{"_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-07-13 00:52:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":2,"date":"2019-07-12 23:00:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-06-12 07:11:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":1,"date":"2019-07-12 23:00:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-12 07:11:31.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-07-12 23:00:07.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-07-13 00:52:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"160b3a500647459d317dc77c9649f91f46258ace","author":{"_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"},"date":"2019-06-11 22:48:28.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"9f45c27cf8f015c352b638d6066c31b77461bf55","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 00:21:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/37/664737/1/check/openstack-ansible-linters/0646566/ : SUCCESS in 5m 57s\n- openstack-tox-docs http://logs.openstack.org/37/664737/1/check/openstack-tox-docs/53b67c7/html/ : SUCCESS in 2m 51s\n- openstack-ansible-deploy-aio_metal_mistral-centos-7 http://logs.openstack.org/37/664737/1/check/openstack-ansible-deploy-aio_metal_mistral-centos-7/2117301/ : SUCCESS in 1h 14m 11s\n- openstack-ansible-deploy-aio_metal_mistral-debian-stable http://logs.openstack.org/37/664737/1/check/openstack-ansible-deploy-aio_metal_mistral-debian-stable/a8402ae/ : SUCCESS in 56m 12s\n- openstack-ansible-deploy-aio_metal_mistral-ubuntu-bionic http://logs.openstack.org/37/664737/1/check/openstack-ansible-deploy-aio_metal_mistral-ubuntu-bionic/43334b8/ : SUCCESS in 54m 03s\n- openstack-ansible-deploy-aio_distro_metal_mistral-centos-7 http://logs.openstack.org/37/664737/1/check/openstack-ansible-deploy-aio_distro_metal_mistral-centos-7/71a1e7b/ : SUCCESS in 1h 09m 46s\n- openstack-ansible-deploy-aio_distro_metal_mistral-ubuntu-bionic http://logs.openstack.org/37/664737/1/check/openstack-ansible-deploy-aio_distro_metal_mistral-ubuntu-bionic/c7a6107/ : SUCCESS in 1h 00m 31s","accounts_in_message":[],"_revision_number":1},{"id":"fde80261cd244711f31d0c9bca7e96d683416e1d","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-12 07:11:31.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0e3cd10b6a6f108e874a1fa99be3f954b9e6d4d5","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-07-12 23:00:07.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"51212ec03a692ad3f28146025ef0f905e45cbe4c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-12 23:00:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"79f0d76a13fdfe3749a3213c8a36bc8a2795b9eb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-13 00:52:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"74e27cc2a18739a2c62ea13980cf55c9cc62076a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-13 00:52:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/37/664737/1/gate/openstack-ansible-linters/7243fe7/ : SUCCESS in 5m 09s\n- openstack-tox-docs http://logs.openstack.org/37/664737/1/gate/openstack-tox-docs/39e2125/html/ : SUCCESS in 3m 43s\n- openstack-ansible-deploy-aio_metal_mistral-centos-7 http://logs.openstack.org/37/664737/1/gate/openstack-ansible-deploy-aio_metal_mistral-centos-7/59f81f7/ : SUCCESS in 1h 17m 43s\n- openstack-ansible-deploy-aio_metal_mistral-debian-stable http://logs.openstack.org/37/664737/1/gate/openstack-ansible-deploy-aio_metal_mistral-debian-stable/bffb57e/ : SUCCESS in 1h 05m 06s\n- openstack-ansible-deploy-aio_metal_mistral-ubuntu-bionic http://logs.openstack.org/37/664737/1/gate/openstack-ansible-deploy-aio_metal_mistral-ubuntu-bionic/9b48410/ : SUCCESS in 59m 44s\n- openstack-ansible-deploy-aio_distro_metal_mistral-centos-7 http://logs.openstack.org/37/664737/1/gate/openstack-ansible-deploy-aio_distro_metal_mistral-centos-7/5b40927/ : SUCCESS in 1h 17m 22s\n- openstack-ansible-deploy-aio_distro_metal_mistral-ubuntu-bionic http://logs.openstack.org/37/664737/1/gate/openstack-ansible-deploy-aio_distro_metal_mistral-ubuntu-bionic/2d6298d/ : SUCCESS in 1h 05m 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f1e952108037b84af16d5e59fa3dbf93a997550d","revisions":{"f1e952108037b84af16d5e59fa3dbf93a997550d":{"kind":"REWORK","_number":1,"created":"2019-06-11 22:48:28.000000000","uploader":{"_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"},"ref":"refs/changes/37/664737/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_mistral","ref":"refs/changes/37/664737/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/37/664737/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/37/664737/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/37/664737/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/37/664737/1"}}},"commit":{"parents":[{"commit":"296f04acc6b004507bb1b4a1729e34a5be2beae7","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/296f04acc6b004507bb1b4a1729e34a5be2beae7"}]}],"author":{"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","date":"2019-06-10 21:49:37.000000000","tz":-420},"committer":{"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","date":"2019-06-11 22:48:28.000000000","tz":0},"subject":"Fix distro installs on Ubuntu","message":"Fix distro installs on Ubuntu\n\nBeginning in the Stein release, Ubuntu distro packages are now using\nPython3. This requires additionally installing and using the uwsgi\npython3 plugin.\n\nInstall the \u0027python3-mistral\u0027 package instead of individual service\npackages since they provide service config files that conflict\nwith the ones OSA provides.\n\nDepends-On: https://review.opendev.org/#/c/662571\nChange-Id: I4b9846a12382f3ef4acdffce8a6f2338ed45d6ff\n(cherry picked from commit 9dc10013800621a4773abaa33d9991a29e4649e7)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/f1e952108037b84af16d5e59fa3dbf93a997550d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/f1e952108037b84af16d5e59fa3dbf93a997550d"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
